)]}'
{"id":"openstack%2Fdesignate~844344","triplet_id":"openstack%2Fdesignate~stable%2Fxena~Ibd01f6ad3046ca29eec02032e9115183f89a6a4b","project":"openstack/designate","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-06-03 03:14:17.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"Ibd01f6ad3046ca29eec02032e9115183f89a6a4b","subject":"Add proper quota error messages","status":"MERGED","created":"2022-06-01 22:43:07.000000000","updated":"2022-06-03 05:38:48.000000000","submitted":"2022-06-03 05:36:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844344","meta_rev_id":"5935d2374e29dc94d321cf94c86a1d036f0f14f3","_number":844344,"virtual_id_number":844344,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-03 05:36:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-06-02 23:29:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-02 18:25:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2022-06-02 23:29:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-01 22:43:07.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2022-06-01 23:34:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-02 18:25:09.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"d26bd0c95e016d182c5479afade8b095c7268f8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-01 22:43:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e57197d45ee481a5f27c2884b52dac4585b15afe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 23:34:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b51f159f6174469797bfb31c7f0da903 : SUCCESS in 41m 59s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/4852cad7690441329dd13060d1d5c1e8 : FAILURE in 27m 55s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/520cc40cb7f041bb8a5b0df6c952b039 : SUCCESS in 30m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cb6644c9dd864c98b40f933ee9692adb : SUCCESS in 29m 39s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/75a9de53e8f34c92a6c150774c5b6cdc : SUCCESS in 49m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f7d7450f332a469d82ff365a726bb817 : SUCCESS in 30m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/82f645d93d7f460fa0381889aca30138 : SUCCESS in 33m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13ad4513ede14e3d8930b65acf989e75 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/146a9502c1f747df95bc638c0f252498 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cf7dd3456cd4c948441cbbf42e7e231 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7b99aff3d74466f960467476bd8dfa8 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20659855fb7f44dc88803a90e3790a21 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89e476d0f5ff436fa1528d925932c68d : SUCCESS in 7m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f1996a76b7304cc0a248606de8dfb631 : SUCCESS in 31m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f1d12409421ca58c546be87a53808cba02f1f413","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-02 18:25:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b841a8af889c71669902c053efa2b3c82b63f455","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-02 23:29:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"63eecbef8e8206d1e6b89b46704f5871e7c8ab31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 23:31:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ae66e098dd9ca1adf1cf2b917f945027b6d4216","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 01:05:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d4d22c7cc84c4bbe98a8c1392106f031 : FAILURE in 31m 56s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e6c31d9872064dc099d63847ecd823a7 : SUCCESS in 29m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/518bb7e4a8ee487fbc36a605b1be2928 : SUCCESS in 30m 55s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9704a7c8a7e84e03a063d5a1cf4c0c00 : SUCCESS in 49m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e059202dd6a74bab9d12c09c4c7b948a : SUCCESS in 40m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a87f5028aa024defad7726ff886b5cb2 : SUCCESS in 36m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91f837ee8bef4224b2f5f7a7ec9271ca : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb807a2f5ff4470bb21dcf4e924a20a8 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6d8775156df45f59e29e45a33f77d99 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b73e8d84a8e04992b849bc05180db862 : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":1},{"id":"30c47c919b2d61e3534e4a0efcd69b6a750a4d34","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-03 03:14:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b57f782874db796aa3dce607198de9e52305607","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 04:11:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2d49d9794f0f415fb44c626bef5ff5b5 : SUCCESS in 45m 29s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/0c39209e68ea4e0b8fbc53fb21dde460 : FAILURE in 43m 48s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/41dbc081902b47d999a3d7a57c723fa2 : SUCCESS in 42m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d8b0f600bbeb4106a3d2a58ff2c21b8b : SUCCESS in 40m 30s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4f92424d8d874a9a946b5e2badb24e9c : SUCCESS in 55m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/77ecabb4a2584457926565131c60f055 : SUCCESS in 30m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/496362b01c554196990efd4aaa20d636 : SUCCESS in 45m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0067200ca70f4d028a8b8677fd32ba5c : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40548e12ff9f4a969e25f72a203145b2 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3dd505b8186f46f091afbcd8ad95c04c : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbcb285331744bcb89184da203d3038d : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e2f5f98ba83457aa98e6a9ab3ab584a : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a63f4a662843452797b80fc1c25a0657 : SUCCESS in 7m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2c99082e93e046ebb0997b868ec1a134 : SUCCESS in 31m 43s","accounts_in_message":[],"_revision_number":1},{"id":"6b7f2d632a8336876faf784c47ad99aed9a24e1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 04:11:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"03661a9e999a2e0cae98bf71551355a58ec5f916","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 05:36:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d0029b3bc2bb4dcc97eab4a74980a8dc : SUCCESS in 38m 46s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/58e32b9ef2534557a02b5fa4971d7686 : SUCCESS in 34m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6b2dd026cff64a63a3b63a3f92c42a55 : SUCCESS in 36m 38s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/29cf1354625545cca8f57bc441b6e2dd : SUCCESS in 56m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/82903dfc9af44a33a4ddd79de4ea9d78 : SUCCESS in 35m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a32695ffc284478e98b30e5e9f2c5780 : SUCCESS in 36m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7a3c80c2b864f999730638b94b50873 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/785fb74714214e86ba0584c4a3b6826b : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/745113ce54c346b0ab52ad4413449df8 : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/491173717d2f4384bd94f8a7c8ba2d72 : SUCCESS in 9m 09s","accounts_in_message":[],"_revision_number":1},{"id":"c9d6401653fb701be0acf70684a1d00473399560","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 05:36:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5935d2374e29dc94d321cf94c86a1d036f0f14f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 05:38:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d897cd8e5c71452f99d4d4bb455511b8 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6b595bd4c380321eeecc88b2a97ea0c5870415ac","revisions":{"6b595bd4c380321eeecc88b2a97ea0c5870415ac":{"kind":"REWORK","_number":1,"created":"2022-06-01 22:43:07.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/44/844344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/844344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/844344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/844344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/844344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/844344/1"}}},"commit":{"parents":[{"commit":"6a4b524a7fbf9c94bcb1c192e4304bcecd8693ae","subject":"Merge \"Fix support for scoped tokens and default roles\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6a4b524a7fbf9c94bcb1c192e4304bcecd8693ae"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-02-15 02:17:12.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-01 22:43:06.000000000","tz":0},"subject":"Add proper quota error messages","message":"Add proper quota error messages\n\nThis patch adds a message to OverQuota containing information\non the actual quota reached and fixes a minor visual bug\nwith QuotaResourceUnknown.\n\nChange-Id: Ibd01f6ad3046ca29eec02032e9115183f89a6a4b\n(cherry picked from commit 907a029625a7f95369ca5a2f5173efefdb04d28f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6b595bd4c380321eeecc88b2a97ea0c5870415ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6b595bd4c380321eeecc88b2a97ea0c5870415ac"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
