)]}'
{"id":"openstack%2Fdesignate~844353","triplet_id":"openstack%2Fdesignate~stable%2Fwallaby~Ib91e49081855e5f4ec9f46a060d1587dcfeaac29","project":"openstack/designate","branch":"stable/wallaby","topic":"bug/1960764","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-06-04 01:57:54.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"Ib91e49081855e5f4ec9f46a060d1587dcfeaac29","subject":"Fixed incorrect message when zone import failed due to quota","status":"MERGED","created":"2022-06-02 03:12:34.000000000","updated":"2022-06-04 04:04:24.000000000","submitted":"2022-06-04 04:02:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844353-bug/1960764","meta_rev_id":"70e2dfa9773ad1673dbd3ec210b316761be5b6c1","_number":844353,"virtual_id_number":844353,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-04 04:02:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-04 01:57:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-06-04 01:41:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-04 01:57:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-02 03:12:34.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2022-06-02 04:18:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-02 18:22:39.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"7095a303c5da9ea23b5cc42db32e9c577aab1e79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-02 03:12:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"869bea952dccde760239b2642893074014279c4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 04:18:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/be41a63f92e1499c891140f8627c3c70 : SUCCESS in 41m 43s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/784cd4049c544ea09986888a24a6f502 : FAILURE in 17m 50s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/6a08a0166f954a31a8fc2c195e9c8dc5 : SUCCESS in 40m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/672dbdf37eeb41c6b796ca7ca3ac0b8d : SUCCESS in 35m 16s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1a62c67ea43e42b6a6b86395c31050af : FAILURE in 1h 04m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/73decb62441a4111a897568b88c4d3bf : SUCCESS in 33m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bc376c9cfee3408fbe80fcaf7239ff88 : SUCCESS in 42m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/933e2de0e7b74c0e829a47acdd210de9 : FAILURE in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b76d0f88be3340758fed5137d8bf0267 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de6dcc25d7dd4224a37f407d9eda3ea3 : FAILURE in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21f6e1e00dd7445e857262c38aa67716 : FAILURE in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5e2870ef42b44ce8db7e0d59b23c951 : FAILURE in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c59eaf786240440680e1e3d46b95bb42 : SUCCESS in 7m 37s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/34d9acb65f454c8c97679621264b5c15 : SUCCESS in 26m 33s","accounts_in_message":[],"_revision_number":1},{"id":"fb8a3b363317c1b3e7f30aefa7bce3d9362e40d3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-02 18:22:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3a402a9ca8f60dcf954b6dbd690511eadf804604","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-02 22:12:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4649657c0e5a71277de08f47883f7a8fef51502c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 22:58:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4b2567d61dc8473dbc7cdd2c160b01f5 : SUCCESS in 39m 45s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/abc57bde3cc44236b6a79292f40ef413 : FAILURE in 18m 31s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/fe19d5ace5354033ae952d7dee6d7f84 : SUCCESS in 35m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f694e11e21d24f80b16f54ffd73ede80 : SUCCESS in 39m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ea71f1b575c1426b854ab5e702aa433c : SUCCESS in 33m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cdb5d972f4834e1b8d1cb186cc957e20 : SUCCESS in 37m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb0c1f0508b540f198268bae8baea50d : FAILURE in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f28bcc8c16349709e5d79575aa3628d : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df9175240c3d486b8dd91feb65fe890e : FAILURE in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d10bb9427cff41ddbb2636877438a11a : FAILURE in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8acee1c14c424dd5b411e77d54c5e36f : FAILURE in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dbb5dea11df4d52b9c192ec832c2923 : SUCCESS in 7m 50s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/db714e3cb6f64f0fb74748c8acfd774c : SUCCESS in 38m 03s","accounts_in_message":[],"_revision_number":1},{"id":"857fb0ecb865f65f67a56e31b17757daa73b3f88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-03 17:39:07.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4dde83b3d3038de876b940d1f7742d93d3498438","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-03 17:55:36.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"0badcfc3392108cf01b65d93cad40bb536ad2320","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 18:54:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/28b563e7c28b4aec8b929f40d6031aca : SUCCESS in 34m 08s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/10a1a27823ae45c2bc7dbfce37ea5025 : FAILURE in 28m 38s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/5d792b28902f4cf3b7e1e7de537c3e2f : SUCCESS in 53m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/23f19a96e3fb4619b1e87334f2dc574f : SUCCESS in 31m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1e5600e7c5484b70b98fd6f91b85e576 : POST_FAILURE in 49m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0a95768c0944409f8f2fd63e1975d432 : SUCCESS in 53m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5608cf4871e49038b95606eac5f0ca7 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/562005233b604647a4b11f14f34b34e3 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee168dc4aeb74db3b212d2cf0a7f783a : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b32f6ac273a453184f71b0319a7878d : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62d277c11e494c76af1f96a0ef61d1af : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3478a492c6d742b582c6722f25c81db2 : SUCCESS in 6m 55s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/c73ecb7e3ba8408e935f9dbe3961ab44 : SUCCESS in 26m 41s","accounts_in_message":[],"_revision_number":3},{"id":"e694d14d60652ec9f229c69cb03a21e7fc158f6f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-03 20:16:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2ff9efeee9a50809986bcb6393c67c7b003dab41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 21:12:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e8d05cd3b83c4c758f926403bb1f5313 : FAILURE in 18m 51s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/a80086c47dd64f95be9e36f5872e28d6 : FAILURE in 35m 02s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1f787f26a2514ebcb26275cd65c89f9b : SUCCESS in 47m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3e8d8be1bc574202bfd611c1b9f6feac : SUCCESS in 43m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/66c78cf5fdf34d978eb6a135b25fbe3c : SUCCESS in 45m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e25fe89043b4492f9446a45bee853069 : SUCCESS in 50m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bafb8ad45ad341aa8cf9d4fe669063b4 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c39c00d8692d4293b343e2b2871a6bfa : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b51c271723e14d2ba871028c3612547a : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd339b2470bd4db3ba7d5159733f9ef1 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07a969930e264025aa191c57552f2d8c : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffdc11d528214bffa2988a172de7d730 : SUCCESS in 7m 21s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/cd82a11404004333989d267ee1760cd7 : SUCCESS in 26m 10s","accounts_in_message":[],"_revision_number":3},{"id":"a7f73ee9fb5080dade5d44d0ce0194906b904089","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-03 22:56:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9d5616f4ddf4aaed645e7cd0b8cbe7a9c9c48fa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 23:54:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/eb22a6f000574846b85e16421e484e5e : SUCCESS in 49m 36s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/a444dc7eda1a4e5dacdaffafed21b2fb : FAILURE in 31m 31s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/37eaa92bcf734450a4a0edac2853a631 : SUCCESS in 47m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3a12edc777fb4a96b19e06add4c63fc7 : SUCCESS in 41m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c1e32a00047b4d5384bea1e49d8f3a77 : SUCCESS in 46m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f08d8ba12c40466cbd452d234933ef5d : SUCCESS in 52m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c2e79a3ddb246c2922c9d1a4dcbe346 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b323f4245a684fc4afc925149309b64b : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f4f774772984ace83654c74e3beab2b : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a40bdb90a5c44bccb54a6ce1b5b5c9ee : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4744c00aa73c4c7395c6b83b7395bedd : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f206a4f901b435d8ee96d7fb6b572ba : SUCCESS in 7m 18s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/41b354af9c8f422495581077635c2ca0 : SUCCESS in 25m 33s","accounts_in_message":[],"_revision_number":3},{"id":"38917fe48b7066718c3fc4cd02dd3d4b6a882f4b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-04 01:41:33.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"354cd5a27ead5724a3308519345da3ead8d6e21b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-04 01:57:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f3416c89c642ba3aed1c5a52d9aea0efd5730f89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 01:58:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"332e1d51cc01050549764b8f078752067c4a2fdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 04:02:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c7047f0a392b49128e1487135ea87a77 : SUCCESS in 41m 47s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1d1a3d615d4b4dbea18735b814648575 : SUCCESS in 43m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/359cb32cb8264a119c463f922633ffc2 : SUCCESS in 45m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f76d232b34d04b9ca747cbb6bf0b5494 : SUCCESS in 38m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/221ebff4cf844360816758552fc50553 : SUCCESS in 41m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1d1f5a29a384b25ac9ed526de8e1f0b : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6443b770a7d44730a85544f8c80b8a3c : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8c64f5b25254c7ea54b15dc398d6669 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a706a7cce734ca68ffe23a39be4f599 : SUCCESS in 7m 45s","accounts_in_message":[],"_revision_number":3},{"id":"91c979a9f85281585f621eb844e1a43003450832","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 04:02:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"70e2dfa9773ad1673dbd3ec210b316761be5b6c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 04:04:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86f2995088aa47e785c54646bdb975cd : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"29a8069ef54758cd0613a6b43e6ce186a65190e9","revisions":{"83412c904ca35f83e103462a1262a8f8aa398b6d":{"kind":"REWORK","_number":1,"created":"2022-06-02 03:12:34.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/53/844353/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/53/844353/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/844353/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/844353/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/844353/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/53/844353/1"}}},"commit":{"parents":[{"commit":"2e55c3e24bfaf1b03241aa1a4a538d03c29d69ad","subject":"Improve quota API validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2e55c3e24bfaf1b03241aa1a4a538d03c29d69ad"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-02-14 01:36:18.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-02 03:12:32.000000000","tz":0},"subject":"Fixed incorrect message when zone import failed due to quota","message":"Fixed incorrect message when zone import failed due to quota\n\nCloses-Bug: #1960764\nChange-Id: Ib91e49081855e5f4ec9f46a060d1587dcfeaac29\n(cherry picked from commit 7e543c2d22516254cbaab111e9dad87a408ab80d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/83412c904ca35f83e103462a1262a8f8aa398b6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/83412c904ca35f83e103462a1262a8f8aa398b6d"}]},"branch":"refs/heads/stable/wallaby"},"8819d8eb4ba8637c108fd59acc8df46799e73a7f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-06-03 17:39:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/53/844353/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/53/844353/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/844353/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/844353/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/844353/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/53/844353/2"}}},"commit":{"parents":[{"commit":"a4530a8ba3ab2fb731923dfc7f51510ca1badd62","subject":"Merge \"Fix a typo in the tsigkey policy file\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a4530a8ba3ab2fb731923dfc7f51510ca1badd62"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-02-14 01:36:18.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-06-03 17:39:07.000000000","tz":0},"subject":"Fixed incorrect message when zone import failed due to quota","message":"Fixed incorrect message when zone import failed due to quota\n\nCloses-Bug: #1960764\nChange-Id: Ib91e49081855e5f4ec9f46a060d1587dcfeaac29\n(cherry picked from commit 7e543c2d22516254cbaab111e9dad87a408ab80d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8819d8eb4ba8637c108fd59acc8df46799e73a7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8819d8eb4ba8637c108fd59acc8df46799e73a7f"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"},"29a8069ef54758cd0613a6b43e6ce186a65190e9":{"kind":"REWORK","_number":3,"created":"2022-06-03 17:55:36.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/53/844353/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/53/844353/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/844353/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/844353/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/844353/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/53/844353/3"}}},"commit":{"parents":[{"commit":"a4530a8ba3ab2fb731923dfc7f51510ca1badd62","subject":"Merge \"Fix a typo in the tsigkey policy file\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a4530a8ba3ab2fb731923dfc7f51510ca1badd62"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-02-14 01:36:18.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-03 17:55:36.000000000","tz":0},"subject":"Fixed incorrect message when zone import failed due to quota","message":"Fixed incorrect message when zone import failed due to quota\n\nCloses-Bug: #1960764\nChange-Id: Ib91e49081855e5f4ec9f46a060d1587dcfeaac29\n(cherry picked from commit 7e543c2d22516254cbaab111e9dad87a408ab80d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/29a8069ef54758cd0613a6b43e6ce186a65190e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/29a8069ef54758cd0613a6b43e6ce186a65190e9"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
