)]}'
{"id":"openstack%2Fdesignate~844354","triplet_id":"openstack%2Fdesignate~stable%2Fxena~If00b37e72dabd1bc66a78d9a832c3847c12f6692","project":"openstack/designate","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-06-02 19:38:58.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"If00b37e72dabd1bc66a78d9a832c3847c12f6692","subject":"Simplify create zone import implementation","status":"MERGED","created":"2022-06-02 03:27:01.000000000","updated":"2022-06-02 20:42:08.000000000","submitted":"2022-06-02 20:39:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844354","meta_rev_id":"daf94f986f41520e583ffc02eb6d3c01b8fc1495","_number":844354,"virtual_id_number":844354,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-02 20:39:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-06-02 19:38:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-02 18:22:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2022-06-02 19:38:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-02 03:27:01.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2022-06-02 04:27:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-02 18:22:25.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"57b33e03585e11a853f2f96a423843eb56a6fa73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-02 03:27:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"643294787267fe9badd1894b29d022cb1a2bdfbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 04:27:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0a66eeb6d7554fa2b7d75996e727d527 : SUCCESS in 34m 22s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/48a8da9c37bf425b866c0f1afb10afc0 : FAILURE in 25m 13s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4ce8d6b37f774e73b2c2e2dc6c43444b : SUCCESS in 42m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/91a3fbaf4866456f842f8f78b3a95648 : SUCCESS in 30m 57s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9d58d584f72848dd8c21013e64406e42 : SUCCESS in 58m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/950044b75718487c9bc335ea5564e92b : SUCCESS in 39m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cdc25875b6b4485d94d6692fb561c87d : FAILURE in 40m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4aace0c01a41454bb85e892d5a08fd18 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c89b02f9fc834bccb42e5fb4ef094d51 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41a67f074d6d4bb38885e91841920343 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76c167b532034d628809e023a75935c2 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7bcbf401ab4e49688d0f3b5d7abc4e1d : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/478774048a5c40cfac1378a779d61d26 : SUCCESS in 7m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/02d77b6516c847258b93be386867bdf9 : SUCCESS in 25m 33s","accounts_in_message":[],"_revision_number":1},{"id":"221bb876b5da805a5920423e1eb50f2f8ad44844","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-02 18:22:25.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e0e5731c100db0d1a079632d2ed13a596b9cf17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 19:22:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d6ea361c67114892b68f7bdc1878c1c7 : SUCCESS in 37m 14s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/7a00f07a65424e38872902893605948d : FAILURE in 26m 19s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2ea7f1fbb64a4bd28bdfa729cae46ee3 : SUCCESS in 38m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/064c8811621743f3a1d1eb08b1d34d98 : SUCCESS in 28m 49s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a7929b5d12804df98f4a9f569627736f : SUCCESS in 58m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/054e2e5b828f4586877f5c00b24736d0 : SUCCESS in 27m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8ca20c83e18749549d3e9acb99f52ca1 : SUCCESS in 41m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77254e6a25e743568902a60894f3e9f7 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33c6782932c54ff0be4e5082a3cda8d1 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15c181d2882642a9b44ca7335086eac1 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3522099d38346cea3c27a9d546f1469 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a4499302d4d429396861cca49cb5243 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01f9e2dd24ae48b299004efd8d460476 : SUCCESS in 7m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/97f678a9257b4643b4662df0687d89c8 : SUCCESS in 26m 38s","accounts_in_message":[],"_revision_number":1},{"id":"84a1207014aba3cec3acea70dff72d9ff4878351","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-02 19:38:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aadd3e2abccc04143d49d0c1067c25d3ef0d90b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 19:39:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2179c916470f196c428d6850798fce269a11c0c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 20:39:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fe18483e5d0f4de69311ed8f9878aca7 : SUCCESS in 32m 43s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/b5f5f6d17ed342a9a645128a15d05396 : SUCCESS in 31m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f1c2ad04973a414f835f795790bf6cb1 : SUCCESS in 30m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3f5db272803242f69e80d71dc77a2d43 : SUCCESS in 49m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1acfb489fa364ca9bf9a99576981c217 : SUCCESS in 30m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d944d3ee0be74f7ea39b50db8b9c47e7 : SUCCESS in 28m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96e89463179844678e13fdbf04adff47 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/077496fb5b3d4729b0113512b7420f94 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b81119625af488085daddcd4c354724 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c48617078814307a1b0b78913f20403 : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":1},{"id":"b8a26e84428ee8289d89663b624eba5fbc983d7d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 20:39:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"daf94f986f41520e583ffc02eb6d3c01b8fc1495","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 20:42:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/468bc6ef3d4d474ab450afc1f01a437b : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e6922d74f9fed36cda39ceb0633a65b106e98cc9","revisions":{"e6922d74f9fed36cda39ceb0633a65b106e98cc9":{"kind":"REWORK","_number":1,"created":"2022-06-02 03:27:01.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/54/844354/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/54/844354/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/844354/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/844354/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/844354/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/54/844354/1"}}},"commit":{"parents":[{"commit":"e3d7854b07f080dbd0f35021034d83d98e79d21a","subject":"Merge \"Validate worker actions before retrying poll\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e3d7854b07f080dbd0f35021034d83d98e79d21a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-02-03 06:01:52.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-02 03:26:59.000000000","tz":0},"subject":"Simplify create zone import implementation","message":"Simplify create zone import implementation\n\nRemoved unnecessarily complex threading code.\n\nChange-Id: If00b37e72dabd1bc66a78d9a832c3847c12f6692\n(cherry picked from commit eed5042c41f8cd557dd3b53b28e17d520178af64)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e6922d74f9fed36cda39ceb0633a65b106e98cc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e6922d74f9fed36cda39ceb0633a65b106e98cc9"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
