)]}'
{"id":"openstack%2Fdesignate~844370","triplet_id":"openstack%2Fdesignate~stable%2Fxena~I782bf38a68f24a2e7133ff2afad7c96e2ae6b7f0","project":"openstack/designate","branch":"stable/xena","topic":"bug/1958409-stable/xena","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-06-03 07:35:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I782bf38a68f24a2e7133ff2afad7c96e2ae6b7f0","subject":"Fix zone update when adding new Bind9 target to pool.","status":"MERGED","created":"2022-06-02 03:04:22.000000000","updated":"2022-06-03 07:36:29.000000000","submitted":"2022-06-03 07:35:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844370-bug/1958409-stable/xena","cherry_pick_of_change":828679,"cherry_pick_of_patch_set":2,"meta_rev_id":"c28a651dab4bb64b3ca958e427bc3e41a49e9e39","_number":844370,"virtual_id_number":844370,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-03 07:35:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-06-03 03:13:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-02 18:24:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2022-06-03 03:13:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"CC":[{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-02 03:04:22.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"state":"CC"},{"updated":"2022-06-02 04:04:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-02 18:24:42.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"421627136a10d4d07013705933f625fa45d25ee4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-02 03:04:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"00778b44489a18a5ba2ec5e2b02410e8cfbc1592","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 04:04:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4ea028adc3774923811c0963b69c42d4 : SUCCESS in 34m 46s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/cd2734b19018410d83fe69a2457c6d24 : FAILURE in 26m 42s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8a57a3d3e05f4304b11688bc18b08089 : SUCCESS in 33m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/55e255dabb7b4b8c8cc89fa7e851c8fd : SUCCESS in 34m 46s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/feb6b8b6945a43eda002e6d2872fd038 : SUCCESS in 56m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a15938030a6d4fe2abb54c3407780f6e : SUCCESS in 36m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ccfc3e4cf0394289ae76b702d1df0041 : SUCCESS in 41m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae291fd5956a42328865aaeee3ef74b3 : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50073a9b0491454796d819274874ea4e : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a5a663b8646479bbe821c747d7f3e6c : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4652ef4f80f44a7ea1f2b49ac7a74ae3 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42e9eb4eb5a04a67a5b943e93f3ba8c9 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fe0d8b4c7874b2492541bc56d2cd169 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fefd769d9cb4ffaaef7b7012093f5a8 : SUCCESS in 4m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3ffc37a49af94927a926c29e8f757bb3 : SUCCESS in 35m 02s","accounts_in_message":[],"_revision_number":1},{"id":"561cab2354ec51cd676644b4ddc0175fad36b84e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-02 18:24:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d964ecbd5734aa28a498310de6647110f8a6eb6b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-03 03:13:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4bfe260ce777d3dd042bccfbf01a7d9214dd8758","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 03:14:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6987ff8434df5560eef914f5d61994a3491c5f1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 04:27:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bedb17809e0546cd9847ce141bca9f90 : SUCCESS in 36m 35s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/056b8f0aada44b5db33d6f66dee8a7a7 : SUCCESS in 34m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ec0da22f1ebc4822a19d143050197e47 : SUCCESS in 36m 03s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f821d315fa45463cbac5f68f4229719c : SUCCESS in 1h 08m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5fd6ee7aeedc4c08a699e34a96a2e684 : FAILURE in 35m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7d6ccabdad3e41b8b82f0233cc27c1a0 : SUCCESS in 33m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bf5955b78914f96ac665b5ec0195e36 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a240cfd683304818a116970f51808221 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc0522d93ad24576b1693ee787eda24e : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d80579c306294766b7a58cda3da25194 : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d68a4ae6a1fa456eb0577fad3b76b594 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"369c1e19141c3072c122b37aa3524e2e02558e55","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-03 04:40:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e1fc36b3bee73b66469ede4c78ab8a6051801de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 05:54:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b68db8acf0684a5786e33d92cab76196 : SUCCESS in 48m 08s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/2514a7f1fe5a4732ba1053b08924a39a : FAILURE in 26m 17s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ef670d6fb15543039e248d6e3848b1ec : SUCCESS in 45m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0a0a0dc3e970435ca3a87da814dd76a6 : SUCCESS in 47m 12s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9d26e875f8fc40ffb7b39a98305e1e93 : SUCCESS in 1h 07m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c7b03438752f4555b86437324343e2d3 : SUCCESS in 47m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b413c3e133314bf4991ee87e8d4de0df : SUCCESS in 45m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6938234efca84f55a619456b66462b40 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72327f3ea80248fdaf547a4b7f4b6e07 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5ec5b36e3974f07b778f05e6bd29b62 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdde88b8ccb24ed6901061287aeb3060 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cba64f6885f64e91a609cfb3ddbe9c8a : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/579299cdc55a45688108eab8bf888abb : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17549763582b48b2a0e38b27d71d32a6 : SUCCESS in 4m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a192cba96bfe4a73af08b479b0910e54 : SUCCESS in 31m 09s","accounts_in_message":[],"_revision_number":1},{"id":"9614a797b37528626fed7a2e7e472d580a320854","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 05:55:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"76b5a072b97ab4c9858df5d75f84287f2b35924f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 07:35:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a28ff7036af6491791a15586318f0dc5 : SUCCESS in 44m 59s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3b7fc5683e2a4f0398fbe134cf1cae91 : SUCCESS in 42m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c4651208badb41c7a83e566830cca460 : SUCCESS in 42m 07s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/02d07a2c66f845f7accf5ce309b48573 : SUCCESS in 1h 08m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/62ac1997eaa445a48577e6ae08e02ff1 : SUCCESS in 45m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ee0ff19b376d45f5be310384bc5944b6 : SUCCESS in 40m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bc7c0fbbb18429496512c52944558ea : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efa26b6ee64c4f2e89cd625ddc7a770f : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac8e5397d588413581fb288d313237f8 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b322b386783d4a99957eb4a8282ac548 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a95b93af6ad0475f8aa8e7b5cb7d8a5e : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"ca8c8c57e982c4ca6b776094caaa947e8a0bbca7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 07:35:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c28a651dab4bb64b3ca958e427bc3e41a49e9e39","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 07:36:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeccdab98f004d06a0dacbdcaca91eb1 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29ec2e5c80fb4084b53a854ca13056e4 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b1b404e359a516ab468e3568412f4c69a7a0f5b","revisions":{"5b1b404e359a516ab468e3568412f4c69a7a0f5b":{"kind":"REWORK","_number":1,"created":"2022-06-02 03:04:22.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/70/844370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/70/844370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/844370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/844370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/844370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/70/844370/1"}}},"commit":{"parents":[{"commit":"e3d7854b07f080dbd0f35021034d83d98e79d21a","subject":"Merge \"Validate worker actions before retrying poll\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e3d7854b07f080dbd0f35021034d83d98e79d21a"}]}],"author":{"name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","date":"2022-02-10 09:40:35.000000000","tz":60},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-02 03:04:22.000000000","tz":0},"subject":"Fix zone update when adding new Bind9 target to pool.","message":"Fix zone update when adding new Bind9 target to pool.\n\nWhen a new Bind9 is added to the pool, Designate will trigger \u0027modzone\u0027\nrndc command that will fail, unless zone is already present on the backend.\n\nThis change will first verify that the zone is present on the backend,\nbefore an update attempt. If zone is not present - zone creation will be\ntriggered.\n\nCloses-Bug: 1958409\n\nCo-authored-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\nChange-Id: I782bf38a68f24a2e7133ff2afad7c96e2ae6b7f0\n(cherry picked from commit 66cc876eadd9a4ded3afa35aed675c15c60f149d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5b1b404e359a516ab468e3568412f4c69a7a0f5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5b1b404e359a516ab468e3568412f4c69a7a0f5b"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
