)]}'
{"id":"openstack%2Fdesignate~844371","triplet_id":"openstack%2Fdesignate~stable%2Fwallaby~I782bf38a68f24a2e7133ff2afad7c96e2ae6b7f0","project":"openstack/designate","branch":"stable/wallaby","topic":"bug/1958409-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-06-02 22:13:14.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"I782bf38a68f24a2e7133ff2afad7c96e2ae6b7f0","subject":"Fix zone update when adding new Bind9 target to pool.","status":"MERGED","created":"2022-06-02 03:04:41.000000000","updated":"2022-06-02 23:52:58.000000000","submitted":"2022-06-02 23:51:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844371-bug/1958409-stable/xena-stable/wallaby","cherry_pick_of_change":844370,"cherry_pick_of_patch_set":1,"meta_rev_id":"50cdc4a8ff0c8cbb3a6f6deb530aa087183ab922","_number":844371,"virtual_id_number":844371,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-02 23:51:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-06-02 22:13:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-02 18:22:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2022-06-02 22:13:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-02 03:04:41.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"state":"CC"},{"updated":"2022-06-02 04:18:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-02 18:22:45.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"2c4b41efee2e051e1934d652fdf5bdf33a70085a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-02 03:04:41.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"1fea80cc1d5065b5ce6460fa3545f792786d8882","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 04:18:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/01e47fd7912045cd948565fb6a09828e : SUCCESS in 47m 20s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/4db5d3925b26488e8689a3a7d51f6af5 : FAILURE in 38m 12s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4fbef4f72f3940f08262f4b6a33b6303 : SUCCESS in 45m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3778984f1f2f487794f649cea3b5cd93 : SUCCESS in 49m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/be3d7217a80c476294ebe9be7ba5dc34 : FAILURE in 1h 08m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/aa9a9060357a45c7a9caa77cdd19fb02 : SUCCESS in 45m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b1ac5afe19864ad6b898be2def2c0fd4 : SUCCESS in 51m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3aa31a1171c24e97a37b87cf7d5154e7 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bdefc1437e844dab00b052a6ced127f : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c58f503cfc1b4cb48a17914a34d87077 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75434dbcf3ca4a938c4a5fcfff242840 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4da9865f154e4aa0a30ba6ca4798640a : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/171904d11ef842f395fe9f02df69e34e : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/001e971a5d83467f8c66fe53a8e24385 : SUCCESS in 5m 51s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/d9e659e723b14063ae187147cbde9bea : SUCCESS in 30m 38s","accounts_in_message":[],"_revision_number":1},{"id":"893df8edaaaa9c2213d1d22af1e4a93aa51d2223","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-02 18:22:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ac59711d622f1d90ef71471866352ec2a2eb1879","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-02 22:13:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"555ac31c8c7ca669fef6bee4f816bb5546127800","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 22:57:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/48fe20893aae43939503dce2b47ef435 : SUCCESS in 33m 24s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/28c7c8f8d5c34e07874034d1a200422d : FAILURE in 18m 52s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/d74523ddb5af40c88b16d077d059ef0d : SUCCESS in 33m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fa01f3c2aec0416a9efb25d05ded73c2 : SUCCESS in 39m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/715561350ffc47d19c7b3e6fa3d3aa4b : SUCCESS in 33m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e557e0e6b1764d73b99bd9e5d6debe43 : SUCCESS in 42m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed141fdd5e2747c4b8c3cafdcf670009 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cbfd734740b4fcfac5f7bb79a23d1ad : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05339d916c0d425a9a8176e411730846 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7ccf135615141f1bac39ff1e989fd6b : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd607d977c94491e9cb1ed9136e17788 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68d4692e5b7445ed9aa8880894a3793f : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/386b508e97a24358baa1dfb27fd227cb : SUCCESS in 3m 46s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/a3bd8545b3a2445d9faacc3bf5901701 : SUCCESS in 25m 15s","accounts_in_message":[],"_revision_number":1},{"id":"fc665262c75adade79e09a42a2b4bcdd61abde0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 22:57:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4ec4e5007b942cca5413faf0f7a121dcf4243434","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 23:51:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a9d32693f9bc4a8c9e1369b7e2f269d7 : SUCCESS in 47m 31s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/91ba022d976e4c1e9f5b069951d67d59 : SUCCESS in 47m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f59a6f8df9f44e4687971dd497f76257 : SUCCESS in 45m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ab86a9b1e02845469d1c5b63a3882694 : SUCCESS in 47m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0ec736cc21b947098032c590975fd2e7 : SUCCESS in 46m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/751cb45fbd5940358e04de0a40b7eaca : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3031da7ab989468bb95a3af1028357dc : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82b844d1141041abae7454e327983404 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d4b5527e09f4df68a05840d3263a728 : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cef47fd5f2641ef8db7d2b9363fe8c9 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7b1f1a9f0544899675190434f5c4690f0d22c9e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 23:51:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"50cdc4a8ff0c8cbb3a6f6deb530aa087183ab922","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 23:52:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f63345b93e2842318ea09cee64e021f6 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff4343c579ab4e51b7c4ff88465d5c4b : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b6915f00cca2439977722d996f6045309176f49f","revisions":{"b6915f00cca2439977722d996f6045309176f49f":{"kind":"REWORK","_number":1,"created":"2022-06-02 03:04:41.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/71/844371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/844371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/844371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/844371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/844371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/844371/1"}}},"commit":{"parents":[{"commit":"2e55c3e24bfaf1b03241aa1a4a538d03c29d69ad","subject":"Improve quota API validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2e55c3e24bfaf1b03241aa1a4a538d03c29d69ad"}]}],"author":{"name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","date":"2022-02-10 09:40:35.000000000","tz":60},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-02 03:04:41.000000000","tz":0},"subject":"Fix zone update when adding new Bind9 target to pool.","message":"Fix zone update when adding new Bind9 target to pool.\n\nWhen a new Bind9 is added to the pool, Designate will trigger \u0027modzone\u0027\nrndc command that will fail, unless zone is already present on the backend.\n\nThis change will first verify that the zone is present on the backend,\nbefore an update attempt. If zone is not present - zone creation will be\ntriggered.\n\nCloses-Bug: 1958409\n\nCo-authored-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\nChange-Id: I782bf38a68f24a2e7133ff2afad7c96e2ae6b7f0\n(cherry picked from commit 66cc876eadd9a4ded3afa35aed675c15c60f149d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b6915f00cca2439977722d996f6045309176f49f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b6915f00cca2439977722d996f6045309176f49f"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
