)]}'
{"id":"openstack%2Fdesignate~844376","triplet_id":"openstack%2Fdesignate~stable%2Fwallaby~I4e6ea0b6b717d2a1b5cc420874d1bb8fb290e04b","project":"openstack/designate","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-06-02 22:11:51.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"I4e6ea0b6b717d2a1b5cc420874d1bb8fb290e04b","subject":"Fix designate-manage pool update bugs","status":"MERGED","created":"2022-06-02 03:57:11.000000000","updated":"2022-06-02 23:43:02.000000000","submitted":"2022-06-02 23:41:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844376","cherry_pick_of_change":844375,"cherry_pick_of_patch_set":1,"meta_rev_id":"53c7c0c7c2c63cd06c3329ffdcb76d6165597ce3","_number":844376,"virtual_id_number":844376,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-02 23:41:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-06-02 22:11:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-02 18:21:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2022-06-02 22:11:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-02 03:57:11.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2022-06-02 05:17:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-02 18:21:10.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"52c53c3dbb58124338353962a1a9b0572493033a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-02 03:57:11.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"013fc49c3b4435bb2c0a0297efd67152dbfb0746","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 05:17:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e7bb59db49e043faaadb5094ceb3ea5d : SUCCESS in 46m 07s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/8c3720c1cf594d35b5a32694a0bb55e4 : FAILURE in 32m 43s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/75eb98d0a0c74262982f008ba3e8b059 : SUCCESS in 48m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/637adf264cb14ea39b4f95d2921130b8 : SUCCESS in 45m 27s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6d3d1aaddce8496396fa6ad92c81424e : FAILURE in 1h 10m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/57831c9369284b9392239c6c1e5ef33d : SUCCESS in 50m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/615b8560cd454e63ada58eaf5ed95c6a : SUCCESS in 51m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e0a85a39cd04d22bc37f0a439f7abf3 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26d15046fdf24c828c9697f5cac6ff97 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8559c9c880084ccc8a389c567bcce31f : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65cb75030bcb4c8b8590ed7b62c2ff76 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85211a10e7ad40d09567fc6e9a6c061c : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/715e0d70958546f1862bae952703ce1e : SUCCESS in 7m 21s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/6d699c450abb47fb88772e726044fb70 : SUCCESS in 25m 39s","accounts_in_message":[],"_revision_number":1},{"id":"da789075ae98fbe97d9260b8fb0136539af8e830","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-02 18:21:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"338b2eb4de6232f2b3dcba6f8fe0b29d3bc3526a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-02 22:11:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dae8cedabacd17a59d6207fa64cbeec6ad49c5a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 22:57:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8de6a819260e4e86a9be701203585809 : SUCCESS in 38m 34s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/e273e0ce935d40ca9eb8d51852df28fa : FAILURE in 18m 55s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/b467f494659c43a0b2b76a61ad1a0cec : SUCCESS in 44m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ae982c9af2644a949fe840c40b503c57 : SUCCESS in 35m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fbca655a649d46a9a1eb32ae0d39ef37 : SUCCESS in 40m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5cef9efde2ef407f91ec3baf717d9826 : SUCCESS in 33m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41e2d195647c47329af60ff9a79b20a5 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/579a70e454314908aa77cf8369769747 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ecb31e9637747d4a1b7aa0551427eb4 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18babea031b546bcae35e87af1184a70 : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1c68ac2c26449df9001881d43778b3f : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39ced56aab4142ce99b3e9f58d9a67ac : SUCCESS in 9m 09s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/603aa7fef35b475ca0db2653dd79aecb : SUCCESS in 35m 24s","accounts_in_message":[],"_revision_number":1},{"id":"899e87d0bbee73584d264977c14a09e88d01f675","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 22:57:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d70cda8bb3facf64e1206544938b7f97657f572f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 23:41:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/eb68ffe727754d6b8be457de2e259fb9 : SUCCESS in 37m 53s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/816562f7beef46079fbe0c505b044adb : SUCCESS in 40m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/726b677ad9ef4de7bc8608a8e94a74ee : SUCCESS in 40m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/401cc0086b9c48f98ddb8c070af1160f : SUCCESS in 42m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1ac93b1e64214672ae95f4b44005c067 : SUCCESS in 39m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa06dd652d084ae98876e0f3b43eb7e6 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/155c182da66549529e5fdeb85e93d606 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bb89355ed534e9580f6081a8f01cc6a : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b1d3111a4d74cc89f1eb8018c2410de : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":1},{"id":"0c4ded7b0e691e5b21ecd4ccb83f24fa50c71016","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 23:41:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"53c7c0c7c2c63cd06c3329ffdcb76d6165597ce3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 23:43:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a4918a94e94cd1add0a58ef335a10c : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"899cd6e7cc7fb17d4ccc8519d8257774443df939","revisions":{"899cd6e7cc7fb17d4ccc8519d8257774443df939":{"kind":"REWORK","_number":1,"created":"2022-06-02 03:57:11.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/76/844376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/844376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/844376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/844376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/844376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/844376/1"}}},"commit":{"parents":[{"commit":"2e55c3e24bfaf1b03241aa1a4a538d03c29d69ad","subject":"Improve quota API validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2e55c3e24bfaf1b03241aa1a4a538d03c29d69ad"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-24 07:14:46.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-02 03:57:11.000000000","tz":0},"subject":"Fix designate-manage pool update bugs","message":"Fix designate-manage pool update bugs\n\nThis patch addresses a few problems with the manage\ncommand for pools and NS records.\n\n- Fixed an issue where having multiple NS records\n  would break the pool command.\n\n- Fixed a scenario where manually (non-managed) NS\n  records could break the pool command.\n\n- Fixed a potential edge case that could break the\n  pool command.\n\nThe biggest change is that we now only manage the NS\nrecord for the zone itself. This was always the case,\nbut because we didn\u0027t check for this specifically,\nother NS records would conflict with the command.\n\nChange-Id: I4e6ea0b6b717d2a1b5cc420874d1bb8fb290e04b\n(cherry picked from commit bff3d5f6e31fe595a77143ec4ac779c187bf72a8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/899cd6e7cc7fb17d4ccc8519d8257774443df939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/899cd6e7cc7fb17d4ccc8519d8257774443df939"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
