)]}'
{"id":"openstack%2Fdesignate~844379","triplet_id":"openstack%2Fdesignate~stable%2Fwallaby~Ib45eae335160d53bb25e3745fcbe451bffa2cc31","project":"openstack/designate","branch":"stable/wallaby","topic":"fix-errmsg-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-06-02 22:12:10.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"Ib45eae335160d53bb25e3745fcbe451bffa2cc31","subject":"Improve wording for validation error messages","status":"MERGED","created":"2022-06-02 04:09:49.000000000","updated":"2022-06-03 00:15:04.000000000","submitted":"2022-06-03 00:13:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844379-fix-errmsg-stable/xena-stable/wallaby","cherry_pick_of_change":844378,"cherry_pick_of_patch_set":1,"meta_rev_id":"7a46ec89442cedff1802e680ed9d9880ea37b8f5","_number":844379,"virtual_id_number":844379,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-03 00:13:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-06-02 22:12:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-02 18:21:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2022-06-02 22:12:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-02 04:09:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2022-06-02 05:13:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-02 18:21:17.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"a5a5605da75eb7dd22359d697afaf29a9c7ecc0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-02 04:09:49.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"22136109935f81491dc07ac0918716e7951b7d7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 05:13:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7eac620e97754c30801b361afed7ae86 : SUCCESS in 38m 03s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/2c5e62b58a7d4f52b8ec9fa4a4ceac3d : FAILURE in 20m 28s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/235f791d10a74a6d9b1c672658fd4bc4 : SUCCESS in 42m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/497461cd7e1a422a811d3843daa6af88 : SUCCESS in 38m 58s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d5509c3dff664343ba4adf9b563f4355 : FAILURE in 1h 01m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1e6319cf53044197927c2468fbcb0671 : SUCCESS in 43m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a19432a2c2964ddc9318b3124bb6024d : SUCCESS in 39m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b42ed9184214e73bb375d6c115e1e1f : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e2a774a8fe74cd8b4e74d440972584b : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f48a72adecac4cb6a8b2461f4e7ae8bf : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06267b5a69d241b7bc0e3ef0c33cb433 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95695ed6a34442199c8992c54b133baa : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83e7254f27cf45aea008e1005cdaf94d : SUCCESS in 7m 08s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/b37a121bceb6424581001d0f74522779 : SUCCESS in 26m 04s","accounts_in_message":[],"_revision_number":1},{"id":"abd7aab25f5d3e052320deb5d96270d879a82d2b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-02 18:21:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ee08326efd342e82d197e31796f948c6e5b5e147","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-02 22:12:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"23975e0591e1985b202795e8aa625e9a078c67f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 23:03:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fbcbf82967624bffab2edb64ee1edc67 : SUCCESS in 45m 38s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/9aa6d7f9cd2c44b0853c439567ac9bf3 : FAILURE in 29m 57s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/cd8212e495c043cb91e0a0e77d68f424 : SUCCESS in 47m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ec7fddbdc845423db4cb554f87afeaea : SUCCESS in 44m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9a44fa7f31d74fe8b8fac663879ccbc3 : SUCCESS in 46m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b5bf0a7479b14ee9a68ae5582a6d8d9b : SUCCESS in 46m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28c4c4992e9b446ba23f5fda66ce7de0 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21327f52f6844f0cbacbe3d52c31ab98 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37b9daabb6fd4c908f3f83aae203f82a : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d942bb946694cc799d5f7dae58a7ade : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62b746bfafbb4b50ad1c0a5ac2337692 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35aedf40b9704167ae73391c2f5cc336 : SUCCESS in 6m 46s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/723a4f4710174f809f45207de6f67c29 : SUCCESS in 26m 45s","accounts_in_message":[],"_revision_number":1},{"id":"c5e3ccea85bd6a03518913c58655acc826b33d5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 23:04:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d22c0aff8d39648716f6a5b0a7bc04323f51d3cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 00:13:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1834c0cfb2b64bb78af74e0fb2a90d24 : SUCCESS in 39m 48s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/43c254c433f948029fa6e0d0837acbcb : SUCCESS in 37m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fb669954588a48f2b5e54b2a5262e8f6 : SUCCESS in 39m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d58ded4bdf5b4289a0c0a9eb39370023 : SUCCESS in 38m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/90e480963a8c4556b1f614b515665f68 : SUCCESS in 41m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49f253ec35af4cb6b17474ae15b7dd6a : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3092870f047f4082a323346e791b4eed : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a83058f16f6d4317a42d8b69eba385a0 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf2d4bed66ab4736ab44829c0a1cfece : SUCCESS in 8m 47s","accounts_in_message":[],"_revision_number":1},{"id":"722903b4358765c9b13954165e26c7639eb54f44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 00:13:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7a46ec89442cedff1802e680ed9d9880ea37b8f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 00:15:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c381274863b44c71a956ad0fb8d47e3d : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1eb5cb418451d20809b4f009db8821861a61566a","revisions":{"1eb5cb418451d20809b4f009db8821861a61566a":{"kind":"REWORK","_number":1,"created":"2022-06-02 04:09:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/79/844379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/79/844379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/79/844379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/79/844379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/79/844379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/79/844379/1"}}},"commit":{"parents":[{"commit":"2e55c3e24bfaf1b03241aa1a4a538d03c29d69ad","subject":"Improve quota API validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2e55c3e24bfaf1b03241aa1a4a538d03c29d69ad"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-01 12:27:46.000000000","tz":60},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-02 04:09:49.000000000","tz":0},"subject":"Improve wording for validation error messages","message":"Improve wording for validation error messages\n\nCorrect some grammatical issues and harmonize the responses\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Ib45eae335160d53bb25e3745fcbe451bffa2cc31\n(cherry picked from commit 296a5adaf95a221b985c026899662a611d936895)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1eb5cb418451d20809b4f009db8821861a61566a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1eb5cb418451d20809b4f009db8821861a61566a"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
