)]}'
{"id":"openstack%2Fdesignate~844475","triplet_id":"openstack%2Fdesignate~stable%2Fxena~I2ff56bb8e079fa032ca3b211b9d46b3a0906c0c3","project":"openstack/designate","branch":"stable/xena","topic":"logging-stable/xena","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-06-02 22:21:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2ff56bb8e079fa032ca3b211b9d46b3a0906c0c3","subject":"Remove LOG.exceptions from __call__","status":"ABANDONED","created":"2022-06-02 20:22:42.000000000","updated":"2022-06-02 22:21:50.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":819168,"cherry_pick_of_patch_set":1,"meta_rev_id":"db6807938d197962ac18d7731d13a55de8c5b735","_number":844475,"virtual_id_number":844475,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-02 21:18:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-02 20:22:42.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2022-06-02 21:18:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"57220976a2f4027a8b93ef66eceda5a6f2d00ff1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-02 20:22:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2f3559fa41f2479c677794f7abc3c0870d142e19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 21:18:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4e5f613cbbff43beb30a44e7aa6fd73d : SUCCESS in 31m 34s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/f464d9b439ad4d96a68467bb377d0372 : FAILURE in 27m 28s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/def5e8cae5d94ce8ae167febf677d39a : SUCCESS in 32m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1d820f42bd2f4f3fb0859fc094c92ca2 : SUCCESS in 30m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9dda1a0ffca846028eb3ff93c2cc07cf : SUCCESS in 53m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4f8462a2c27b40c980a0ed98317f408c : SUCCESS in 34m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/662a590fa8724f4bb823091e11603721 : SUCCESS in 35m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8c9a45ec26e4fe3ba3c38082bee52c8 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/564ff4828d314dc28e742abae93892f4 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/446e3be590834e38aee069c0fcae7792 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25310274aa58413594d0846dc68f1134 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07fdf6d88d3d45daaf6c5bfb6f387f30 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06dff86a54814712a57a8195f28b3c9e : SUCCESS in 7m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fef3fb697d0b484daafbe1e17c378731 : SUCCESS in 28m 26s","accounts_in_message":[],"_revision_number":1},{"id":"972f6093eba3a9d48d0723c8dfe5c0a2daad2362","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-02 22:04:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db6807938d197962ac18d7731d13a55de8c5b735","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-02 22:21:50.000000000","message":"Abandoned\n\nNot needed on this branch","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b6875e85343881a2f99e5ffe897378187552650","revisions":{"1b6875e85343881a2f99e5ffe897378187552650":{"kind":"REWORK","_number":1,"created":"2022-06-02 20:22:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/75/844475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/75/844475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/844475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/844475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/844475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/75/844475/1"}}},"commit":{"parents":[{"commit":"e3d7854b07f080dbd0f35021034d83d98e79d21a","subject":"Merge \"Validate worker actions before retrying poll\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e3d7854b07f080dbd0f35021034d83d98e79d21a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-11-24 22:10:55.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-02 20:22:42.000000000","tz":0},"subject":"Remove LOG.exceptions from __call__","message":"Remove LOG.exceptions from __call__\n\nRemoving the extra LOG.exceptions from call,\nas this is already handled under _handle_exception.\n\n\u003e if not getattr(e, \u0027expected\u0027, False):\n\u003e     LOG.exception(e)\n\nChange-Id: I2ff56bb8e079fa032ca3b211b9d46b3a0906c0c3\n(cherry picked from commit b9a4e36c0cf09bee7ba64c0b775f293915daefca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1b6875e85343881a2f99e5ffe897378187552650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1b6875e85343881a2f99e5ffe897378187552650"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
