)]}'
{"id":"openstack%2Fdesignate~844476","triplet_id":"openstack%2Fdesignate~stable%2Fwallaby~I2ff56bb8e079fa032ca3b211b9d46b3a0906c0c3","project":"openstack/designate","branch":"stable/wallaby","topic":"logging-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-06-02 22:22:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2ff56bb8e079fa032ca3b211b9d46b3a0906c0c3","subject":"Remove LOG.exceptions from __call__","status":"ABANDONED","created":"2022-06-02 20:22:52.000000000","updated":"2022-06-02 22:22:01.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":844475,"cherry_pick_of_patch_set":1,"meta_rev_id":"e872ca65b1932c79421f1c9b9161c9a109c62b7d","_number":844476,"virtual_id_number":844476,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-06-02 21:21:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-02 20:22:52.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2022-06-02 21:21:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"31c89ebde8b18f91c6eaa97bc695e547d2cf6d48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-02 20:22:52.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"5a6a576c42cffc7e8b27be6fc19e27e557fce075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 21:21:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/81fd28c3883d48af974fec5a984d09a4 : SUCCESS in 43m 50s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/c84f1001c2504296aaa82280194c9161 : FAILURE in 17m 34s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1db46a7058484410b8b3c2c69c30d73e : SUCCESS in 42m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e11d44d653a84d33a9e7bf4ebb19ebb9 : FAILURE in 38m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d6df962444f44cc2b90507bdf9a2899d : FAILURE in 56m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/da65ec4fe5ad4e35b13418d8e910443d : SUCCESS in 40m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9703e16563a34bd8a7a656b8ba27d259 : SUCCESS in 47m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61784d7a8485482ca5305927293b624a : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53b272cbb61d466fbe2a5178c496f28c : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ce9e24022cd4b12adc4626380309acc : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df46d3c7c8d54208bfff5b44845abbe1 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a4a657a3dcf4c3e98b0c80d921bd41c : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ee6aa32ba9f40b4b1ae3f349acb32b8 : SUCCESS in 8m 35s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/068d4626205444dc9e81b71fe71e33a0 : SUCCESS in 29m 11s","accounts_in_message":[],"_revision_number":1},{"id":"fd644b5db761357995ebaae6a5bc46496d4e2f27","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-02 22:04:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e872ca65b1932c79421f1c9b9161c9a109c62b7d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-02 22:22:01.000000000","message":"Abandoned\n\nNot needed on this branch","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e5f14dd6e704ec7393e15b34d1e61815a0f10d2","revisions":{"2e5f14dd6e704ec7393e15b34d1e61815a0f10d2":{"kind":"REWORK","_number":1,"created":"2022-06-02 20:22:52.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/76/844476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/844476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/844476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/844476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/844476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/844476/1"}}},"commit":{"parents":[{"commit":"2e55c3e24bfaf1b03241aa1a4a538d03c29d69ad","subject":"Improve quota API validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2e55c3e24bfaf1b03241aa1a4a538d03c29d69ad"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-11-24 22:10:55.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-02 20:22:52.000000000","tz":0},"subject":"Remove LOG.exceptions from __call__","message":"Remove LOG.exceptions from __call__\n\nRemoving the extra LOG.exceptions from call,\nas this is already handled under _handle_exception.\n\n\u003e if not getattr(e, \u0027expected\u0027, False):\n\u003e     LOG.exception(e)\n\nChange-Id: I2ff56bb8e079fa032ca3b211b9d46b3a0906c0c3\n(cherry picked from commit b9a4e36c0cf09bee7ba64c0b775f293915daefca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2e5f14dd6e704ec7393e15b34d1e61815a0f10d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2e5f14dd6e704ec7393e15b34d1e61815a0f10d2"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
