)]}'
{"id":"openstack%2Fdesignate~844956","triplet_id":"openstack%2Fdesignate~master~I81a4883fd48653c3b0d60ac07210fc0de55ac12a","project":"openstack/designate","branch":"master","topic":"bug/1977826","attention_set":{},"removed_from_attention_set":{"21778":{"account":{"_account_id":21778,"name":"junbo","email":"junbo4242@gmail.com","username":"junbo"},"last_update":"2022-06-09 04:27:28.000000000","reason":"junbo replied on the change"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-06-07 21:58:27.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"I81a4883fd48653c3b0d60ac07210fc0de55ac12a","subject":"Check zone_id in notification_handler to avoid UUID format issue.","status":"MERGED","created":"2022-06-07 10:49:37.000000000","updated":"2022-08-26 09:45:01.000000000","submitted":"2022-06-09 06:27:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844956-bug/1977826","meta_rev_id":"c85578bffef7f325d6728648888ee5352f3ee819","_number":844956,"virtual_id_number":844956,"owner":{"_account_id":21778,"name":"junbo","email":"junbo4242@gmail.com","username":"junbo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-09 06:27:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-08 14:51:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-06-08 01:27:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-08 14:51:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-07 12:05:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-07 21:21:54.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-06-07 23:06:07.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-08-26 09:45:01.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"}],"messages":[{"id":"565508e772aaa56ef9806c9ed6108c22de8c2376","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21778,"name":"junbo","email":"junbo4242@gmail.com","username":"junbo"},"date":"2022-06-07 10:49:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4c495ffd06560554cafdcead5ede49b90d912cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 12:05:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9ad7e2b2bdd24f8f883e16de44c609cc : SUCCESS in 45m 38s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/efb925fe78cf41a59d534c7c2613d269 : SUCCESS in 45m 21s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0dd27f3d72404377b8ca39f5c4b340cc : SUCCESS in 54m 41s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/7246f7e1f5034b518b83603b10baeac3 : SUCCESS in 48m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0d22b832832d420db9117a5c597737a6 : SUCCESS in 44m 56s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c76295fc70814791884ea20c67e92fb2 : SUCCESS in 1h 09m 11s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6b9fd92432c84715a0234a4bce8ef514 : SUCCESS in 56m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4aadc342a3b04d5896e47feada6caa4b : SUCCESS in 44m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2125e33688d349aa9fe92d4d41085c6a : SUCCESS in 37m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b072e49a148400b8f6b5d4b213c1788 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4290f4a0577f44fe93ef610d46c2fff3 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/439ce2e1466a41a991f7af61a940cb2a : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/998e99fc3d63467ea04b433b09b6670e : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd58f8e6514e4697a76ac0c2025973ea : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1121c3a3be1a4052b98fb8bc268b3533 : SUCCESS in 9m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9afc78db5a014e018eb2e7195cf2218c : SUCCESS in 29m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b1289a244d1343079cd05c40ed62bfd4 : SUCCESS in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92ab7d4b34118ecf0552ec6b08c42a2fd25cf9aa","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-07 21:21:54.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"73dcd3f5f50ddbde7d08aad5f73235e1a478223f","author":{"_account_id":21778,"name":"junbo","email":"junbo4242@gmail.com","username":"junbo"},"date":"2022-06-07 21:56:27.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9497965a1659180f1deb2d8d9a440bd6599c93b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21778,"name":"junbo","email":"junbo4242@gmail.com","username":"junbo"},"date":"2022-06-07 21:57:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ae01357342b667e775fe3f2494172fb5c63742c0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-07 21:58:27.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd4a45120670a3ee333469637116a5a8be144633","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 23:03:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/50cc84a1e2a348e9b1fc6a030544412c : SUCCESS in 44m 59s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/83316931e0c24e7dbf55027164936695 : FAILURE in 13m 29s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9c1d6606193474bb173a1610f5a561e : SUCCESS in 52m 37s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/70950e23f1014d67940138df726dab77 : SUCCESS in 44m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/419d2c19841746c3b9a859287aff7760 : SUCCESS in 41m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b82a574a4b314c65b7a0cb48c294d590 : SUCCESS in 1h 01m 51s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/360ccf46a7914d5cacac4c3cf227ce00 : SUCCESS in 54m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/537ab154a41a40d5804fdaca806c288a : SUCCESS in 41m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6abe9fd71bb74f2bbadb310dc1e82334 : SUCCESS in 45m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75f3716be2d744059e27cea318a1ab10 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6e043d9980c40bca4b872557ad2c251 : FAILURE in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4b4c53f5aac4b5e8f5b0e2d59d09f7c : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e07331c7df83496aa0111c5135ce2547 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4714b9ee26c4441aad2bac2f066a2ec5 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e43f758ecd014c05b4db8cf00c078135 : SUCCESS in 7m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c1d7e4a129c945e186e8616684518fd5 : SUCCESS in 29m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/fb04fbc6394940d3a50cabec80d81249 : SUCCESS in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b33e8877b4e2c95535f38fac08e1b04cae287859","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-07 23:06:07.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"58333d8dfa7a538bb83dae10f64d407276efe9a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21778,"name":"junbo","email":"junbo4242@gmail.com","username":"junbo"},"date":"2022-06-08 00:33:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b27bd6f4ce5df8c322842f582404b0234fb40dd4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-08 01:27:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8e409e5dbc9afe74060fef77a3f43975ad11dede","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 01:48:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3da5cd1b5631439aac181c3f472d9781 : SUCCESS in 32m 10s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/acc0d74dee7d421f844f1fba4ded98f5 : FAILURE in 10m 54s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2f5a37058c64dc99fab2b9bbdb84a5f : SUCCESS in 38m 50s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/0adf9758097d409987528fa412423e4e : SUCCESS in 36m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e4eb6d9f6db64e09928ceae7674adc58 : SUCCESS in 35m 20s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/df21e0c2239c429c80db5ce79e9f22a6 : SUCCESS in 1h 09m 56s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/094d9edc69b34ddcb36743615abc72ce : FAILURE in 48m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9b4b2490e9b24ad08b7789ddd6b6260a : SUCCESS in 46m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3632c12a27e54172b83d3dc49f6d23cf : SUCCESS in 34m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2d5f7b1218d409f96cd6656bb7ea5e2 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70d5b94750b34295a7784cc43d37152b : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f036f39637c481283d9b58f734450de : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2c288ddf65f40c58b7a3de91dbe2e14 : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1793cc09f4444fd91898d660a315877 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eac75ef20a14e6181b94badc564bd92 : SUCCESS in 8m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b79227d00fc747f4acb8531633972284 : SUCCESS in 30m 01s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/44df2fe31a2445a5b8176889e2abbf14 : SUCCESS in 4m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b8db3acc8ae89d833ec80989d9ad570429df508c","author":{"_account_id":21778,"name":"junbo","email":"junbo4242@gmail.com","username":"junbo"},"date":"2022-06-08 09:12:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d3e6d43e0c9754c5524e3dba251ef0905d4d0cb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 10:08:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ceb47a3efec34cbc8c7dc015ad7d5b95 : SUCCESS in 35m 36s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a4d2dd47ae9a4e699baac3b5db757dc7 : SUCCESS in 44m 17s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c1b1f163e114f398cb0736f9f5367da : SUCCESS in 42m 25s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/acc17857c0c84bd78681f8cd3a0bd573 : SUCCESS in 42m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c3aad451f4ce482cad25519357ed7c94 : SUCCESS in 40m 15s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/30543cb108f3414db024f413c04679a9 : SUCCESS in 49m 51s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/09ee35f120b848c99f0601a94415970e : SUCCESS in 53m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/285cfdf9c44742b095ddc0fcde3ae638 : SUCCESS in 40m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/567a918750f142ca8ddcf4d42051641f : SUCCESS in 42m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab505c0215d34d218da7865fb3a9a2d5 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e6ad32c96b545838be9f5dd24ebeaf5 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/364fcdba19d04db08d80c989c7589184 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45faf31d49464075b6fdb69745666bdf : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c75bb8d57b194b4eb26cdaba5de7289f : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81484546c72d42749c425b9aba8de5f9 : SUCCESS in 7m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e579b9f79fa04719b23b6cdb5ad69615 : SUCCESS in 39m 09s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/bc6f0b8e302a4905a0824e51f1058c7a : SUCCESS in 4m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"66911235181e66a853e890b31620864f06df2456","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-08 14:51:20.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3fb66102d4ea38e6078838257a357c0b56923a2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 14:52:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f2941ed73ad07e9a657aace2d0c1e8be372608c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 16:12:29.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8c7173c4d84e4e9499c965d693cddd27 : SUCCESS in 36m 51s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/0bc8930186a4420287f1a28aadc04488 : SUCCESS in 38m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f56b2392add94880a1490b54ea0705b9 : SUCCESS in 1h 08m 30s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/811749a6209b4b9cb71daa83fca965ba : FAILURE in 20m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c3737546814349e0bc9c0be49c82f437 : SUCCESS in 48m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/775b573ae7bd457194d3c4026ad9179b : SUCCESS in 49m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8079f451c69842548a29ae8fca96e19f : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c80a3d3211b42b7bda232590af3810b : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1436ef6a21574159b1607c8a9009ff52 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11e0d043d8644532af9adeb97c847d91 : SUCCESS in 7m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0ab67da33013445882e8d54242ef524e : FAILURE in 25m 05s","accounts_in_message":[],"_revision_number":3},{"id":"b8f718bd1b45b9251332b5e287bf4d6c392e0c82","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-08 16:26:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"60972de12f94b10740ac1af81d0209c9285afa79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 17:27:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ac5d805fdaa2431399fc99eb7d84085b : SUCCESS in 50m 05s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7c741198836a4cd1b5cca5de736e5f8d : SUCCESS in 50m 51s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe75fffc24ac4593baa0eb31bc8cd13c : SUCCESS in 47m 19s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/44ba26b3c94f45179328d9e53c7b5a38 : SUCCESS in 43m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/567ab7e0674d4b3ba6b82eed42463e04 : SUCCESS in 42m 43s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/1a502c518b844013b9f3f745370d1a25 : SUCCESS in 58m 50s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/abaebd34404b43cfa71a36bc905cc9a7 : SUCCESS in 57m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bf00a67f5a5a46d2a6d64ab2f527fb8f : SUCCESS in 41m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bd8fd38050784e75bd3c5ace3f878b0d : SUCCESS in 43m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9706f0623ed4e18883d70232155cb61 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3a48f6d08ac48c1863659145b2d1b50 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5aa3fcf060724ab2900bd5e6b59673dd : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79f52f9d43b146f5ba9e8d231eee9fb0 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d4840085667417d91be73701807e050 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33964598f91c4ac9ae3d608711f5effe : SUCCESS in 8m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/646c32914c61461a8ec38893e7875d40 : SUCCESS in 29m 45s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9971809ab36d40ceb2d6d736933ada35 : SUCCESS in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b967fd25b3818658194cbcd5bba29af8982b004e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 17:27:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"81e4d07df5586681ba2e7cb81e8c48d2da914394","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 18:32:11.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8f48e8a3d21e41c58caf3140aed83844 : SUCCESS in 41m 55s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3065975e60384e7aa37ca964d1095510 : SUCCESS in 38m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ed9206977dfa41a8bb2b2874da49918a : SUCCESS in 37m 43s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9ecb63f6088d4055b15fc54e7f40b172 : SUCCESS in 1h 03m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/42d84854af034ddc86363be65ab05e7f : FAILURE in 43m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1f9691d3e8214a0696eb0c227ed6f6c3 : SUCCESS in 37m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af52ceffd3584859a1d6767e7e09afe6 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/053eebfd7dbd434e8706e411dadbff58 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0eaefa56def4267b58d0cea07337b21 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4a748d2f5884861b448d38e5b58d4b3 : SUCCESS in 8m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5a870457b7774c1f85b5820cc4c8f9f2 : SUCCESS in 29m 54s","accounts_in_message":[],"_revision_number":3},{"id":"90f3a732770d8ba33020335435bc7904586ed973","author":{"_account_id":21778,"name":"junbo","email":"junbo4242@gmail.com","username":"junbo"},"date":"2022-06-09 04:27:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6ae186b6f3ab2a96262557076c1390b2a0b7549b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 05:36:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8ab3c9e0e127447dbf97278de829eae6 : SUCCESS in 46m 28s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7a9816eaf05449269db3b5f09338b9ab : SUCCESS in 54m 04s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c39ca49b0f9549e988264cb1885ab40f : FAILURE in 27m 48s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2560f568401948ac99f80dc720d19e4b : SUCCESS in 44m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ba7a4a099b624bd4b574604c5b6e2ff2 : SUCCESS in 44m 23s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8b040739d89b4298bab3ff1e19285c7d : SUCCESS in 42m 13s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0f9f869fa7914ea192b726e26cfe67ec : SUCCESS in 1h 03m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7778cc46d06647dda7f66de4cfca22fb : SUCCESS in 42m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/589b6e9384c141dab0c5f439407906f9 : SUCCESS in 47m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be017b2d794340d2b9c72a8c2013fc5a : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3697d4442d2a4d858bd114022054144c : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/684931b0838f42da9901129f59e12031 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9dd97aeef5cc44f598bc781ce8c7effa : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/360e02800a60494caf8fcfe93ac6d034 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f357bb470524cecbe574c025623423d : SUCCESS in 6m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e3b27af635e74395819737829448f116 : SUCCESS in 27m 20s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/010dbce6a6944e36bd6ac34df0a206c1 : SUCCESS in 3m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"27b538b42c81722b5ecad4895e8a8eb9e3e9616b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 05:37:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fdc5a5843d4c6993aec419a4a2d8559758d216d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 06:27:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5c90143291f245dc966b9ee0dda14ca7 : SUCCESS in 38m 08s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/fd4136c5de974f5cbb62cd7a6c91c860 : SUCCESS in 32m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/08750644f19a4ddb91194632ca8f7558 : SUCCESS in 36m 09s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b444a96dcbfa4b47b8a38fe4de22e783 : SUCCESS in 48m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1bfc0d053ded4727991a273bfc0cb004 : SUCCESS in 45m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/36168352d3114b828ccf50932d14610c : SUCCESS in 32m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c34ae0981d904489bc843ba11b72b3ea : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4839f68c1ed84d4ab80f7b5addb12544 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33b3ecf69dd84e5984866d09d8cff1bd : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14f118568b24465eadc665a55b8a8787 : SUCCESS in 6m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/158a944b98c0436d8c7f3e903156e769 : SUCCESS in 32m 04s","accounts_in_message":[],"_revision_number":3},{"id":"f54d04937a5e4983971695a76f7a01f3c84446c6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 06:27:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0742cc398f11e85319320a5e6077f418f9d8310a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 06:29:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a1fd846a159486795f6ca13b690a728 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3},{"id":"c85578bffef7f325d6728648888ee5352f3ee819","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2022-08-26 09:45:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"42dac06e0cdef9eccabc7085d0ff9da47563b318","revisions":{"67ce011805b0f1cdea0d9768e518bfeb5c1d3029":{"kind":"REWORK","_number":1,"created":"2022-06-07 10:49:37.000000000","uploader":{"_account_id":21778,"name":"junbo","email":"junbo4242@gmail.com","username":"junbo"},"ref":"refs/changes/56/844956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/56/844956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/844956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/844956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/844956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/56/844956/1"}}},"commit":{"parents":[{"commit":"e8fdabfc38a1fc0a946cc9f4efa2bafb036d1df2","subject":"Update neutron integration link.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e8fdabfc38a1fc0a946cc9f4efa2bafb036d1df2"}]}],"author":{"name":"junbo","email":"junbo4242@gmail.com","date":"2022-06-07 10:47:27.000000000","tz":480},"committer":{"name":"junbo","email":"junbo4242@gmail.com","date":"2022-06-07 10:47:27.000000000","tz":480},"subject":"Check zone_id in notification_handler to avoid UUID format issue.","message":"Check zone_id in notification_handler to avoid UUID format issue.\n\nCloses-Bug: #1977826\nChange-Id: I81a4883fd48653c3b0d60ac07210fc0de55ac12a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/67ce011805b0f1cdea0d9768e518bfeb5c1d3029"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/67ce011805b0f1cdea0d9768e518bfeb5c1d3029"}]},"branch":"refs/heads/master"},"1b4ba4a2bccf042358766ec93a57c4bf473a0075":{"kind":"REWORK","_number":2,"created":"2022-06-07 21:57:24.000000000","uploader":{"_account_id":21778,"name":"junbo","email":"junbo4242@gmail.com","username":"junbo"},"ref":"refs/changes/56/844956/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/56/844956/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/844956/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/844956/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/844956/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/56/844956/2"}}},"commit":{"parents":[{"commit":"e8fdabfc38a1fc0a946cc9f4efa2bafb036d1df2","subject":"Update neutron integration link.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e8fdabfc38a1fc0a946cc9f4efa2bafb036d1df2"}]}],"author":{"name":"junbo","email":"junbo4242@gmail.com","date":"2022-06-07 10:47:27.000000000","tz":480},"committer":{"name":"junbo","email":"junbo4242@gmail.com","date":"2022-06-07 21:57:15.000000000","tz":480},"subject":"Check zone_id in notification_handler to avoid UUID format issue.","message":"Check zone_id in notification_handler to avoid UUID format issue.\n\nCloses-Bug: #1977826\nChange-Id: I81a4883fd48653c3b0d60ac07210fc0de55ac12a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1b4ba4a2bccf042358766ec93a57c4bf473a0075"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1b4ba4a2bccf042358766ec93a57c4bf473a0075"}]},"branch":"refs/heads/master"},"42dac06e0cdef9eccabc7085d0ff9da47563b318":{"kind":"REWORK","_number":3,"created":"2022-06-08 00:33:08.000000000","uploader":{"_account_id":21778,"name":"junbo","email":"junbo4242@gmail.com","username":"junbo"},"ref":"refs/changes/56/844956/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/56/844956/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/844956/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/844956/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/844956/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/56/844956/3"}}},"commit":{"parents":[{"commit":"e8fdabfc38a1fc0a946cc9f4efa2bafb036d1df2","subject":"Update neutron integration link.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e8fdabfc38a1fc0a946cc9f4efa2bafb036d1df2"}]}],"author":{"name":"junbo","email":"junbo4242@gmail.com","date":"2022-06-07 10:47:27.000000000","tz":480},"committer":{"name":"junbo","email":"junbo4242@gmail.com","date":"2022-06-08 00:33:10.000000000","tz":480},"subject":"Check zone_id in notification_handler to avoid UUID format issue.","message":"Check zone_id in notification_handler to avoid UUID format issue.\n\nCloses-Bug: #1977826\nChange-Id: I81a4883fd48653c3b0d60ac07210fc0de55ac12a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/42dac06e0cdef9eccabc7085d0ff9da47563b318"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/42dac06e0cdef9eccabc7085d0ff9da47563b318"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
