)]}'
{"id":"openstack%2Fdesignate~845507","triplet_id":"openstack%2Fdesignate~master~Ie89f5eff1cfb30b0e4da32c82f69c97ee7513c9c","project":"openstack/designate","branch":"master","topic":"rpc","hashtags":[],"change_id":"Ie89f5eff1cfb30b0e4da32c82f69c97ee7513c9c","subject":"Removed RPC calls from MDNS and moved them to the Worker","status":"ABANDONED","created":"2022-06-12 22:16:36.000000000","updated":"2022-06-13 23:20:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"a217101ab0f14e7262af598dc1a00d39109e3831","_number":845507,"virtual_id_number":845507,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-12 23:23:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-06-12 23:23:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c337ddef0c945c36e4bd0950be3e99714fbf7882","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-12 22:16:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd90af63e1ec269c67d34ea1bdb20ed8025cc897","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-12 22:25:20.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"8265cc4869a91ddc369817505c917886900b647e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-12 23:23:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/781955fe902748c99683e74f8872af86 : SUCCESS in 35m 42s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c57222dcc77146218feab43204792d13 : SUCCESS in 36m 02s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/999f3efc8d274fbca4466840e95d6391 : SUCCESS in 1h 01m 28s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/155b7910a8534ad681033f69b9170563 : SUCCESS in 43m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ccb3a0fa79224f8d85248a24791c1564 : SUCCESS in 48m 28s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0a39eef222454dc7bec005ad9247d0c0 : SUCCESS in 1h 01m 33s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f4012d221301473cab0ee8ee13ea7423 : SUCCESS in 55m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d352ab20f00e493e94bf15efffbe82ad : SUCCESS in 45m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d0732ba52fed49f894f81f94fc4b89fc : SUCCESS in 40m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f9a76ed8f324d758d439b906597d769 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4d55e66a1124a7d8a25774208bf114f : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29aec8caf90e4ff9b0b27997f893452a : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a7e48cda3d74cbe81b638f2ccb274a7 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9d784508d7d4069be3203f3c2badb0b : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/601afd823b284192b9daf7e7a8906b3c : SUCCESS in 8m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2b0ce349525a43578a7d66529b9cfbaa : SUCCESS in 31m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c95887ba08494ea6aa205c040016c6e8 : SUCCESS in 4m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a217101ab0f14e7262af598dc1a00d39109e3831","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-13 23:20:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87df04e6b87d83033ec0d4d2842fcd917932a2f6","revisions":{"87df04e6b87d83033ec0d4d2842fcd917932a2f6":{"kind":"REWORK","_number":1,"created":"2022-06-12 22:16:36.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/07/845507/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/07/845507/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/845507/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/845507/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/845507/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/07/845507/1"}}},"commit":{"parents":[{"commit":"d7579909dbc01cbf7f5b2e291a312649399b67a0","subject":"Removed unused metrics backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d7579909dbc01cbf7f5b2e291a312649399b67a0"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-06-12 22:14:47.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-06-12 22:14:47.000000000","tz":-420},"subject":"Removed RPC calls from MDNS and moved them to the Worker","message":"Removed RPC calls from MDNS and moved them to the Worker\n\nThis patch moved the last remaning RPC call away from the\nMDNS service. This means that the MDNS service now only\nhandles incoming DNS queries.\n\nChange-Id: Ie89f5eff1cfb30b0e4da32c82f69c97ee7513c9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/87df04e6b87d83033ec0d4d2842fcd917932a2f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/87df04e6b87d83033ec0d4d2842fcd917932a2f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
