)]}'
{"id":"openstack%2Fdesignate~845526","triplet_id":"openstack%2Fdesignate~master~I606fcc1d8f8f4fe74eecfc261f01f1b785f29f6e","project":"openstack/designate","branch":"master","topic":"fix-queue-config","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-06-13 02:21:16.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I606fcc1d8f8f4fe74eecfc261f01f1b785f29f6e","subject":"Do not declare queue at a pipeline level","status":"MERGED","created":"2022-06-13 01:03:28.000000000","updated":"2022-06-13 03:14:36.000000000","submitted":"2022-06-13 03:12:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845526-fix-queue-config","meta_rev_id":"db223d8c68e9a2748f01786d88b9ea57a3fa99ee","_number":845526,"virtual_id_number":845526,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-13 03:12:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-06-13 02:12:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-13 01:58:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2022-06-13 02:12:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-13 01:58:26.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-06-13 02:11:30.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-06-13 02:14:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6504365c55aed4cb46023a851178295237ba3617","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-13 01:03:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee132b96cea4cc57d67912cdb232df48268d18bb","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-13 01:58:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bd785f830ce306f37fcd55ffd969fcfe29855f30","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-13 02:11:30.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"696cd5e33a992c86af83327613dadd40cf3bedc8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-13 02:12:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9d7b364e9395e4bcc26f483bb78c29b908ca846","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 02:14:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/30d0b0150e8c432894f3a73a90e5ddb3 : SUCCESS in 38m 39s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/de78d8a540a64cc6a440154b13b06a37 : SUCCESS in 45m 23s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ec22ba33f034d80a3174b5f67e3d226 : SUCCESS in 41m 47s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/03e936cc275a4ea18732e947b5d8e612 : SUCCESS in 38m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bd529a19ac0448cdb56b6e7ab4bb778b : SUCCESS in 36m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/636e36f5015241138f1897656d4302b6 : SUCCESS in 54m 27s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5d839b48601a4fb995f03d462a40d479 : SUCCESS in 57m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7715b827338d42ec8d7bb577418f2994 : SUCCESS in 38m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/94c45acf8ab241e58a643cb0ca92e0bb : SUCCESS in 35m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5d6074b2ce14e31932df2939f51f076 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5cda2fb1f614d138fd18bd4c75c8513 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ae939f1bc1946f384f137529dee8776 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7525fc6917c043dab3830d9b8ae432ed : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c8d9129cff64202b280f9dbcc00efcb : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aabd86e7e2849bdbaf22db5a47654a8 : SUCCESS in 7m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/77f5d58a828047b5ae71c2da16df42d5 : SUCCESS in 26m 03s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/48c7a4e5a0a549aba1aca17c6ef567f9 : SUCCESS in 4m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ceff5f79976a6c36c375cd12ca7097240067b567","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 02:15:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f9cbfc0d264f6f78da1e0399bb68e21488aa0518","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-13 02:21:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31c6511ab42a3e1182b78ecc8a564d69ca08bd0a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-13 02:22:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea2d889ff71988d32350637e4b8d000fbdcfedd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 03:12:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5f360d6f1afd4fca80b7b1ca8e4ffa2c : SUCCESS in 49m 46s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8726610323e84370b259c13f904c52bd : SUCCESS in 43m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/38bd7b89ea51411883823ee7f173a4bb : SUCCESS in 44m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b4ded9475f1e4528b300cca3b69936f5 : SUCCESS in 55m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2635d64e65004d07821115f51ceb8255 : SUCCESS in 40m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8de370c4ac3b4f1c8e4cd86604d33c80 : SUCCESS in 43m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/265b5d43f1be4ed0a9755a37415ffa85 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cad4cdec21d249b5a6cd4dbf6ce9aa35 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/222dac3746444920942c359ccea1f9c9 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a165dfffdc73497ba051df8387837404 : SUCCESS in 9m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/57084f45adb347e39fbd6f268c61f454 : SUCCESS in 27m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1384fbe10f46e2f8ad088f1b91ab29dbcc73d31f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 03:12:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"db223d8c68e9a2748f01786d88b9ea57a3fa99ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 03:14:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7335357e88542a186b9a630e1646957 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0ccec12126d7c0fed128f9289ad50f98ea0a21f","revisions":{"f0ccec12126d7c0fed128f9289ad50f98ea0a21f":{"kind":"REWORK","_number":1,"created":"2022-06-13 01:03:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/845526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/845526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/845526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/845526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/845526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/845526/1"}}},"commit":{"parents":[{"commit":"404db698ca59a5f9254377d568bd42d7b1a5225f","subject":"Merge \"Fixed incorrect quota exception message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/404db698ca59a5f9254377d568bd42d7b1a5225f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-13 01:02:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-13 01:02:28.000000000","tz":540},"subject":"Do not declare queue at a pipeline level","message":"Do not declare queue at a pipeline level\n\nZuul deprecated declaring shared queues at a pipeline level with\nrelease 4.1.0[1]. This removes pipeline-level declarations to use only\ntop-level ones.\n\nSee [2] for details.\n\n[1] https://zuul-ci.org/docs/zuul/latest/releasenotes.html#relnotes-4-1-0-deprecation-notes\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2022-May/028603.html\n\nChange-Id: I606fcc1d8f8f4fe74eecfc261f01f1b785f29f6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0ccec12126d7c0fed128f9289ad50f98ea0a21f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0ccec12126d7c0fed128f9289ad50f98ea0a21f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
