)]}'
{"id":"openstack%2Fdesignate~846103","triplet_id":"openstack%2Fdesignate~master~Ib3e9d87c29a612e4c77212bd3a414c53ee1d9fe0","project":"openstack/designate","branch":"master","topic":"hacking","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-06-16 17:23:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib3e9d87c29a612e4c77212bd3a414c53ee1d9fe0","subject":"Bump hacking version","status":"MERGED","created":"2022-06-16 07:32:03.000000000","updated":"2022-06-16 17:25:18.000000000","submitted":"2022-06-16 17:23:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"846103-hacking","meta_rev_id":"bbfaf08f770218c2e3faf865c2d7c5ccecda4452","_number":846103,"virtual_id_number":846103,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-16 17:23:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-06-16 16:30:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2022-06-16 16:29:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-16 07:51:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2022-06-16 16:29:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-16 07:51:48.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-06-16 09:19:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-16 16:29:40.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-06-16 16:30:55.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"5da78b498ea20938ee55f3be0bba3b32171ee033","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-16 07:32:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f08f0460a7235f28b217b033ab793daba4b0beb3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-16 07:51:48.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6fa08576d795d60cf540bba8a1a8a0bc513d8253","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 09:19:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d59ba7dd19114c7a827cc7a0466caa33 : SUCCESS in 4m 16s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0f95820a032d4c849bfa19e020380a96 : SUCCESS in 40m 41s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5307c9ec914645da8e731106a8ccae36 : FAILURE in 14m 24s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/aea386dd3b8d4e4d8df224976d8239b9 : SUCCESS in 53m 43s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/759a184e5b3e4d0489ac2157579f74fa : SUCCESS in 41m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8c63627c23a4450780280940bbbb2948 : SUCCESS in 51m 13s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/cb6a1f1088794a8f8dc07ee504ac567e : SUCCESS in 1h 00m 57s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/40d6b6a7ba0f49c6b306c37947748230 : SUCCESS in 1h 00m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0c0a4afbe05c4904b7296798be812529 : SUCCESS in 37m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d5ba44077dec4d67aaecb1c132266091 : SUCCESS in 47m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3357b63f580e462bafa6662726844b8e : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecfd084790494ca0a5e82fc6ee6c44b1 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cef0371436d6472babe7f029b92db67c : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9888978fecb94864bd687ef75273a09b : SUCCESS in 6m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad4844f7108f453ca6e0fea31eea03c1 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/978a80f845644f4088a61b89041205a8 : SUCCESS in 8m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/91f945ae937f4c93a5e96eb5018d33e5 : SUCCESS in 2m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b6db1d89ef345be8c85a70bdcb68aa5 : SUCCESS in 4m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f5a5c3647d7845ebafb8f7ade22dd440 : SUCCESS in 5m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8059c792a0f6ffbde9e4ee1f4484b139235fa3dd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-16 16:29:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c11f4e665ee06b2c68f824bcad836ba5aa9f824f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 16:30:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22f41a74383eb8dfcb1e1640f9b829c93e759355","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-06-16 16:30:55.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"757f5c544f0f467a3a0d3d8c2b44e337beb9d2ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 17:23:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0e7c7a2c457d493e8790a2ec29b11d90 : SUCCESS in 3m 58s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/edc7a2cfaeb14127ab1c7fad5b0073c2 : SUCCESS in 32m 38s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/92e8f4d540104f77adfc36801cde0d68 : SUCCESS in 32m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f108219510c2498c9a68dc0edb0b3fc3 : SUCCESS in 30m 12s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4bb33288389e42c0932b1bf960a4db36 : SUCCESS in 52m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/545cf1307b804b8aa0ab99563b34c5e6 : SUCCESS in 35m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/467ef9dd16f24fb9a58a03f06e4e9abe : SUCCESS in 33m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c633e78d049f434aa9c6500850d3ab3f : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/392e3fba66f845ed94039ce81c66f42c : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc6699951d004b69bd27442088751607 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b3d7a2e77c34f6fa26909f3b060150d : SUCCESS in 7m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0e76838180d24a5ab023ecd4c066624e : SUCCESS in 2m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee0bd3fb6a164530a21661c19dbcf085 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"1711d05e5cea82964be500581a566a43cc0847ac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 17:23:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bbfaf08f770218c2e3faf865c2d7c5ccecda4452","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 17:25:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/9cf9a4ea5e4e4f0fa31884252660b11a : SUCCESS in 53s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1cd64c2acb74cddb30435e75cec93b1 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/228f23f792e64699b4fd244f23d2b2a4 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cdf81bb56e0cd2ee503153d31ceab24570f123fa","revisions":{"cdf81bb56e0cd2ee503153d31ceab24570f123fa":{"kind":"REWORK","_number":1,"created":"2022-06-16 07:32:03.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/03/846103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/846103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/846103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/846103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/846103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/846103/1"}}},"commit":{"parents":[{"commit":"404db698ca59a5f9254377d568bd42d7b1a5225f","subject":"Merge \"Fixed incorrect quota exception message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/404db698ca59a5f9254377d568bd42d7b1a5225f"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-06-16 07:31:18.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-06-16 07:31:18.000000000","tz":120},"subject":"Bump hacking version","message":"Bump hacking version\n\nWe can also drop some excludes that no longer affect us\n\nChange-Id: Ib3e9d87c29a612e4c77212bd3a414c53ee1d9fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cdf81bb56e0cd2ee503153d31ceab24570f123fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cdf81bb56e0cd2ee503153d31ceab24570f123fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
