)]}'
{"id":"openstack%2Fdesignate~846104","triplet_id":"openstack%2Fdesignate~master~Id03b49a3f7dc9bb9affb160099faf5bb1c67a291","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-06-18 08:25:05.000000000","reason":"Dr. Jens Harbott (frickler) replied on the change"}},"hashtags":[],"change_id":"Id03b49a3f7dc9bb9affb160099faf5bb1c67a291","subject":"DNM: Test jsonschema bump","status":"ABANDONED","created":"2022-06-16 07:32:58.000000000","updated":"2022-06-18 08:25:08.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d949c8923fc280685a38d096bef665bcaab79036","_number":846104,"virtual_id_number":846104,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-16 08:52:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-16 08:52:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-17 15:34:51.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"}],"messages":[{"id":"5f5c2e1c8856e5dc729d179bfabb2ba702e55881","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-16 07:32:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a6bded3440e1f350a3bcc5dadc1dd9696cfa940","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 08:52:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/30faaefcc5a9428d865ad2f8863cafa6 : SUCCESS in 43m 50s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6b3a26180ef542c19e3d76562ae2261c : TIMED_OUT in 1h 14m 23s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee833ac5c67c4201be83f47313427a25 : TIMED_OUT in 1h 13m 26s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8e7a7c8a1855427ba58be2398c93f02b : SUCCESS in 45m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/771feca0b0a54bb1b3f99315bd7ad2ad : SUCCESS in 44m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/77bdb9f30dfe4f5f9b43bc31c775f8d4 : SUCCESS in 1h 03m 17s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/aa58bf1f51ae42bb9436329429a4b499 : SUCCESS in 1h 01m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0527caf1ccb44b709c74cb6c2f2be500 : SUCCESS in 40m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/adc3556fbe27433a9a3b1e8645cce2e6 : SUCCESS in 45m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe0c3eaa18234af68a9e324cad23337f : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a21db46d6c944f6a81fefe8ecbbd73d4 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/722f3085e6a54fb486f742b0c62f865c : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/263dc57fe71547f58253833435f3cfd9 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18e439b989864abe96145cc04a634cbc : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3086669b6e5d43209f8265f4a218a0f4 : SUCCESS in 7m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2efa15cc04364d04b0215346c1e11cf9 : SUCCESS in 27m 59s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f3253a64f7a547908227780a4f2c1ce6 : SUCCESS in 4m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a1119069a0a8979053b7c64fb30c022b532c753","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-17 15:34:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"999df984a07ceceade8b9433e0a8242d76f2da17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 16:53:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1036177e5ca24d97bafec88ebe0d6e6c : SUCCESS in 46m 17s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7ea06361c79f402da9e5c5bd3e31ece8 : FAILURE in 19m 47s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a25209fe86a148e79172a06495832fc3 : TIMED_OUT in 1h 13m 16s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/749c19df94834392a609f619e3bbe47d : SUCCESS in 47m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fc680acebc1b4be786474c3c51e89c03 : SUCCESS in 48m 23s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9d8729c338be49d18aed4ee70873808f : SUCCESS in 1h 08m 25s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1219cd74711e4b288d095e408f74d569 : SUCCESS in 1h 06m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d70a1444b4ab4338b575b5953f4aac6b : SUCCESS in 47m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e35c434aea16463ea7b84e52d4fa3bf3 : SUCCESS in 43m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/badd335104cf4acea32772da2e52a76b : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d76273558ee49c9bcd8c50c0389eca2 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b79ce514d2047b1967c83b77196a566 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e85acc9d08ab452ba1895defb1f514f0 : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15c59bb649b548c886de1227846e893e : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bfbb7ff17594b6e991fb430793c1b45 : SUCCESS in 8m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a05f97479b0b497eb8013cfe122b89f0 : SUCCESS in 28m 01s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0a4a08dfb1fd4aeb9c4bb957ac0fb120 : SUCCESS in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b64ddc7cae32adfdee1b1910f7dc1bdabb6a2759","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-17 17:13:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b584bcc384aa1dd3073264c750d5017d03cfac8c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-18 08:25:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d949c8923fc280685a38d096bef665bcaab79036","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-18 08:25:08.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cdaa256793bccc507a4ee2d348ea4ba70c75d8ce","revisions":{"cdaa256793bccc507a4ee2d348ea4ba70c75d8ce":{"kind":"REWORK","_number":1,"created":"2022-06-16 07:32:58.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/04/846104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/04/846104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/04/846104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/04/846104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/04/846104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/04/846104/1"}}},"commit":{"parents":[{"commit":"2cb4caa363255c158f0a2996adc2e4a42573ec07","subject":"Remove invalid use of jsonschema.compat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2cb4caa363255c158f0a2996adc2e4a42573ec07"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-06-16 07:32:35.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-06-16 07:32:35.000000000","tz":120},"subject":"DNM: Test jsonschema bump","message":"DNM: Test jsonschema bump\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/845859\nChange-Id: Id03b49a3f7dc9bb9affb160099faf5bb1c67a291\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cdaa256793bccc507a4ee2d348ea4ba70c75d8ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cdaa256793bccc507a4ee2d348ea4ba70c75d8ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
