)]}'
{"id":"openstack%2Fdesignate~847432","triplet_id":"openstack%2Fdesignate~stable%2Fussuri~I782bf38a68f24a2e7133ff2afad7c96e2ae6b7f0","project":"openstack/designate","branch":"stable/ussuri","topic":"bug/1958409-stable/ussuri","attention_set":{},"removed_from_attention_set":{"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2022-07-02 04:38:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I782bf38a68f24a2e7133ff2afad7c96e2ae6b7f0","subject":"Fix zone update when adding new Bind9 target to pool.","status":"MERGED","created":"2022-06-24 04:51:35.000000000","updated":"2022-07-02 04:39:32.000000000","submitted":"2022-07-02 04:38:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847432-bug/1958409-stable/ussuri","cherry_pick_of_change":828679,"cherry_pick_of_patch_set":2,"meta_rev_id":"645288b81025f736d483a9e45049254feb8fc138","_number":847432,"virtual_id_number":847432,"owner":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-02 04:38:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-06-24 21:03:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2022-06-24 14:51:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-02 02:27:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-02 03:37:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"}],"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-24 04:51:35.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"state":"CC"},{"updated":"2022-06-24 05:46:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-24 14:51:25.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-06-24 21:03:18.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2022-07-02 02:27:24.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"0a9231a3d5caebc5e66c520faf429325591e1cef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2022-06-24 04:51:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"20fc22f0a748e913d8177359d6064e5edb6fa065","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 05:46:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fce5e5e453b64b6ab40eb095c9fc8994 : SUCCESS in 33m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c4376c801acb4dc1bcef8c79568c8df4 : SUCCESS in 28m 36s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/eb92e9795f00479e836c77e66557ea65 : SUCCESS in 48m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d317020602cd4c539420cde456330b50 : SUCCESS in 32m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6706c780d2ff44e0a8ae0f5aa962a6b5 : SUCCESS in 30m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/045c924c3ec04e199b9f9bed20461677 : SUCCESS in 9m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0b6516831164f289eed3c591e804fd0 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d1b5cadf4c04fa9934933806f75335a : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b08fded8f5474d60a6aca0fad67d5a66 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54a551d5b6644f28b2a5689cc04eca90 : SUCCESS in 11m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c8dc295a1024f5e8fd2c1fdd0d4a666 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"a1996181c848e4f143d087db4c4419a2a9b55cd3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-24 14:51:25.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"69cf81ec7d898d1e94504e39f39f5c60d033f3d3","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-06-24 21:03:18.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a821ddc921682f79b0242132f690cd35ca9c11f9","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2022-07-01 09:08:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f25cd31d00c377f07d731e4e899fd0916aaaff7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-02 02:27:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7b0ea3371f5599efe521057e44f6b833082d8321","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-02 03:37:12.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"57a725977f15c7e53a9ce875f8cefd36a915dcc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-02 03:38:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d26589a926cac09e5cfb0aec257658998acd15c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-02 04:38:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/12e2fc8a011040118d6b912a58eba8fe : SUCCESS in 33m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cd42c9d1957a41d2af7ffe9a285869de : SUCCESS in 32m 49s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f3b87cc2022a4862a2e3f8aba32f571a : SUCCESS in 59m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/adca891485b64ffd82ac3f4412894d8f : SUCCESS in 34m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/92baa3949f0a4956a324cbae8bb26969 : SUCCESS in 32m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3029827b7a31444292be09809c045b8f : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6aef57ec4794ab7a4a7b5354bc369d2 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/39bf363e8cec43fdbe89f69031310683 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9169a12df4c4b65be581d97d55e4ffb : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d2c40705d7b4781bc4b1912e429ecae : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"4c78018ec089a78df7bfabaaadb8545da1b81f74","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-02 04:38:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"645288b81025f736d483a9e45049254feb8fc138","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-02 04:39:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd8e338155264b83befec36e7a3ce89e : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/756d5a5521a5476ca1255c18f2e994d1 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5e569a3c7646042131926b090f570ca5082503e3","revisions":{"5e569a3c7646042131926b090f570ca5082503e3":{"kind":"REWORK","_number":1,"created":"2022-06-24 04:51:35.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/32/847432/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/847432/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/847432/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/847432/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/847432/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/847432/1"}}},"commit":{"parents":[{"commit":"8a84e936b2cab5badcfb817914d289bcb9a348c3","subject":"Merge \"Checks for invalid denylist regex patterns\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8a84e936b2cab5badcfb817914d289bcb9a348c3"}]}],"author":{"name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","date":"2022-02-10 09:40:35.000000000","tz":60},"committer":{"name":"Sven Kieske","email":"s.kieske@mittwald.de","date":"2022-06-24 04:51:35.000000000","tz":0},"subject":"Fix zone update when adding new Bind9 target to pool.","message":"Fix zone update when adding new Bind9 target to pool.\n\nWhen a new Bind9 is added to the pool, Designate will trigger \u0027modzone\u0027\nrndc command that will fail, unless zone is already present on the backend.\n\nThis change will first verify that the zone is present on the backend,\nbefore an update attempt. If zone is not present - zone creation will be\ntriggered.\n\nCloses-Bug: 1958409\n\nCo-authored-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\nChange-Id: I782bf38a68f24a2e7133ff2afad7c96e2ae6b7f0\n(cherry picked from commit 66cc876eadd9a4ded3afa35aed675c15c60f149d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5e569a3c7646042131926b090f570ca5082503e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5e569a3c7646042131926b090f570ca5082503e3"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
