)]}'
{"id":"openstack%2Fdesignate~848194","triplet_id":"openstack%2Fdesignate~master~I13162711b521472ec8b796ced34d0b7107aaed36","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-08-31 00:26:52.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I13162711b521472ec8b796ced34d0b7107aaed36","subject":"Add simple API versioning to the Designate API","status":"MERGED","created":"2022-06-29 22:42:50.000000000","updated":"2022-08-31 18:06:49.000000000","submitted":"2022-08-31 18:05:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848194","meta_rev_id":"ee8f49ce230da9ec5dd71740b499c6318bbbf364","_number":848194,"virtual_id_number":848194,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-31 18:05:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-07-21 22:32:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-31 00:26:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-07-22 15:52:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-31 00:26:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-29 23:41:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-21 22:32:36.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2022-07-22 15:52:30.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"6713092a1086a8f04475debdd14723a2c3415b54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-29 22:42:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dec56bd19f5c29cbbb822524b0ef83888051db1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-29 22:49:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"465fc275e056eba67f1c1c2c29e32a2afda3673a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 23:41:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a4c57df097a84f0ba9db16c4e3df4a5a : SUCCESS in 3m 43s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7331f9c386cf4cbcb7ff77619bdc2c17 : FAILURE in 31m 54s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/cf32798f0f224f9b963690dc64a7512f : FAILURE in 44m 52s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a8be9f1c2d7242cda2e6645343565ef1 : FAILURE in 38m 20s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/83ea66b2a79b4e7885fae2536609a1b3 : FAILURE in 32m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1448c409e4ea4a9abb5ca9d248e81e88 : FAILURE in 33m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/96a9fc73a5844078b85e002bbfe547c5 : FAILURE in 49m 04s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/396844c57f274d019effd55fc88c30e7 : FAILURE in 50m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/96c1d34c0460446f9cb254c9b93d1f1a : FAILURE in 42m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ce3a3645ba374972a35156a2bf915e91 : FAILURE in 37m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ef7f25045b946e5960d7d77dabf17b0 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33027da591684c0190c1ecb80074a777 : FAILURE in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/734c49c7b709413f9b2ac0172ac55a6f : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83d4405c587e4334993c42d80d72214b : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ffc2ea678d34f0aaf8aa7b725582641 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50f6966699e24ecfafd9a3aabdfc2642 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a0d53302af1464ba829e3f538614550 : SUCCESS in 3m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bbf174b4244144ca8cf4896b7ab0c99a : SUCCESS in 36m 48s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8a5f9b0bcfb84d35be964b8fd8d2b3fc : SUCCESS in 4m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"84be68aa549527f5e8e7aa94e04db6d625b59768","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-30 17:41:14.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ebd43413b5e7ecb2572727f9289c80f36423a205","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-30 18:30:54.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"c0221642c50716b24270456a288abdec3436dc97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 19:36:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f8d360219187401ab239285e0efca435 : SUCCESS in 4m 19s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5639d50ba8014f309b03f4aefcce1c0b : SUCCESS in 38m 47s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/eaabf8dda0044112866d15d29038d316 : SUCCESS in 45m 35s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/622e2983a0694919b3468a7c509ab003 : SUCCESS in 39m 18s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ce378d7d2fa34ec9a31632a0acf2e938 : SUCCESS in 39m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1991964a84114f299b00eb6d5897f4c9 : SUCCESS in 41m 06s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/aed1e8f977fe4423b9e0cd1160c7d90e : SUCCESS in 1h 02m 33s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8caa09d302454d07a918bc552fa3f0e2 : SUCCESS in 1h 00m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8a1f1ef3aafe4139b97a13b2b2d23630 : SUCCESS in 40m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cfcf09154223468c9de66fc9f1ae5b7e : SUCCESS in 37m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94f40d746e7e4bf6b0421b00d1419910 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efd2cf09af264681b6975fd229b9af1b : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1f9739e974f47ec9d8c9b5bda931f86 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c36aa09859c14106832862f58f147e2f : SUCCESS in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6851b0068d1a447f9d96c5e0ce9faea6 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d531d83c74644e5aabd7dce24f0c9269 : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f13f285f38d14194adf3eb75be070517 : SUCCESS in 4m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3aabdf1ec29440459ecf92235f5edcd9 : SUCCESS in 31m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b61a30fb5da74b7280a3dd6a0ab4620d : SUCCESS in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e468d92cc7dd3c2373960d3c614c7e49cdb8ea4a","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-07-21 22:32:36.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d0a63fafb4929ed83f5990e8b9987e6b642faabf","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-22 15:52:30.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b02ca4729d13cdeff2fcb0c9f951ecac80db81c1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-31 00:26:52.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5452b8576eef587a2d0685f9e35d3133ce7a97ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 17:09:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f11be0e10b7d685cf31e01a15ef4d9236fb12557","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 18:05:25.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c6e9e76439c45aa94b5c8fa34468e8e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4627dd5f55214736a412cec7d8b8ba1e : SUCCESS in 4m 46s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9246766bc67b4bb3874da020532f9122 : SUCCESS in 51m 07s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/28bfbef2e55844649c25697316119393 : SUCCESS in 45m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/23ae05b134134954aad4d76f11f38138 : SUCCESS in 39m 58s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0ec02e09ec8242e49efb53d1baac50f0 : SUCCESS in 49m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5204ad9d915a4957a982a1c94414a0db : SUCCESS in 35m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8853771ccc15433f9a2ed5ee2b4682fd : SUCCESS in 36m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dabc598290fa4594b961e15f8ab63040 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b49780e3cc7f4ec3aa7f385c9074261a : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/791ca6cc5d7342a397c466f5669696d5 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a447b18c90ed47bca3022e02711d3d0e : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93f2518f17164785ad694c1e40e78f6e : SUCCESS in 4m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7d8c815f77fc493f9ed1b2722e9475ba : SUCCESS in 35m 51s","accounts_in_message":[],"_revision_number":4},{"id":"440710f834bc14206193598bc666b3faaf69212b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 18:05:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"ee8f49ce230da9ec5dd71740b499c6318bbbf364","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 18:06:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14b69def04cd4e6abfd3ba2841187acd\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/c9b665abf43b4f319e4408b35dc2dd43 : SUCCESS in 42s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45e2d80df81d4227b8b8164f037fc248 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fb8905b0eda46dc9389f6d9b28d237e : SUCCESS in 57s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"39652d7a518c7b26cc878cbb23b34f94ef8a1252","revisions":{"086c94300daaa3e6b9daaffb58b54eafe33b6567":{"kind":"REWORK","_number":1,"created":"2022-06-29 22:42:50.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/94/848194/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/848194/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/848194/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/848194/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/848194/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/848194/1"}}},"commit":{"parents":[{"commit":"d05232fc075ffe6bfaf7a537334fe081bb41a65c","subject":"Merge \"Removed RPC calls from MDNS and moved them to the Worker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d05232fc075ffe6bfaf7a537334fe081bb41a65c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-29 22:39:38.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-29 22:42:49.000000000","tz":0},"subject":"Add simple API versioning to the Designate API","message":"Add simple API versioning to the Designate API\n\nAdds simple API versioning for Designate. Designate API minor releases are\nadditive to the API major revision and share the same URL path. Subsequent\nminor versions are a superset of the previous versions of the same major\nrevision.\nUpdates the version discovery document to be compliant with OpenStack\nversion discovery.\nRestructures the API reference to provide more information about the\nversioning.\n\nChange-Id: I13162711b521472ec8b796ced34d0b7107aaed36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/086c94300daaa3e6b9daaffb58b54eafe33b6567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/086c94300daaa3e6b9daaffb58b54eafe33b6567"}]},"branch":"refs/heads/master"},"fa4924d165ca2233b97acdf70184ae7a9c61a7f0":{"kind":"REWORK","_number":2,"created":"2022-06-29 22:49:00.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/94/848194/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/848194/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/848194/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/848194/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/848194/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/848194/2"}}},"commit":{"parents":[{"commit":"d05232fc075ffe6bfaf7a537334fe081bb41a65c","subject":"Merge \"Removed RPC calls from MDNS and moved them to the Worker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d05232fc075ffe6bfaf7a537334fe081bb41a65c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-29 22:39:38.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-29 22:48:54.000000000","tz":0},"subject":"Add simple API versioning to the Designate API","message":"Add simple API versioning to the Designate API\n\nAdds simple API versioning for Designate. Designate API minor releases are\nadditive to the API major revision and share the same URL path. Subsequent\nminor versions are a superset of the previous versions of the same major\nrevision.\nUpdates the version discovery document to be compliant with OpenStack\nversion discovery.\nRestructures the API reference to provide more information about the\nversioning.\n\nChange-Id: I13162711b521472ec8b796ced34d0b7107aaed36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa4924d165ca2233b97acdf70184ae7a9c61a7f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa4924d165ca2233b97acdf70184ae7a9c61a7f0"}]},"branch":"refs/heads/master"},"a2eae6c8f2fb0fe69dc47699b5cab11cb3c8757f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-06-30 17:41:14.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/94/848194/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/848194/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/848194/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/848194/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/848194/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/848194/3"}}},"commit":{"parents":[{"commit":"d05232fc075ffe6bfaf7a537334fe081bb41a65c","subject":"Merge \"Removed RPC calls from MDNS and moved them to the Worker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d05232fc075ffe6bfaf7a537334fe081bb41a65c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-29 22:39:38.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-30 17:41:14.000000000","tz":0},"subject":"Add simple API versioning to the Designate API","message":"Add simple API versioning to the Designate API\n\nAdds simple API versioning for Designate. Designate API minor releases are\nadditive to the API major revision and share the same URL path. Subsequent\nminor versions are a superset of the previous versions of the same major\nrevision.\nUpdates the version discovery document to be compliant with OpenStack\nversion discovery.\nRestructures the API reference to provide more information about the\nversioning.\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/848323\nChange-Id: I13162711b521472ec8b796ced34d0b7107aaed36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a2eae6c8f2fb0fe69dc47699b5cab11cb3c8757f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a2eae6c8f2fb0fe69dc47699b5cab11cb3c8757f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"39652d7a518c7b26cc878cbb23b34f94ef8a1252":{"kind":"REWORK","_number":4,"created":"2022-06-30 18:30:54.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/94/848194/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/848194/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/848194/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/848194/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/848194/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/848194/4"}}},"commit":{"parents":[{"commit":"d05232fc075ffe6bfaf7a537334fe081bb41a65c","subject":"Merge \"Removed RPC calls from MDNS and moved them to the Worker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d05232fc075ffe6bfaf7a537334fe081bb41a65c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-29 22:39:38.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-30 18:30:54.000000000","tz":0},"subject":"Add simple API versioning to the Designate API","message":"Add simple API versioning to the Designate API\n\nAdds simple API versioning for Designate. Designate API minor releases are\nadditive to the API major revision and share the same URL path. Subsequent\nminor versions are a superset of the previous versions of the same major\nrevision.\nUpdates the version discovery document to be compliant with OpenStack\nversion discovery.\nRestructures the API reference to provide more information about the\nversioning.\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/848323\nChange-Id: I13162711b521472ec8b796ced34d0b7107aaed36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/39652d7a518c7b26cc878cbb23b34f94ef8a1252"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/39652d7a518c7b26cc878cbb23b34f94ef8a1252"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
