)]}'
{"id":"openstack%2Fdesignate~849514","triplet_id":"openstack%2Fdesignate~stable%2Fyoga~I9b08e25a007bea598442da377dc227538f6e35f7","project":"openstack/designate","branch":"stable/yoga","topic":"bug/1926429-stable/yoga","attention_set":{},"removed_from_attention_set":{"5572":{"account":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"last_update":"2022-07-14 21:13:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9b08e25a007bea598442da377dc227538f6e35f7","subject":"Minmum TTL value is zero","status":"MERGED","created":"2022-07-12 15:10:24.000000000","updated":"2022-07-14 21:14:44.000000000","submitted":"2022-07-14 21:13:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849514-bug/1926429-stable/yoga","cherry_pick_of_change":848941,"cherry_pick_of_patch_set":1,"meta_rev_id":"e5347c8630f16d4e77826a2f11d840cf8aa4b45c","_number":849514,"virtual_id_number":849514,"owner":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-14 21:13:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-07-14 20:09:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-13 23:46:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2022-07-14 20:09:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-12 16:13:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-13 23:46:44.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-07-14 20:09:35.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"b0d81795ddd72b3120e8b16790263c8261fef848","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-07-12 15:10:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"56f78324b090bd55b5fd44437063043c8f55fc3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 16:13:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba28ae630c9a4ba08a55cc7da82daf35\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/aefebfe9ac24484fb13bbab95d0eec00 : SUCCESS in 40m 44s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/b95da131b0e84278a3acde9ccea17607 : RETRY_LIMIT in 24m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/cda84227d7534bacae9858599bd79524 : SUCCESS in 46m 21s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/7e5b22497f5b4545aa30154da1892d34 : SUCCESS in 43m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/94150f7e1e224d41ac6978861526a353 : SUCCESS in 41m 46s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5ca320c63e0f4a18864c684256545c91 : SUCCESS in 54m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1a8894384e6245938aa820cf0ea9ebc1 : SUCCESS in 36m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/28b14966af9041f5a5aebc2643bd92f1 : SUCCESS in 45m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f201b1808bea46d695f3fbc77e12c908 : SUCCESS in 6m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5180255e45b5491dad9feadbd5214b8e : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fadb0cc5a4c14fc789cbba838ea0402d : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5d579c6a5134044ab4592bd50eba0f1 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d63119a0f4a34c319880f293899f72be : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90f438c16df94f59a6bbdf5c2cc30f63 : SUCCESS in 8m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/08cb0c39498644219efef2171fba56c2 : SUCCESS in 37m 03s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b72d09146cb54310bd98d14577a2f1d1 : SUCCESS in 5m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d6a3cc1e87f6a44273e2c36379f12e5a298252e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-13 23:46:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"384eb94b74f6a37ce14b325aa1c84cf7d16dbf98","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-07-14 20:09:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1bea7774f9d6778cc5959bfc5533d21b1bd51a65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 20:10:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e7f79049dae38ae2d2197fe4aa0cee135a409f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 21:13:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/681548a3fe244f10898e50a5b1d8be61\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/93c02d248c2e4ca8b6efae1e2d7d897e : SUCCESS in 1h 01m 44s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/188d056fe72f41e78548dcff38df47a3 : SUCCESS in 49m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8bcaca58adac4b9b838b8093853c13d3 : SUCCESS in 53m 03s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bf6dabf472dd4b35b5eb23edf041b85c : SUCCESS in 57m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6487b3e8851f4038a80477302d37b798 : SUCCESS in 44m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/38828aaa8ad7435896a59840bb47ad3a : SUCCESS in 49m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a24cd916c9c3492184ae15c2aedaa6b3 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/981bd64d0179455293c47258a87a60ad : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3960570804c4f32ab88e28f909623a6 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9b52ee049204925a2a40b993af4475a : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60ea222e52954969b3d921af50997aad : SUCCESS in 9m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d522432622b34faabf330189d802d7a0 : SUCCESS in 48m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a2029c76effe62ce02ab3d5b30d2e4a6ea3d7bd2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 21:13:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e5347c8630f16d4e77826a2f11d840cf8aa4b45c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 21:14:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fe7756b1283440cad6ffe68a8d0693f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db85ff9a676641648483c07dfe8c53ca : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2dc40a15894d2c2010b7176ff6f1dff0c8621503","revisions":{"2dc40a15894d2c2010b7176ff6f1dff0c8621503":{"kind":"REWORK","_number":1,"created":"2022-07-12 15:10:24.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/14/849514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/849514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/849514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/849514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/849514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/849514/1"}}},"commit":{"parents":[{"commit":"3808102790ca8ccb559c865ddea15876b5a19939","subject":"Fix misleading release note for RBAC changes.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3808102790ca8ccb559c865ddea15876b5a19939"}]}],"author":{"name":"dkehn","email":"dekehn@gmail.com","date":"2022-07-07 10:58:04.000000000","tz":0},"committer":{"name":"Don Kehn","email":"dekehn@gmail.com","date":"2022-07-12 15:10:24.000000000","tz":0},"subject":"Minmum TTL value is zero","message":"Minmum TTL value is zero\n\nAccording to IETF https://tools.ietf.org/html/rfc2181#section-8 the\ndefinition of the ttl value is unsigned and can have a minimum\nvalue of 0. This path changes the minimum value of 1 to allow for\n0 in recordset creats and updates. Unit test have also been modified\naccordingly.\n\nCloses-Bug: #1926429\nChange-Id: I9b08e25a007bea598442da377dc227538f6e35f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2dc40a15894d2c2010b7176ff6f1dff0c8621503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2dc40a15894d2c2010b7176ff6f1dff0c8621503"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
