)]}'
{"id":"openstack%2Fdesignate~849515","triplet_id":"openstack%2Fdesignate~stable%2Fxena~I9b08e25a007bea598442da377dc227538f6e35f7","project":"openstack/designate","branch":"stable/xena","topic":"bug/1926429-stable/yoga-stable/xena","attention_set":{},"removed_from_attention_set":{"5572":{"account":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"last_update":"2022-07-14 21:17:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9b08e25a007bea598442da377dc227538f6e35f7","subject":"Minmum TTL value is zero","status":"MERGED","created":"2022-07-12 16:15:59.000000000","updated":"2022-07-14 21:18:28.000000000","submitted":"2022-07-14 21:17:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849515-bug/1926429-stable/yoga-stable/xena","cherry_pick_of_change":849514,"cherry_pick_of_patch_set":1,"meta_rev_id":"35740ce36f9b91dbcf0df5a0c21366d8bc6c31ed","_number":849515,"virtual_id_number":849515,"owner":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-14 21:17:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-07-14 20:09:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-13 23:47:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2022-07-14 20:09:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-12 17:26:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-13 23:47:10.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-07-14 20:09:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"417892001d8530e62aeaf253064273952122aa17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-07-12 16:15:59.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"9251da48f8fc517df27221b12739ead652bd91b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 17:26:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2879aedd4564063ad68818c5bd7bc54\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d6cd309cc5644f1faa8293ed4fe268ff : SUCCESS in 48m 20s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/f630c95164b84d838327f2151a12cab1 : RETRY_LIMIT in 14m 40s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f0c8b77d7c6b45f29484ae6148356c06 : SUCCESS in 45m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/abc39f1dbe60476b87baf781fb8de0e2 : SUCCESS in 43m 10s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f0dd96d945d7401d94c82a87ff95333e : SUCCESS in 1h 08m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f5409db7c8464b948eb6f7ef159e0cb6 : SUCCESS in 46m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/419aaeac2b2b4efca3028c8ceac002a8 : SUCCESS in 44m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79821ea01763467aaefa72c05fe54ee3 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7499ebb588734f1ca1268af21d63adfc : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28881100e20b4130865b257c6e0060a7 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32a165f2a01a44f7850065ebec4bf9f2 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a03c3cd06c540ad80322dffabf1943a : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f132535f53b48668ccb586463e14c2b : SUCCESS in 7m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bf0fbb25896c4ea3b748a32cac63ba4e : SUCCESS in 33m 20s","accounts_in_message":[],"_revision_number":1},{"id":"1a58f7fe5ac70986450db038cc08f390957d522f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-13 23:47:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bb67d3314eea86bd15d898f3c1c614a65e8543a9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-07-14 20:09:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1fd28eecc728d5fed8c0c3b4032cf6475a614aac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 20:10:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"925f97a5e773f02b0e76a7e3c2fe67b76565a038","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 21:17:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d6e9f015bd44db9ba236290f8f72021\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3baf4e383adb4e57ac1161bda1abab01 : SUCCESS in 41m 16s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e38fbe50b5bf4b36b64d2da6161ec039 : SUCCESS in 47m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/01929ae3e77743a0b5dbe625a6cb4437 : SUCCESS in 42m 19s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ed4ca10a0dd342b6895788e7f7015439 : SUCCESS in 1h 06m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b74afb2ae22c40fb8ab89f1b0a77d2be : SUCCESS in 48m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d7d9bbaeea6e4a02af9c7042adb31455 : SUCCESS in 44m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6908bdf82e4a4236929bec07f610ce31 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b865488a915a42608d65d369add1fdf0 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e61eb47c482431aa30754a9104d7e8b : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7cdb2752977492b9da8adad46728dd3 : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":1},{"id":"9cfd920ac4a8605f77523a723b55fcedda36a345","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 21:17:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"35740ce36f9b91dbcf0df5a0c21366d8bc6c31ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 21:18:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f449e4de0d3346d591b20ad18bc0a072\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db4e5a4efe064a6f803efcb2dbf73b93 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"410c1e8d60a6b3844050dd4d3ecc46f640356bfb","revisions":{"410c1e8d60a6b3844050dd4d3ecc46f640356bfb":{"kind":"REWORK","_number":1,"created":"2022-07-12 16:15:59.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/15/849515/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/15/849515/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/15/849515/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/15/849515/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/15/849515/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/15/849515/1"}}},"commit":{"parents":[{"commit":"2159c7a5aa4669cdb31ec8c5780c3b445f32d7cb","subject":"Fix misleading release note for RBAC changes.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2159c7a5aa4669cdb31ec8c5780c3b445f32d7cb"}]}],"author":{"name":"dkehn","email":"dekehn@gmail.com","date":"2022-07-07 10:58:04.000000000","tz":0},"committer":{"name":"Don Kehn","email":"dekehn@gmail.com","date":"2022-07-12 16:15:59.000000000","tz":0},"subject":"Minmum TTL value is zero","message":"Minmum TTL value is zero\n\nAccording to IETF https://tools.ietf.org/html/rfc2181#section-8 the\ndefinition of the ttl value is unsigned and can have a minimum\nvalue of 0. This path changes the minimum value of 1 to allow for\n0 in recordset creats and updates. Unit test have also been modified\naccordingly.\n\nCloses-Bug: #1926429\nChange-Id: I9b08e25a007bea598442da377dc227538f6e35f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/410c1e8d60a6b3844050dd4d3ecc46f640356bfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/410c1e8d60a6b3844050dd4d3ecc46f640356bfb"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
