)]}'
{"id":"openstack%2Fdesignate~849644","triplet_id":"openstack%2Fdesignate~stable%2Fwallaby~I9b08e25a007bea598442da377dc227538f6e35f7","project":"openstack/designate","branch":"stable/wallaby","topic":"bug/1926429-stable/yoga-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"5572":{"account":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"last_update":"2022-07-14 23:42:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9b08e25a007bea598442da377dc227538f6e35f7","subject":"Minmum TTL value is zero","status":"MERGED","created":"2022-07-13 14:08:08.000000000","updated":"2022-07-14 23:43:28.000000000","submitted":"2022-07-14 23:42:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849644-bug/1926429-stable/yoga-stable/xena-stable/wallaby","cherry_pick_of_change":849515,"cherry_pick_of_patch_set":1,"meta_rev_id":"1984ae3ead589bcf8224d7f1576e8c0066c7e504","_number":849644,"virtual_id_number":849644,"owner":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-14 23:42:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-07-14 20:10:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-13 23:47:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2022-07-14 20:10:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-13 15:09:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-13 23:47:15.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-07-14 20:10:08.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"095183f60cb4e9b2f9bcb4610cc0a5448894e1c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-07-13 14:08:08.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"dacd8bd20d4fe0c3abc72eccb74acadae07b2932","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 15:09:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c3131063a2543e496addf5ccb5fbb46\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d862c57bf94d4e9098a07769a71b5be6 : SUCCESS in 45m 47s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/ed70b2ee632c480e9ba6b825b4cbb653 : RETRY_LIMIT in 24m 41s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/15f1a9fec057408096000b4950f4aa93 : SUCCESS in 51m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1e0d0b228bbf403a9da9399c4dfb40f7 : SUCCESS in 53m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b42ac7c17b0c4344ad4f5473300f6905 : SUCCESS in 46m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/38376e8011d04e95be14482689080b09 : SUCCESS in 46m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d489f82042484a39877fe26c4f39fbd4 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/996d680f28074eb2b31569a160dcbe43 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/babb34633279409982fcc72df95c0d21 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c71984e8bf3f4b35bcdd4fd267cdd19d : SUCCESS in 8m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9c98b25beb04cdaadb3970bf9a237e8 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fef866617c4940b68c7b39526820d992 : SUCCESS in 11m 51s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/39a17d9b9c624c4c9b9e66b53101c179 : SUCCESS in 40m 58s","accounts_in_message":[],"_revision_number":1},{"id":"49e9318508e42a959b347e4a8d350dc75bc325fa","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-13 23:47:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e6527b229e4900839116c3b94a92e1648914fc99","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-07-14 20:10:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"343d883abeb3a7c1197c16ff40a29568b46fbd3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 20:10:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6022a5dab6cbb8de9481318180b64a9012401431","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 21:17:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/96799c1ddce3403589d27d11e00df7c3\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f284a824a592401691891326274d3a5e : SUCCESS in 50m 38s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3993078e1e0746258f922fd34e8b3945 : SUCCESS in 44m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/95a6cced77644750b31f62b6f9a7b907 : SUCCESS in 59m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fdd45af0e1df4100a525ac63e8a4e64a : FAILURE in 47m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/aac23f65c9b94ab2aa449ae36cdc4193 : SUCCESS in 53m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5af30909b2794899b5d29ff2854bdf1b : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c660093dfbcf4f3e8c52a86a0fab6606 : SUCCESS in 9m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90be32370d754194860c807ac0f6eff2 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f846c44981e34da0899bfe1fa1d26ec7 : SUCCESS in 9m 42s","accounts_in_message":[],"_revision_number":1},{"id":"9d45362358ef9ed37c9d1971ae45b34a4226fa73","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-14 21:33:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"294d39216f37a0594ab14e8eddf00af3ab57ce70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 22:46:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/043ac34ee5114ca0945e68ac51dc6ac1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/df95a795bd304d659c8a2490e3404bbe : SUCCESS in 53m 52s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/39cf6c14d6f94fe3b525fe6c9fdb21d7 : RETRY_LIMIT in 24m 37s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1056cfba2f6d4e398dc625e4816f8048 : SUCCESS in 47m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1919e4384d584dd986bbeed69c3b4548 : SUCCESS in 52m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/176832145ca2457ebe8cba885f515feb : SUCCESS in 51m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2fd80bbdf6e24171a26272bdbb46b13e : SUCCESS in 50m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e36dbf65ec9498f85d48f66e548a31f : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b498c1b6fbc84e5d99d632ed060701c4 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f62c47e982ce4b88802a2637949ec7f7 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffb10f10368e4d2bb07da55953f2848f : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfd80c3087ce464cbe649c7f0b2bfb46 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/272fcb2ee04a40a19844f433e5e2a9ea : SUCCESS in 6m 43s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/85504243dae04f87a4c1a913626293de : SUCCESS in 34m 57s","accounts_in_message":[],"_revision_number":1},{"id":"add295d9d81c4d6269a67d5515f7461819e08a0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 22:46:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c4aa3144ab5db59543270b59556866b0bcc320e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 23:42:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04d3eb9bab9847e9b39d607300338582\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/069f4b9088644c6ab78f80f7a6e7474c : SUCCESS in 45m 35s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8338ae27aa4040eb93fd2453e523de77 : SUCCESS in 45m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1840971899ce47f3805731950a5085b8 : SUCCESS in 45m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e1d70bb16bad42a9813a5f8fdc8242bd : SUCCESS in 54m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9d2c59fefabd414eb0b5b25a3e9fa46e : SUCCESS in 47m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39e56ddd3b6e42e4976369a13e17e9bf : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84b8dfe9dd8e497da232a2198cc5e8cb : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9779e3adbf243a29d9143a5eb27e385 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b086dae48b3746a8a15901a105bc330d : SUCCESS in 8m 55s","accounts_in_message":[],"_revision_number":1},{"id":"602111f1c0acec245f2a494a017afb6ac372b605","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 23:42:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1984ae3ead589bcf8224d7f1576e8c0066c7e504","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 23:43:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b23c5f291ddf4101aa0e58be76b0a3dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f85716abf7aa4bdd9d6c24ba27ae462e : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8cef906600117b71903b54c4d7c1e072902f7a46","revisions":{"8cef906600117b71903b54c4d7c1e072902f7a46":{"kind":"REWORK","_number":1,"created":"2022-07-13 14:08:08.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/44/849644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/849644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/849644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/849644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/849644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/849644/1"}}},"commit":{"parents":[{"commit":"f2557474951e96677939dc768c98a02543380b72","subject":"Fix misleading release note for RBAC changes.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f2557474951e96677939dc768c98a02543380b72"}]}],"author":{"name":"dkehn","email":"dekehn@gmail.com","date":"2022-07-07 10:58:04.000000000","tz":0},"committer":{"name":"Don Kehn","email":"dekehn@gmail.com","date":"2022-07-13 14:08:08.000000000","tz":0},"subject":"Minmum TTL value is zero","message":"Minmum TTL value is zero\n\nAccording to IETF https://tools.ietf.org/html/rfc2181#section-8 the\ndefinition of the ttl value is unsigned and can have a minimum\nvalue of 0. This path changes the minimum value of 1 to allow for\n0 in recordset creats and updates. Unit test have also been modified\naccordingly.\n\nCloses-Bug: #1926429\nChange-Id: I9b08e25a007bea598442da377dc227538f6e35f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8cef906600117b71903b54c4d7c1e072902f7a46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8cef906600117b71903b54c4d7c1e072902f7a46"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
