)]}'
{"id":"openstack%2Fdesignate~849831","triplet_id":"openstack%2Fdesignate~master~I3d54a47f73929377122249c4d971e70e553725a6","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-07-21 19:16:47.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I3d54a47f73929377122249c4d971e70e553725a6","subject":"Cleaned up and fixed record objects and tests","status":"MERGED","created":"2022-07-15 00:06:59.000000000","updated":"2022-07-21 20:39:59.000000000","submitted":"2022-07-21 20:38:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849831","meta_rev_id":"968d1d76308f4ed60841959ef9fef09e8a351bb2","_number":849831,"virtual_id_number":849831,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-21 20:38:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-07-21 19:12:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-21 14:53:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-07-21 19:16:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 01:29:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-20 22:15:27.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-07-21 19:12:34.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"08828dd7f8e2b78286f6ea6daab4f88f75b98a19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-15 00:06:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b094d568d43e1ddd5c5c2ca69f7c28c94cbe358","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-15 00:08:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"17c94ed275104ff1d236c481b8a51a947c0baaf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-15 00:09:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"431032d5a381d291bdc15021f4bb97ddb1c1e8d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 01:29:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ce14e3f17c1430f8600859a24f1d8ba\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7130e760075d4b859f673639324ff7b7 : SUCCESS in 53m 07s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/aa03fbc319f4434db08fb856a9134b01 : SUCCESS in 1h 10m 29s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/61d97111109346f8a0c224057192225c : SUCCESS in 57m 43s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ea442e73b0e94d25a307497fcb683596 : SUCCESS in 46m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/85aa71bb53474b988ece6f129ff1ce18 : SUCCESS in 51m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8f21925a1b0c445589d0ec45db5dbca6 : SUCCESS in 1h 13m 28s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0f09dfaae46843f1ae2528b862e05223 : SUCCESS in 1h 10m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/935d55e3c2a54713ba1893e2ed3e56bd : SUCCESS in 46m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a42e0504de5d48fb9cb171205dfbdab7 : SUCCESS in 50m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f451d00d6c64e32846b235ce5ffa477 : FAILURE in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7db2e1031e5541e992b97184e91d7615 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b62efedc873148d2a5944f3be1ae4d43 : FAILURE in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b2ab5700b2440eb9caffe4711f6c2c7 : FAILURE in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a79a7e2967fc41e6994ec8eace06667d : FAILURE in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71da1ae9ecc34124b9483d2e2f9a09ed : SUCCESS in 8m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ce801a71b112410295415770a8ff3f2c : SUCCESS in 37m 17s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e648ccacd0d645e2998fd11c6cf9af4c : FAILURE in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d6cb746931b2eac43a337411bc3429f75a2ec771","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-15 01:43:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"05ee66a8b753c17b9bbacb8b08e69eea4cc6c437","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-15 01:50:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c7a7be5a24c91025cf7b2a190506a11a3b799f01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-15 01:52:52.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"62083c5956e61b15923e25948f423cd6c17e319e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-15 01:59:10.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"57c7d8719c62e5c30e5f613effc0c57ebf1b8b9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 03:18:29.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42da9ead9f5f4a1d8c9bd6300195d764\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7adc1ec3525448f58b384a43f6c555bd : SUCCESS in 52m 40s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7a194123845f4b0cb176667473c3a0d4 : SUCCESS in 1h 12m 22s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c714618b94cc430bb33141be939cae19 : SUCCESS in 1h 04m 26s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/fbf88b6f6b1d40fb90bda6fc30afeb56 : SUCCESS in 54m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2a08603b269d4575888f7482bc7eb76e : SUCCESS in 50m 48s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ea390aabf1e646c8bcd4455009d408c5 : SUCCESS in 1h 12m 57s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d14bb9232f004abd85ef8eafa09fecb3 : SUCCESS in 1h 12m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/359942a3a99a4098b8b48bd71f3460db : SUCCESS in 56m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/71230536640d4b70a5f431447316576a : SUCCESS in 55m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94e178040f5245aaa2151f4614ceeeb6 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae8f8c45e5ae4b0aa8ea0538a110387b : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a5e610b44c242818e5ad48183455fdc : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3a685c035f74fc5bc8d02ace473a50c : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5268ca704e854d46bf09f85f5b72b8a4 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6157c3dbb8bc4fd9925ca9d75682241b : SUCCESS in 8m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/78feae4133974f9f9756124ae38c9c6f : SUCCESS in 33m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/57e33be3717b42378221e235d0fc60a9 : SUCCESS in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"05ea1537e455c281ea677787614bed9dedc2dedb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-19 01:23:19.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"925eb619dbcb2a64d7467abf7d3e75c083b18aa0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-19 01:24:21.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"026cf689055ceb9ef5c3f6f41f0bacd165c25e28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 02:22:19.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1a3705ef1dc42a1a20ed1e5e9fce26f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dd92a121e2694fe7ab0cf34ba5c30759 : SUCCESS in 47m 09s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b438dfd9380443618c86a512f85f6592 : SUCCESS in 51m 56s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8a35e87371c411187516017c3c7805a : SUCCESS in 43m 43s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/436867d360d04a71a66213474545af39 : SUCCESS in 43m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1ef4161869624130a66ea9f09aa94a9e : SUCCESS in 35m 43s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/71248d3cc3ed4447afdc5c79dd0f154d : SUCCESS in 56m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8bfb1018e3e04b9ea964d247c94f58f1 : SUCCESS in 54m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0e5848c5ecfa4d5e9b5b81b49d3ace80 : FAILURE in 41m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/241638234b5c4e77989abbf87e994d76 : SUCCESS in 38m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/592e8a1abc2f4edc8b7bf0d2a1ffbaeb : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b29db41ade1b4d87807dfafa72157220 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffa04dcdec6e4db9ba6a284864cb5245 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4310e69b9a524c278e0ba22f51276288 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec69d989d54a4826b78a543138d03730 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/009a822b87254b9bb9805370f10a9956 : SUCCESS in 7m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6079f9edaacc44f78959ca7b6bdf2437 : SUCCESS in 35m 23s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/25d31b1ab2b04598abba93e37f0350fd : SUCCESS in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e50b1decf7a27646fe2996a20fc021ffd9ae4739","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-19 02:58:17.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"a6b2d3f584498ce39c459be33a089078114341fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 04:30:07.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/924dc28e07e24b49aac66f1af337dc99\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f5eeec85e47b4f8db1081f972c0f56e6 : SUCCESS in 44m 30s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/eb66a473c7784ebb9ed4ae96a4d1eb43 : SUCCESS in 1h 05m 06s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e910acad27da4a57b02187c298657daf : SUCCESS in 1h 07m 50s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/a741aa5286b14c65bdff505e7fb105f5 : SUCCESS in 50m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/824f49024fc149a8b6d10ee1813240f6 : SUCCESS in 50m 12s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d0b1226948064daca5ff003fd331ed0d : SUCCESS in 1h 16m 25s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/77bda30c0a76475f94d16baa6d7cf6d7 : SUCCESS in 1h 07m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d08b7c98ff764f89bea7175d39b477b4 : SUCCESS in 38m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ee3fdd4d6a294b1eb5215fce75cf767e : SUCCESS in 40m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f603c8a182447c5b1166c09f679d2fe : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbd1525be6cb448ab120a604fe157576 : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/705aa268d5fd44b9b541acf7c3eda856 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10322668a06244259d509e9e6a2b0567 : SUCCESS in 7m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f3d61f497974100ae76876a88edc151 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ffe8155cda346dcb837b5a3e044ef20 : SUCCESS in 9m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7fdabd385de2486581b4a58f28f12225 : SUCCESS in 42m 18s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5510b2af3d0e43e893b62e4eabd2a956 : SUCCESS in 6m 32s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"fe45e6e2f93de298c903a94541599c132f983d48","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-07-20 22:15:27.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2350540b9454c399c2ee4b55bfa2f6030d813742","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-21 04:25:50.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"6f9e36f924bfdde712db15c5271b9785e9d25c75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-21 04:39:56.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"6c5a7e61095eaa324bf3a03eb06ac4cabf8e07ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 05:56:23.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efa0f4cb4acd42859856db0c0614f7a0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3d9c750f33834c318206f09f1f82b746 : SUCCESS in 49m 39s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/79c480a5952247f3afa4ba5326ccf2a4 : SUCCESS in 1h 07m 45s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f7b86a4b9da461ebce1bfb8b8496883 : SUCCESS in 59m 51s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/434d88184d8343cab1de15e42f05d08a : SUCCESS in 55m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7111d33b083d4d248c87c43e71d9a970 : SUCCESS in 46m 56s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d116de9b566144eba342e4a556dbd2c6 : SUCCESS in 1h 14m 00s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3387629c1411439eb67cd186131c138f : SUCCESS in 1h 07m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/797e8a34dfbd482ca5bdfb3045ea4692 : SUCCESS in 54m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5153af3ab9ef439ebd280c66ab2bf4a4 : SUCCESS in 49m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9671a4481c5419dbb30b7544795c3dd : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20e6e9d683b04b7a885e5cc16c9484ea : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3cc53c0fbab54ffc83b8033d2cf9af4c : SUCCESS in 9m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e870532be33463d86495e13662afbac : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91b5e72f96c2410b94f876c200a7f69e : SUCCESS in 9m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6ac339f9df948f59532c423030a627c : SUCCESS in 8m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0c95b46868714eba8039fc0ec2538fe7 : SUCCESS in 51m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/74c310d0f7ac4b5bbd9db5d642268b40 : SUCCESS in 4m 18s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"50b0a2cc6de5c9a915a5ae8a85e4d031ccf833a6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-07-21 14:53:27.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"02a1ceeac57ae59e2116b6e288c81c47945f780c","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-07-21 19:12:34.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"58c77e72ac93e6f7fa768bab1f4c0d9b45b7c0b8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-21 19:16:47.000000000","message":"Patch Set 12: Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"fcb8ef85c4d910f448045c023a6cfdabe7b8cb54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 19:18:02.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"42a50a86a2351a84edbb9168dac3e56f184bc93e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 20:38:51.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9a6854f32ed45c6ad2aedbcce0503b7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2247f774b668476fb8645262eac33325 : SUCCESS in 54m 46s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/68a8f8886e7a47c6829d97e9c653f238 : SUCCESS in 55m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fbaefbb12e644663a99609129d20a13a : SUCCESS in 49m 59s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5a031131a28c4f17894f22b73260f5d7 : SUCCESS in 1h 14m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7c4b9a7edec24cc6bfe9c8843e6bcd11 : SUCCESS in 51m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c9b4c1045d7e4f00853c186a12921cb7 : SUCCESS in 49m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28b49fd13d504eaba03a97b68c352c01 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc13f8c15ba9447584299b710a5d9db7 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17e906eefa414ca391a47f0258a7b732 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae9cfda9f5c94e5c8699895c79d1853c : SUCCESS in 11m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f0ac3afefb714d01a74617c5517da765 : SUCCESS in 43m 03s","accounts_in_message":[],"_revision_number":12},{"id":"a888702581919fd32046af805c9834b51b3205fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 20:38:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"968d1d76308f4ed60841959ef9fef09e8a351bb2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 20:39:59.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9212267682df4c78ada86826af86ecf1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bc262ad57af48fe9aed16478a1659dc : SUCCESS in 49s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"d10056178ae13c25289f8cdd41c48062da2aab62","revisions":{"297e4dd4ed40f3b050c06bf695ae860caac8752f":{"kind":"REWORK","_number":1,"created":"2022-07-15 00:06:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/31/849831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/849831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/849831/1"}}},"commit":{"parents":[{"commit":"806bbb17c9814a65a5e001032ad5a00435bf1478","subject":"Merge \"Added additional wsgi app testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/806bbb17c9814a65a5e001032ad5a00435bf1478"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-15 00:06:09.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-15 00:06:09.000000000","tz":-420},"subject":"Fixed A, SPF and TXT record unit tests","message":"Fixed A, SPF and TXT record unit tests\n\nChange-Id: I3d54a47f73929377122249c4d971e70e553725a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/297e4dd4ed40f3b050c06bf695ae860caac8752f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/297e4dd4ed40f3b050c06bf695ae860caac8752f"}]},"branch":"refs/heads/master"},"863e96ce9f67b1da856f606c15ad92e33d9f2318":{"kind":"REWORK","_number":2,"created":"2022-07-15 00:08:49.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/31/849831/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/849831/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/849831/2"}}},"commit":{"parents":[{"commit":"806bbb17c9814a65a5e001032ad5a00435bf1478","subject":"Merge \"Added additional wsgi app testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/806bbb17c9814a65a5e001032ad5a00435bf1478"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-15 00:06:09.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-15 00:08:43.000000000","tz":-420},"subject":"Fixed A, SPF and TXT record unit tests","message":"Fixed A, SPF and TXT record unit tests\n\nChange-Id: I3d54a47f73929377122249c4d971e70e553725a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/863e96ce9f67b1da856f606c15ad92e33d9f2318"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/863e96ce9f67b1da856f606c15ad92e33d9f2318"}]},"branch":"refs/heads/master"},"04717c47189164b68d936a02accd496f2cbe67ec":{"kind":"REWORK","_number":3,"created":"2022-07-15 00:09:52.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/31/849831/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/849831/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/849831/3"}}},"commit":{"parents":[{"commit":"806bbb17c9814a65a5e001032ad5a00435bf1478","subject":"Merge \"Added additional wsgi app testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/806bbb17c9814a65a5e001032ad5a00435bf1478"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-15 00:06:09.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-15 00:09:47.000000000","tz":-420},"subject":"Fixed A, SPF and TXT record unit tests","message":"Fixed A, SPF and TXT record unit tests\n\nChange-Id: I3d54a47f73929377122249c4d971e70e553725a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/04717c47189164b68d936a02accd496f2cbe67ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/04717c47189164b68d936a02accd496f2cbe67ec"}]},"branch":"refs/heads/master"},"4b391747f8b4bcc1dafa583ea60492eaaca3d631":{"kind":"REWORK","_number":4,"created":"2022-07-15 01:43:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/31/849831/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/849831/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/849831/4"}}},"commit":{"parents":[{"commit":"806bbb17c9814a65a5e001032ad5a00435bf1478","subject":"Merge \"Added additional wsgi app testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/806bbb17c9814a65a5e001032ad5a00435bf1478"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-15 00:06:09.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-15 01:40:52.000000000","tz":-420},"subject":"Cleaned up and fixed record objects and tests","message":"Cleaned up and fixed record objects and tests\n\n- Cleaned up objects to use consistent exceptions.\n- Fixed all broken tests.\n- Re-named various objects tests to better match.\n\nChange-Id: I3d54a47f73929377122249c4d971e70e553725a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b391747f8b4bcc1dafa583ea60492eaaca3d631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b391747f8b4bcc1dafa583ea60492eaaca3d631"}]},"branch":"refs/heads/master"},"a9f778779f049425b3aa8eda4a63110914af8037":{"kind":"REWORK","_number":5,"created":"2022-07-15 01:50:04.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/31/849831/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/849831/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/849831/5"}}},"commit":{"parents":[{"commit":"806bbb17c9814a65a5e001032ad5a00435bf1478","subject":"Merge \"Added additional wsgi app testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/806bbb17c9814a65a5e001032ad5a00435bf1478"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-15 00:06:09.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-15 01:49:51.000000000","tz":-420},"subject":"Cleaned up and fixed record objects and tests","message":"Cleaned up and fixed record objects and tests\n\n- Cleaned up objects to use consistent exceptions.\n- Fixed all broken tests.\n- Re-named various objects tests to better match.\n\nChange-Id: I3d54a47f73929377122249c4d971e70e553725a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a9f778779f049425b3aa8eda4a63110914af8037"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a9f778779f049425b3aa8eda4a63110914af8037"}]},"branch":"refs/heads/master"},"79c03f82f9f778ddfac160f65b11a03ef7e940aa":{"kind":"REWORK","_number":6,"created":"2022-07-15 01:52:52.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/31/849831/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/849831/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/849831/6"}}},"commit":{"parents":[{"commit":"806bbb17c9814a65a5e001032ad5a00435bf1478","subject":"Merge \"Added additional wsgi app testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/806bbb17c9814a65a5e001032ad5a00435bf1478"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-15 00:06:09.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-15 01:52:46.000000000","tz":-420},"subject":"Cleaned up and fixed record objects and tests","message":"Cleaned up and fixed record objects and tests\n\n- Cleaned up objects to use consistent exceptions.\n- Fixed all broken tests.\n- Re-named various objects tests to better match.\n\nChange-Id: I3d54a47f73929377122249c4d971e70e553725a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/79c03f82f9f778ddfac160f65b11a03ef7e940aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/79c03f82f9f778ddfac160f65b11a03ef7e940aa"}]},"branch":"refs/heads/master"},"a2d7b744cfd8c8b2a1f2c5e18daa012a5b4e1286":{"kind":"REWORK","_number":7,"created":"2022-07-15 01:59:10.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/31/849831/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/849831/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/849831/7"}}},"commit":{"parents":[{"commit":"806bbb17c9814a65a5e001032ad5a00435bf1478","subject":"Merge \"Added additional wsgi app testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/806bbb17c9814a65a5e001032ad5a00435bf1478"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-15 00:06:09.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-15 01:59:02.000000000","tz":-420},"subject":"Cleaned up and fixed record objects and tests","message":"Cleaned up and fixed record objects and tests\n\n- Cleaned up objects to use consistent exceptions.\n- Fixed all broken tests.\n- Re-named various objects tests.\n\nChange-Id: I3d54a47f73929377122249c4d971e70e553725a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a2d7b744cfd8c8b2a1f2c5e18daa012a5b4e1286"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a2d7b744cfd8c8b2a1f2c5e18daa012a5b4e1286"}]},"branch":"refs/heads/master"},"9df839192e0337f274aae158291e9aca5999e72c":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2022-07-19 01:23:19.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/31/849831/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/849831/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/849831/8"}}},"commit":{"parents":[{"commit":"806bbb17c9814a65a5e001032ad5a00435bf1478","subject":"Merge \"Added additional wsgi app testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/806bbb17c9814a65a5e001032ad5a00435bf1478"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-15 00:06:09.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 01:23:19.000000000","tz":0},"subject":"Cleaned up and fixed record objects and tests","message":"Cleaned up and fixed record objects and tests\n\nWe are currently raising the wrong exceptions for\na few record types. This can cause only part of the\nfailures to get raised to the user, since the first\nexception will fail the process. Ideally we want to\ninform the user of all invalid records.\n\n- Cleaned up objects to use consistent exceptions.\n- Fixed all broken tests.\n- Re-named various objects tests.\n\nChange-Id: I3d54a47f73929377122249c4d971e70e553725a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9df839192e0337f274aae158291e9aca5999e72c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9df839192e0337f274aae158291e9aca5999e72c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e389b44b55604f7e98e2e0e2b233934914aa6788":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2022-07-19 01:24:21.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/31/849831/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/849831/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/849831/9"}}},"commit":{"parents":[{"commit":"806bbb17c9814a65a5e001032ad5a00435bf1478","subject":"Merge \"Added additional wsgi app testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/806bbb17c9814a65a5e001032ad5a00435bf1478"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-15 00:06:09.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 01:24:21.000000000","tz":0},"subject":"Cleaned up and fixed record objects and tests","message":"Cleaned up and fixed record objects and tests\n\nWe are currently raising the wrong exceptions for a few record\ntypes. This can cause only part of the failures to get raised\nto the user, since the first exception will fail the process.\nIdeally we want to inform the user of all invalid records.\n\nThis patch fixes this and cleans up some of code as well.\n- Cleaned up objects to use consistent exceptions.\n- Fixed all broken tests.\n- Re-named various objects tests.\n\nChange-Id: I3d54a47f73929377122249c4d971e70e553725a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e389b44b55604f7e98e2e0e2b233934914aa6788"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e389b44b55604f7e98e2e0e2b233934914aa6788"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a3803a3fad4da65ca4df2d88272e97100440e53f":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2022-07-19 02:58:17.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/31/849831/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/849831/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/849831/10"}}},"commit":{"parents":[{"commit":"8abb3117afa5347e100887577eb29578371aac7e","subject":"Cleaned up adapter code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8abb3117afa5347e100887577eb29578371aac7e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-15 00:06:09.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 02:58:17.000000000","tz":0},"subject":"Cleaned up and fixed record objects and tests","message":"Cleaned up and fixed record objects and tests\n\nWe are currently raising the wrong exceptions for a few record\ntypes. This can cause only part of the failures to get raised\nto the user, since the first exception will fail the process.\nIdeally we want to inform the user of all invalid records.\n\nThis patch fixes this and cleans up some of code as well.\n- Cleaned up objects to use consistent exceptions.\n- Fixed all broken tests.\n- Re-named various objects tests.\n\nChange-Id: I3d54a47f73929377122249c4d971e70e553725a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a3803a3fad4da65ca4df2d88272e97100440e53f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a3803a3fad4da65ca4df2d88272e97100440e53f"}]},"branch":"refs/heads/master","description":"Rebase"},"a18213e327d881cfaa3e456f062886bdd85d9d4d":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-07-21 04:25:50.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/31/849831/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/849831/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/849831/11"}}},"commit":{"parents":[{"commit":"3423a3e6568c1755405d4dc31eed1121769eb7a2","subject":"Merge \"Validate matching double quotes in TXT recordsets.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3423a3e6568c1755405d4dc31eed1121769eb7a2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-15 00:06:09.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-21 04:25:50.000000000","tz":0},"subject":"Cleaned up and fixed record objects and tests","message":"Cleaned up and fixed record objects and tests\n\nWe are currently raising the wrong exceptions for a few record\ntypes. This can cause only part of the failures to get raised\nto the user, since the first exception will fail the process.\nIdeally we want to inform the user of all invalid records.\n\nThis patch fixes this and cleans up some of code as well.\n- Cleaned up objects to use consistent exceptions.\n- Fixed all broken tests.\n- Re-named various objects tests.\n\nChange-Id: I3d54a47f73929377122249c4d971e70e553725a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a18213e327d881cfaa3e456f062886bdd85d9d4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a18213e327d881cfaa3e456f062886bdd85d9d4d"}]},"branch":"refs/heads/master","description":"Rebase"},"d10056178ae13c25289f8cdd41c48062da2aab62":{"kind":"REWORK","_number":12,"created":"2022-07-21 04:39:56.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/31/849831/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/849831/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/849831/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/849831/12"}}},"commit":{"parents":[{"commit":"3423a3e6568c1755405d4dc31eed1121769eb7a2","subject":"Merge \"Validate matching double quotes in TXT recordsets.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3423a3e6568c1755405d4dc31eed1121769eb7a2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-15 00:06:09.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-21 04:39:45.000000000","tz":-420},"subject":"Cleaned up and fixed record objects and tests","message":"Cleaned up and fixed record objects and tests\n\nWe are currently raising the wrong exceptions for a few record\ntypes. This can cause only part of the failures to get raised\nto the user, since the first exception will fail the process.\nIdeally we want to inform the user of all invalid records.\n\nThis patch fixes this and cleans up some of code as well.\n- Cleaned up objects to use consistent exceptions.\n- Fixed all broken tests.\n- Re-named various objects tests.\n\nChange-Id: I3d54a47f73929377122249c4d971e70e553725a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d10056178ae13c25289f8cdd41c48062da2aab62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d10056178ae13c25289f8cdd41c48062da2aab62"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
