)]}'
{"id":"openstack%2Fdesignate~849847","triplet_id":"openstack%2Fdesignate~master~Idfd6c959cf943c5aaf95746f16070000bf82ca21","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-08-10 06:00:23.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Idfd6c959cf943c5aaf95746f16070000bf82ca21","subject":"Fixed issues with __repr__ and __str__ on objects","status":"MERGED","created":"2022-07-19 04:20:57.000000000","updated":"2022-08-10 07:06:50.000000000","submitted":"2022-08-10 07:05:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849847","meta_rev_id":"510a15befd324118b83422f0800bd709fe7dcc32","_number":849847,"virtual_id_number":849847,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-10 07:05:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-09 20:41:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-08-10 06:00:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-19 05:34:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-20 19:48:39.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2022-08-09 20:41:07.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"8aef53d9a25e97ff4829ed99275fb47415aa322e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-19 04:20:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f909390dca6e81c6295a77aeb2ee681083f5109","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 05:34:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5085f0943854e18af3385e7e0279364\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cd080e8978ac41208353e245e4c5b64b : SUCCESS in 43m 47s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0d3356bb4cd449adaa2f754d796e728c : SUCCESS in 1h 03m 51s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a5a65fbe9274933b56ff08cad8f00c7 : SUCCESS in 1h 00m 33s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/73c7572194e24ad0890245cb75b69600 : SUCCESS in 45m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f39a4f8d2e53470b833c48e57bb3a32f : SUCCESS in 45m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0be4961a649748b9bc86a8fc4dd7bf34 : SUCCESS in 1h 01m 41s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/69c4aa31575b435b9ef4879ccfbd0f88 : SUCCESS in 1h 08m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d09f891809d0445e8921d8d3e62fb21a : SUCCESS in 41m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6229f1d8469544e6952e6fa6a9c47e62 : SUCCESS in 46m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea74ce44af144f4497b3efbb02cbbaa5 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4806569d6b4547008212b62e7743a6b8 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f391879197e422da50f93503fb1b763 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19262dda5e5b480c92bd6ebeae14327d : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4585a94d635340ddaba97004f68deb50 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6f9e377b1f341319385dcf6a3153520 : SUCCESS in 6m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e76fb5f93a104815944096c09e222b12 : SUCCESS in 31m 35s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8b698a9d9a69431fb432ab90af34f72e : SUCCESS in 4m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06277912678a0e82b91d6a5df274e647248b6100","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-20 01:31:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"680c1f5c0aff9ed56299388b40f0da1cc8dc548d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-20 01:37:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3c6dcc6e06a99fdb3e7167492751fe212e2cb1c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-20 02:13:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"eb035e924f6052815c24581efa0be8cab512698a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 03:37:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fb0a2176c044bc7894951c6dcef43a5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9dd4895d55634efeb5998cf0945acbbd : SUCCESS in 51m 26s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/077daced95bd463f98298c4309b18e44 : SUCCESS in 1h 01m 49s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c058625b9d3540be998e6efee3d3aaf9 : SUCCESS in 1h 03m 38s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3a341c30d7d9473ea44f90e2259d5222 : SUCCESS in 52m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a1d5a2c6156c42d7a97ede36caf03b17 : SUCCESS in 49m 29s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7789c02ec1c14527b7f5b782287a6be7 : SUCCESS in 1h 10m 17s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4e2aaad629b84ed096a88425bae2da72 : SUCCESS in 1h 08m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b49b4a981c5b465aaad6013d0e9cb6f2 : SUCCESS in 42m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ddb163596c4d481885f17a8cadd32459 : SUCCESS in 55m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ddddeeb015b4414b1389a9c70fbaf63 : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20ff79f4d08141b599e28b43c621fe13 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0143be92b7c946f680bc4b969a31d622 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d21316eda6c45348c2a495924509279 : SUCCESS in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d6772a005bc457894de4602ca466366 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14a0a71797674f08aab381e894774fa0 : SUCCESS in 9m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8b1855dd97bd4701a0b8364f9e1277ae : SUCCESS in 37m 40s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/59cbc41e98504a8ebb03855e9810185f : SUCCESS in 5m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ca53134b9bf69408739f39bf28920f2b9804ede2","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-07-20 19:48:39.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"eab8fd7250bce4d722fbe0090220b90a34c9c595","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-21 04:28:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2ffa5ddf8a6e6aa6e8ed329d9ac88555a32e1afd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-21 04:40:56.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"a69a3c6a430744afac39771dc15338f34a712e33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 05:50:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ee0784480ee4a029ea17f18969a8b1e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/da1b7b07ef7e44c6b71e5e8bb603dc9a : SUCCESS in 48m 49s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ed67e285494d4053ad7e621cf2f729fa : SUCCESS in 1h 00m 34s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/aaa33c29481d4b5692e2d75933daa4b0 : SUCCESS in 55m 04s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/79c18851be0845f9af086816c3dfadd5 : SUCCESS in 49m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/42a9ce07d56b42b58607414af3add13e : SUCCESS in 45m 43s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/444da27cb5644440990d9849a9bb4b6e : SUCCESS in 1h 07m 39s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fe96786f7d154701b42bc6cfb6e502a2 : SUCCESS in 1h 06m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/132e3a66fb5c409191ed3ec087a8f295 : SUCCESS in 40m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/df79ce23b01e4d089b5306f596bffa3b : SUCCESS in 48m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac9fb2a56e0847e5b766bc3dd81d95f0 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e8344a74afe4a5dae7fc0273c9d8fdb : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e825d19a352b48558fa6c8cc42545425 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e2e14f6e74b4d57b26dcaf40f0e0166 : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ebb3d4bd5d6c4f4ead471dd082016f7c : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65d0623e87264d82a87678ed77eff000 : SUCCESS in 9m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2fa974cfaf7146d099c52ace3a389f90 : SUCCESS in 43m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/83aea0ec8d5443d3aae98b3243ca8bfb : SUCCESS in 6m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d92d19f942d18bc1891cc15acfa261acbd2b47ba","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-07-21 20:45:09.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a80f51fdd086f588eb599a31bf55aee902db08bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-03 03:40:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f88b8a3c93e6b11f0e572e1b0cd28055236eb313","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 04:58:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88b0d3e798d1454a8edcb212bd811860\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b5009c2210ae4824bc4194d40929ddcd : SUCCESS in 53m 40s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/df976857950d463c8958093bf5a2bcb5 : SUCCESS in 1h 11m 27s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4cd7b9384694f79950ce7c59c59066e : SUCCESS in 59m 34s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1a8c3e4b11554866a0a2e64c27607275 : SUCCESS in 50m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3120f8b82153482381394fb6f365d13d : SUCCESS in 48m 35s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ef339a8fece04155b91cc2b697e985e4 : SUCCESS in 1h 06m 41s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5c9ec8f5333445228d31bdd251ac8d03 : SUCCESS in 1h 05m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c90b1bb553f64901af081b6d8ba6d545 : SUCCESS in 49m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/553be54ba7274b109f228abd13b9898e : SUCCESS in 45m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b9b3e8235744e3eb4afc8e96c472ee5 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b942b1c479d4864b51257804b4290a1 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a789751ee4f1464492a01f97c90bfac6 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b99b0d42d5d44f9a22ebeef074faef3 : SUCCESS in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a192489230724ff39f405626a522fe65 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c38d31b05e474488b7b4cd03a9068f7e : SUCCESS in 9m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ecbc2274ce18445aaca0f291b227fef8 : SUCCESS in 34m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/37d88d19ea5e48578b40a5db194a1169 : SUCCESS in 5m 04s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"53d6bb4f26295a1de20e98b4915a3849cceeeaf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-04 02:17:43.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"627de3d64426f9011c5e3e5ae1af283bddd3379a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 03:39:49.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad017482d92c434e8f09b1dee78a43d3\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6f8ec00e372f4148a682926a28a0ffe0 : SUCCESS in 50m 10s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/63bbf513737a44e285610f644794f650 : SUCCESS in 1h 07m 57s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/be242eb535754423abc917199e322ba1 : SUCCESS in 57m 35s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/9cb37b0d8eb84b178885c75f15b4d98c : SUCCESS in 49m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fee70d696359480e8ca0fde7febdf92f : SUCCESS in 48m 10s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/86474988779a4e65ad3083d68f914b9c : SUCCESS in 1h 13m 23s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0ee3a272d7be45d9b7db03ec34342062 : SUCCESS in 1h 12m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4472f812a00d4b56ab685e51673a2377 : SUCCESS in 44m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8b9478154ddd4c18836acb34498c9535 : SUCCESS in 45m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0eb975a523941c3af1298813c811fc0 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6b6ffcf22ae4989a0c05a1c35e95d76 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64a6d12c6085442f91c22bd04be0227c : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/996ef225c826467ba2167cfc16e7a70d : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b469a39d7a94a988edac3d8ca92df02 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24c8d6bcaa684b658426cac230cd6cf4 : SUCCESS in 9m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ff4a057445fc489781f3cd11ed9152aa : SUCCESS in 37m 09s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/22a239f30ddc40e782c2dbbb408408cc : SUCCESS in 6m 11s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9c21d60481e5ca796aa0fb909b6b00aba6e50bf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-05 01:08:43.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b9a7ed0f97caba380635814dca7202a2dbfba377","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 02:25:20.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8151cb11aed246d3b47d9027594639bf\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3901ad43cb994c35bcff4f7f81daa6df : SUCCESS in 48m 23s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7184ce56b37f4048869e56fee13bae90 : SUCCESS in 1h 01m 29s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7771acf471849d6a61dc35d030e38f5 : SUCCESS in 1h 04m 34s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/27630c1d52934259a67a670b64f756c2 : SUCCESS in 48m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a0cdbeb114fc405e941577ef437ae0c1 : SUCCESS in 37m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f5f20db297574cc0a1fc7d7f748569c5 : SUCCESS in 1h 11m 27s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f09a0a5e38ef42639655323bccdb5c4f : SUCCESS in 59m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ed77b788cd7547e3a2b8c35169c488e6 : SUCCESS in 44m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bcd371f4b82c47fcbe868a48e3863a05 : SUCCESS in 53m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1826fa9acbdc4b1e8eb21efc9f40721c : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66e017f4bb2649f392a290b99d5b470d : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1a1e763a9e34c9394c0488b6f694deb : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16f64b132ec946e4927aba0a54b6a25e : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b18bcd6286a499ea8954554e317e2ce : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d1c5bf26b194bf9a7a27eaffe216209 : SUCCESS in 7m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/61568e6dd90f4a8baecc2cea025f2813 : SUCCESS in 34m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ccc575f7c8434bc0b47896a6a31bb203 : SUCCESS in 4m 00s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5729aec1b25abba21b39f5e50b0ba4125503d9fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-06 03:31:29.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"5af9ffd80a2fc3e65ca406486c63b5dc048a0b81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 04:32:26.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64a005d2db814a1fb666351bbc3de6e1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e7a6ff26f0bd4d47a9d7a69036173d5f : SUCCESS in 48m 29s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b88c89c42b3545efb79b0a88c32239ff : SUCCESS in 54m 17s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb68f4d38a15426ba182e88c2d4396bb : SUCCESS in 46m 45s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/bf9eacaf1c2e4c4c994acfa198a711a1 : SUCCESS in 38m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5dc253236dd24568a06205e8b9aebb54 : SUCCESS in 40m 02s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6aadf4cf75d7460a923424cd25f4975c : SUCCESS in 56m 58s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/546cc782c09f4efb9a5dc62de34e2e5f : SUCCESS in 50m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4d324f38065b459ba401ec3a80c0a035 : SUCCESS in 40m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f42cc134e66c4f4b8fb00c32111cad80 : SUCCESS in 36m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c6058a1d5aa4ddcba5c39c4cead6395 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48910cd3bfbf453494f5745b7fdfbe1e : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e42195a32944794b824a935c6596c56 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb9a5b59e8c14f47ab212ac626f25829 : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/faa408f6387d4e32a54e0b981437eded : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36f2ed1713df445ab5437793a5edc4b5 : SUCCESS in 6m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fc4e6446d1b54433aca1ac51346d48c3 : SUCCESS in 33m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6738d355cde44ea99ef45e8ad5713887 : SUCCESS in 4m 19s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9eff42a980bcb7ed47f4ec9995470d8451320d6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-07 17:28:38.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"8bc18acdd9c837a442ff9128d081710299fa00ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-07 18:45:28.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a65354b0c2c4479bc1ca5680efe444a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4f739beb05af42fab68d17a2b27e4f97 : SUCCESS in 40m 00s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ef7fd74b854b47878983cbfc7d5bdc46 : SUCCESS in 1h 12m 08s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/58f20661a6c745f2a46d01e375ed2e41 : SUCCESS in 44m 11s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/c657a068536c492487b8e401791e5beb : SUCCESS in 38m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2f39558355ca49548c2a206d6ccd707d : SUCCESS in 34m 47s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/20c0b9f42afe49c289773717a682bfe3 : SUCCESS in 52m 58s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b1272da30c604f1ca313f407d718ef12 : SUCCESS in 48m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/542f19c47cae4e4b9a5c5195fb3ba38c : SUCCESS in 36m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6318c5fc12ed4461bfb5eca763da3099 : SUCCESS in 38m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16ea33af39f14501bc407cdaa56f9b5d : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32a555d52f9b4df5940d2760b24df702 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7458a45863e34a1a9a2d45b43474df56 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bb523ab0b8a4bb5b43255e5d6724e6c : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/332b68a2c5964c79a653af77fc6dfb7d : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b5760f160954991889844999ac81663 : SUCCESS in 7m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/de6e7ee0b3f24261bc941f12d2b096db : SUCCESS in 45m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/15523b0b713d43d7b97a88c11613636f : SUCCESS in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ea2a87fe37c4d0834a930a3f8bd96f1ac6d9db0d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-09 20:41:07.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"dacc555d8940cf2362ff581821abd79487bde98b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-10 06:00:23.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"7742506b826b30dff1d281fbe3a8f471f88f70f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 06:01:02.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"c859db5283a767a967500490ffca2b980c7aa770","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 07:05:22.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d67141d6c71b4a3ea22cf534ac41f694\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/07db5666fa9746199d9090177ad509a6 : SUCCESS in 46m 49s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/059228870f3c48afbf97957bfcf224a3 : SUCCESS in 45m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b7160a3a335d426abd1c6ef9e762d1f6 : SUCCESS in 42m 03s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e60fb55b864d4773a3ed8fb4cc3e0436 : SUCCESS in 1h 03m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2269b92ee25243b0b173a57e39d6636d : SUCCESS in 46m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e4afd83565774d659a99f3e3bf0bc025 : SUCCESS in 48m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21a74b47687d40f1932ec12605299c1b : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce3051a18ff1433389fbdb04101459e5 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ae36b12333a4c9eaf3b06ab1ead7aef : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4419a177bb724aae9b6777bdda34b58f : SUCCESS in 8m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1fa02772f99741deb1b65a6bbd376171 : SUCCESS in 32m 42s","accounts_in_message":[],"_revision_number":11},{"id":"0d3d5f0e4472a9e770e3677b720bf33f9f795ecd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 07:05:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"510a15befd324118b83422f0800bd709fe7dcc32","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 07:06:50.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2a07073973e4199b00e7c8e93e11f4a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/540b187be18c4beb86c1f9cda720836f : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"fe387ae91be85391005dafcdcc92f5a3f01bff62","revisions":{"d5bb92b432685591e38bb9ed5f6d3519568fb3e6":{"kind":"REWORK","_number":1,"created":"2022-07-19 04:20:57.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/47/849847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/849847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/849847/1"}}},"commit":{"parents":[{"commit":"8abb3117afa5347e100887577eb29578371aac7e","subject":"Cleaned up adapter code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8abb3117afa5347e100887577eb29578371aac7e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:20:26.000000000","tz":-420},"subject":"Add basic information to represent DesignateObjects","message":"Add basic information to represent DesignateObjects\n\nWe are currently override __repr__ in our base VersionedObject\nand removing some useful information when debugging. Adding the\nclass name back into our implementation of __repr__.\n\nChange-Id: Idfd6c959cf943c5aaf95746f16070000bf82ca21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d5bb92b432685591e38bb9ed5f6d3519568fb3e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d5bb92b432685591e38bb9ed5f6d3519568fb3e6"}]},"branch":"refs/heads/master"},"8b2ef29f0b95caea4dc9b1f3c8a1b5e71e78af1f":{"kind":"REWORK","_number":2,"created":"2022-07-20 01:31:29.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/47/849847/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/849847/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/849847/2"}}},"commit":{"parents":[{"commit":"8abb3117afa5347e100887577eb29578371aac7e","subject":"Cleaned up adapter code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8abb3117afa5347e100887577eb29578371aac7e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-20 01:30:34.000000000","tz":-420},"subject":"Fixed issues with __repr__ and __str__ on objects","message":"Fixed issues with __repr__ and __str__ on objects\n\nWe are currently override __repr__ in our base VersionedObject\nand removing some useful information when debugging. Adding the\nclass name back into our implementation of __repr__.\n\nRe-wrote __str__ to not raise an exception if data is missing.\n\nRemoved some unused _to_string functions.\n\nChange-Id: Idfd6c959cf943c5aaf95746f16070000bf82ca21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8b2ef29f0b95caea4dc9b1f3c8a1b5e71e78af1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8b2ef29f0b95caea4dc9b1f3c8a1b5e71e78af1f"}]},"branch":"refs/heads/master"},"d48eafe276a8617e8bd649997b10c2131d546466":{"kind":"REWORK","_number":3,"created":"2022-07-20 01:37:48.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/47/849847/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/849847/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/849847/3"}}},"commit":{"parents":[{"commit":"8abb3117afa5347e100887577eb29578371aac7e","subject":"Cleaned up adapter code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8abb3117afa5347e100887577eb29578371aac7e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-20 01:37:29.000000000","tz":-420},"subject":"Fixed issues with __repr__ and __str__ on objects","message":"Fixed issues with __repr__ and __str__ on objects\n\nWe are currently override __repr__ in our base VersionedObject\nand removing some useful information when debugging. Adding the\nclass name back into our implementation of __repr__.\n\nRe-wrote __str__ to not raise an exception if data is missing.\n\nRemoved some unused _to_string functions.\n\nChange-Id: Idfd6c959cf943c5aaf95746f16070000bf82ca21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d48eafe276a8617e8bd649997b10c2131d546466"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d48eafe276a8617e8bd649997b10c2131d546466"}]},"branch":"refs/heads/master"},"b02dc4d006499e3ebdc267abf8a2df0a7c195fe7":{"kind":"REWORK","_number":4,"created":"2022-07-20 02:13:42.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/47/849847/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/849847/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/849847/4"}}},"commit":{"parents":[{"commit":"8abb3117afa5347e100887577eb29578371aac7e","subject":"Cleaned up adapter code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8abb3117afa5347e100887577eb29578371aac7e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-20 02:13:34.000000000","tz":-420},"subject":"Fixed issues with __repr__ and __str__ on objects","message":"Fixed issues with __repr__ and __str__ on objects\n\nWe are currently override __repr__ in our base VersionedObject\nand removing some useful information when debugging. Adding the\nclass name back into our implementation of __repr__.\n\nRe-wrote __str__ to not raise an exception if data is missing.\n\nRemoved some unused _to_string functions.\n\nChange-Id: Idfd6c959cf943c5aaf95746f16070000bf82ca21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b02dc4d006499e3ebdc267abf8a2df0a7c195fe7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b02dc4d006499e3ebdc267abf8a2df0a7c195fe7"}]},"branch":"refs/heads/master"},"b4c6d0dbe7e5bdfa20f2c2b72ec242757b45922f":{"kind":"REWORK","_number":5,"created":"2022-07-21 04:28:47.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/47/849847/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/849847/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/849847/5"}}},"commit":{"parents":[{"commit":"a18213e327d881cfaa3e456f062886bdd85d9d4d","subject":"Cleaned up and fixed record objects and tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a18213e327d881cfaa3e456f062886bdd85d9d4d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-21 04:28:29.000000000","tz":-420},"subject":"Fixed issues with __repr__ and __str__ on objects","message":"Fixed issues with __repr__ and __str__ on objects\n\nWe are currently override __repr__ in our base VersionedObject\nand removing some useful information when debugging. Adding the\nclass name back into our implementation of __repr__.\n\nRe-wrote __str__ to not raise an exception if data is missing.\n\nRemoved some unused _to_string functions.\n\nChange-Id: Idfd6c959cf943c5aaf95746f16070000bf82ca21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b4c6d0dbe7e5bdfa20f2c2b72ec242757b45922f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b4c6d0dbe7e5bdfa20f2c2b72ec242757b45922f"}]},"branch":"refs/heads/master"},"81d500ef6425d860fc2d3a3cbf3dbdf6628404d8":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-07-21 04:40:56.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/47/849847/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/849847/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/849847/6"}}},"commit":{"parents":[{"commit":"d10056178ae13c25289f8cdd41c48062da2aab62","subject":"Cleaned up and fixed record objects and tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d10056178ae13c25289f8cdd41c48062da2aab62"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-21 04:40:56.000000000","tz":0},"subject":"Fixed issues with __repr__ and __str__ on objects","message":"Fixed issues with __repr__ and __str__ on objects\n\nWe are currently override __repr__ in our base VersionedObject\nand removing some useful information when debugging. Adding the\nclass name back into our implementation of __repr__.\n\nRe-wrote __str__ to not raise an exception if data is missing.\n\nRemoved some unused _to_string functions.\n\nChange-Id: Idfd6c959cf943c5aaf95746f16070000bf82ca21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/81d500ef6425d860fc2d3a3cbf3dbdf6628404d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/81d500ef6425d860fc2d3a3cbf3dbdf6628404d8"}]},"branch":"refs/heads/master","description":"Rebase"},"efe03d74314180443137bde3a298c01d2035a137":{"kind":"REWORK","_number":7,"created":"2022-08-03 03:40:50.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/47/849847/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/849847/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/849847/7"}}},"commit":{"parents":[{"commit":"d10056178ae13c25289f8cdd41c48062da2aab62","subject":"Cleaned up and fixed record objects and tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d10056178ae13c25289f8cdd41c48062da2aab62"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-03 03:40:07.000000000","tz":-420},"subject":"Fixed issues with __repr__ and __str__ on objects","message":"Fixed issues with __repr__ and __str__ on objects\n\nWe are currently override __repr__ in our base VersionedObject\nand removing some useful information when debugging. Adding the\nclass name back into our implementation of __repr__.\n\n- Re-wrote __str__ to not raise an exception if data is missing.\n- Removed some unused _to_string functions.\n- Changed _from_string to from_string.\n\nChange-Id: Idfd6c959cf943c5aaf95746f16070000bf82ca21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/efe03d74314180443137bde3a298c01d2035a137"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/efe03d74314180443137bde3a298c01d2035a137"}]},"branch":"refs/heads/master"},"bfb0c53a2125a95440743feb8d1876c69835cbfd":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-08-04 02:17:43.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/47/849847/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/849847/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/849847/8"}}},"commit":{"parents":[{"commit":"ef187ddec8599c2833909049f77e53d90cf4eea4","subject":"Merge \"Re-factored central and rpc decorators\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ef187ddec8599c2833909049f77e53d90cf4eea4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-04 02:17:43.000000000","tz":0},"subject":"Fixed issues with __repr__ and __str__ on objects","message":"Fixed issues with __repr__ and __str__ on objects\n\nWe are currently override __repr__ in our base VersionedObject\nand removing some useful information when debugging. Adding the\nclass name back into our implementation of __repr__.\n\n- Re-wrote __str__ to not raise an exception if data is missing.\n- Removed some unused _to_string functions.\n- Changed _from_string to from_string.\n\nChange-Id: Idfd6c959cf943c5aaf95746f16070000bf82ca21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bfb0c53a2125a95440743feb8d1876c69835cbfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bfb0c53a2125a95440743feb8d1876c69835cbfd"}]},"branch":"refs/heads/master","description":"Rebase"},"b9ef3ff2fc57c51bae4d088c86e64dc26ac99295":{"kind":"REWORK","_number":9,"created":"2022-08-05 01:08:43.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/47/849847/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/849847/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/849847/9"}}},"commit":{"parents":[{"commit":"ef187ddec8599c2833909049f77e53d90cf4eea4","subject":"Merge \"Re-factored central and rpc decorators\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ef187ddec8599c2833909049f77e53d90cf4eea4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-05 01:08:36.000000000","tz":-420},"subject":"Fixed issues with __repr__ and __str__ on objects","message":"Fixed issues with __repr__ and __str__ on objects\n\nWe are currently override __repr__ in our base VersionedObject\nand removing some useful information when debugging.\n\n- Re-wrote __str__ to not raise an exception if data is missing.\n- Removed some unused _to_string functions.\n- Changed _from_string to from_string.\n\nChange-Id: Idfd6c959cf943c5aaf95746f16070000bf82ca21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9ef3ff2fc57c51bae4d088c86e64dc26ac99295"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9ef3ff2fc57c51bae4d088c86e64dc26ac99295"}]},"branch":"refs/heads/master"},"0a24ae2808bb46fadcfef8785b9f9d362c90f429":{"kind":"REWORK","_number":10,"created":"2022-08-06 03:31:29.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/47/849847/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/849847/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/849847/10"}}},"commit":{"parents":[{"commit":"ef187ddec8599c2833909049f77e53d90cf4eea4","subject":"Merge \"Re-factored central and rpc decorators\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ef187ddec8599c2833909049f77e53d90cf4eea4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-06 03:31:01.000000000","tz":-420},"subject":"Fixed issues with __repr__ and __str__ on objects","message":"Fixed issues with __repr__ and __str__ on objects\n\nWe are currently override __repr__ in our base VersionedObject\nand removing some useful information when debugging.\n\n- Re-wrote __str__ to not raise an exception if data is missing.\n- Removed some unused _to_string functions.\n- Changed _from_string to from_string.\n\nChange-Id: Idfd6c959cf943c5aaf95746f16070000bf82ca21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a24ae2808bb46fadcfef8785b9f9d362c90f429"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a24ae2808bb46fadcfef8785b9f9d362c90f429"}]},"branch":"refs/heads/master"},"fe387ae91be85391005dafcdcc92f5a3f01bff62":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-08-07 17:28:38.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/47/849847/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/849847/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/849847/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/849847/11"}}},"commit":{"parents":[{"commit":"8beccabe4eb78f7da60748cec84df66946d423e7","subject":"Added additional test coverage for adapters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8beccabe4eb78f7da60748cec84df66946d423e7"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-07 17:28:38.000000000","tz":0},"subject":"Fixed issues with __repr__ and __str__ on objects","message":"Fixed issues with __repr__ and __str__ on objects\n\nWe are currently override __repr__ in our base VersionedObject\nand removing some useful information when debugging.\n\n- Re-wrote __str__ to not raise an exception if data is missing.\n- Removed some unused _to_string functions.\n- Changed _from_string to from_string.\n\nChange-Id: Idfd6c959cf943c5aaf95746f16070000bf82ca21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fe387ae91be85391005dafcdcc92f5a3f01bff62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fe387ae91be85391005dafcdcc92f5a3f01bff62"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
