)]}'
{"id":"openstack%2Fdesignate~849848","triplet_id":"openstack%2Fdesignate~master~I6e1a74fa3ecca0f82735ba36ac4336ab6cdeb56e","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-08-02 22:21:49.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-08-06 03:34:28.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I6e1a74fa3ecca0f82735ba36ac4336ab6cdeb56e","subject":"Added additional test coverage for adapters","status":"MERGED","created":"2022-07-19 04:33:31.000000000","updated":"2022-08-06 05:43:49.000000000","submitted":"2022-08-06 05:42:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849848","meta_rev_id":"e4edc57e9023ce34cc46331e0d4429ac73cc6c16","_number":849848,"virtual_id_number":849848,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-06 05:42:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-08-06 03:28:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-06 03:28:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-08-06 03:34:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-19 05:47:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-19 22:14:31.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-07-21 20:18:38.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"6cb9f4ab97d93730c2aaef9bde6783aa270ef174","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-19 04:33:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bb08da44e0e1fff9f136472a8f67a478e66f8de","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-19 04:47:31.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"e4310bbb4d1a20ed52b84d1311f7c6cb19f58b11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 05:47:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c52554d7b084b0dbca7058151afa24f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c8b07eb39d3d4818884dd1d4114bcb61 : SUCCESS in 47m 03s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d388c07a6bba4f9e8f32f460eb1f9b95 : SUCCESS in 1h 07m 56s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d73a585b3a047e09ac37fa7be3a46b5 : SUCCESS in 57m 24s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3dacfb87174040a793dd9d02ae421e99 : SUCCESS in 47m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/788b0b5df55a4d19bb8834333f1d506c : SUCCESS in 47m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/05b1efca0701412cad0f6c4ffb7c7774 : SUCCESS in 1h 06m 11s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5336272e0ee64c438cf4079f5d91cd7e : SUCCESS in 1h 03m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d86c60e9224846cf9d85d92b5b1c257f : SUCCESS in 51m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/505baa8ccef148f4bf33b9020f9fe096 : SUCCESS in 46m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9ecdc85edaa4ecba888a3905cfc7bcf : FAILURE in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed06c51082f443a0acc70d02d799a8e9 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4004d2d453e7456396a4ec3ac8810812 : FAILURE in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10739aca020f4581921ce96a01aaebfb : FAILURE in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2d04734369742e9bcb04cae5de1eb5a : FAILURE in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6895318b368940ec8688a406247f3786 : SUCCESS in 7m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0461f263fffb495883189bd5b056c388 : SUCCESS in 35m 01s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/62e27a9ce0134a549c78c02a1d89390b : FAILURE in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c1ce72a73ea96f332a9fb17eee2ae35acdfb609","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-07-19 22:14:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"33afcd618707f724eef4eb5fc6bc35de64add077","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-20 01:52:00.000000000","message":"Topic backup removed","accounts_in_message":[],"_revision_number":1},{"id":"7d62f6583f4731ed15954ba01ff5f282427ae0f7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-20 01:52:07.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ad0c2a9f247c4b6a174fde9adf82a735a2b040ed","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-20 01:59:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3f1ab05600c6472d4e3e6f2c43f558412c4e67b8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-20 02:02:23.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"3054854b93aa920ff74d31911f8ce4751fd18f4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-20 02:15:45.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"dedf1a160e9682aa1803607a8c4a2b8602cf0704","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 03:38:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/967ab29e40bd4ff5a7ee401cf61b5018\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2b46082ccd6641b2bab1ffcb05216ef7 : SUCCESS in 43m 21s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d2ef7d6f7c654169aafcbecd9ad7ce37 : SUCCESS in 1h 03m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/de09fd5a81cc446099e13a35dca01913 : SUCCESS in 57m 21s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/dbca8d9843b34c9ea79fce3d97ae356f : SUCCESS in 42m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/886d750643724edea8387246eb34d8ec : SUCCESS in 44m 39s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/cd7db17959df4bcc8f0dc4ef6a88905b : SUCCESS in 1h 08m 25s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/399bac5897fd4d2ebc412938a8b6a1a0 : SUCCESS in 1h 06m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/86a342e0df9a4e9789442041bfa5c0d7 : SUCCESS in 52m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6d90494c03bf4e36bc67dfed3c141b60 : SUCCESS in 43m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a55ec4ceda20429e943481f0c355344f : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79a141727bda40b2813a26bd1428f023 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26c7f68d1b0c428c8a02561ece4e0649 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7351000e6c314898ac812ba95cac5344 : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/689ca50a319f4c01b8f615fa271c6084 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d30da6236be402b85b47af223decc3a : SUCCESS in 9m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d932fb270e7c48ef8aee9a4ad8e27f18 : SUCCESS in 36m 13s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a3f12aeee0d54e12b6013165518b9802 : SUCCESS in 5m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c718f5516a6603c5c5b3150de3255487189c20e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-21 04:28:59.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"a2264ae28f53baa0c83de3d0e76625322deaf015","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-21 04:40:59.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"02f2e62e51183cc13602e1f6e816817a756bc343","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 05:58:45.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1f95f8882524bd0b860b8bf81cced2a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/33d77b91370b428fa66138d2691fb3bf : SUCCESS in 59m 17s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6615043297604f5995d07df255754f37 : SUCCESS in 1h 10m 53s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5924a9bf3a44c83927df40a375e4988 : SUCCESS in 1h 01m 15s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e6a1aad5bc2f495e88af9ddc80451779 : SUCCESS in 59m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e908120485284fe99eaae75d534325dd : SUCCESS in 55m 05s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8bc32d9f097b470ab78efd838a33d1fe : SUCCESS in 1h 14m 28s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/582f808059bf4646b225ae4f9f4ac833 : SUCCESS in 1h 14m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/deca7ffbc5ec468988484e35b1db3f53 : SUCCESS in 53m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/734ca4d6a80e41ecbaf70bee308900c6 : SUCCESS in 50m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0b64fdf7d6b4874b9ee1a12efd920a3 : SUCCESS in 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f4bcf33a2774c1abc70d8fd76cfecfd : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a6eed0514ad428898c02fc54ac4e37c : SUCCESS in 7m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8aa529d9a4774283b9b767d56db39a09 : SUCCESS in 7m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd7b2460f20e4ecbb4e384d2328ebe1e : FAILURE in 7m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bed416fdc531403d853b4edbf434ef35 : SUCCESS in 11m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cd9cb60ca9664ff284b815761aaa51c5 : SUCCESS in 48m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a0fe07bd6b6b414d807504f242d7c330 : SUCCESS in 7m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"66bde2ad7c5f2002312e355b7ee8da3c83935a5e","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-07-21 20:18:38.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c796631fa542784b0085abfadfadd56e8a8615ee","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-21 21:24:30.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3be49cc7f0784c5b2c450f58280673dde03a82a2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-21 21:56:20.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"81b93e501fee772e6a308bb23a259a43ef7ff7fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-23 04:43:05.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8184e6903fe3f81d02ac151682b1e93f99dec86c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-23 04:46:03.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7260e3a42d86ebb56796c2e2da1be8d2f868a0e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-23 06:02:55.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/487e0baab17548ad83c5345c00dfcef0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4eb00db7c14449f7acd1540a46c5382f : SUCCESS in 40m 20s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4b47f1cdcae546c1a9606ac0004885f6 : SUCCESS in 52m 38s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/df925334c724450f9a5a0e1de75dcb78 : SUCCESS in 44m 41s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ee346d0f89364124bdbb9d7babf8b229 : SUCCESS in 42m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6bdf5472a8f14345a7ab569e44b3bdbc : SUCCESS in 33m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e3c3bad3d64b42a9b2e71b46caaae6c8 : SUCCESS in 1h 11m 50s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b90f89c08bee44c7a8fe136eea85d0cd : SUCCESS in 52m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e43c881a1bb94bdbb53cb3092e4b749e : SUCCESS in 35m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ccb01e93dd6845d78e176a7d3a1d1366 : SUCCESS in 53m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84418668e43d4becb5d5a0365fbd6849 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/821865a844c74c7ea9c0ab1a0c92e5d4 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b648757315e40fdbc5555bbef77fa53 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7fb8130084a4ddc97021b8cf0b2fb8b : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/280ee08e25ad46a9bb132139b7db2dcc : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a650734f1114e2cbf8767a9f6dcd36b : SUCCESS in 7m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d54cd0fadd6e43db9dac77e1e8029776 : SUCCESS in 44m 26s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/86b6df73af2e4207b603a846f90a3515 : SUCCESS in 5m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"021a619a007e4dab05f1b26651da653f31a2f5d9","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-07-26 16:00:36.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1433854e516ffe0d9f597491b3f97e6d802f451e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-02 22:21:49.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"be0754aa23e00d72ba30e76196f3f3bf2c63e1b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-03 05:23:25.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"5b4c27e3305f14907e1f2de16a49ea589f9697b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 06:40:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08c1af8aeaaf459e880f27fbbe1160c8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f46254d077d24c479af5c05701af3b7a : SUCCESS in 48m 08s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3c7fd7cefe30471797b0ae9c786f6421 : SUCCESS in 1h 10m 09s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/285296deeec24fe6982361e6b7a831b2 : SUCCESS in 1h 04m 34s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/6be6f56895bc4f65816b100afabea3ba : SUCCESS in 53m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/86bf7d05596b434489ec8fee7b7eb272 : SUCCESS in 44m 16s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c97950d9583d473781b84b318e359ad4 : SUCCESS in 1h 12m 10s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ea42fcbd143a4b91b258b37a6b529ddf : SUCCESS in 1h 09m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dfb51cc0d2d2457ba5de7bf16b6e98dd : SUCCESS in 50m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2fcc393f46bb4be1bd29a02529d659ca : SUCCESS in 37m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4bb9fa2f36b4bc3a57a15d863f36160 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75c9c832140f4436a841968f4aac00a7 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9214339c433341f0847cc969af8ce03c : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d1777b7cf974b33a7cec9c7dec556b4 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5d39312d2be457b991d86e40e098ee0 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a42177b78aa44f87a174883837f984e6 : SUCCESS in 6m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/148f066c08714885b8230dc0985bee42 : SUCCESS in 39m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8c43d91669d242f99905ba4c199622f4 : SUCCESS in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2d1220702740fa0bb30c10e7f4bf4f562053f318","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-04 02:17:49.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"9178bcca1b5468c8c2195a42987157a1d1351416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 03:34:28.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ebb2c14e7d5434a86ac85a5d7d71b1b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3d1c8401680246c2ba90b5cbc980f72f : SUCCESS in 52m 55s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8dc8363f797145619eae46611b7d83ce : SUCCESS in 59m 20s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bfef9e61b29647cabab76a8ffef572c3 : SUCCESS in 54m 55s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/594f847c81734745a1102067c0ce241e : SUCCESS in 52m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5dab7cca4289443c83ffbb5ed92df68b : SUCCESS in 50m 23s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c499d548c33e4f688a88cda962f1b38d : SUCCESS in 1h 07m 24s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6676a566795641f2a1678d41b574e2c1 : SUCCESS in 1h 07m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/620ff31e9b86404db73fccf788c3b3b2 : SUCCESS in 47m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5c237950798d42309fff53d7b9186a24 : SUCCESS in 52m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/943196ea5afc420bb3e7d48b6b2f5fc7 : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6991cdb1ecb46ab99bb14de9322a248 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0e02a93956147da8ea16b151610275d : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58ccf135384f49aeb4e0e19b4cfa8b93 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c8a46abe13f54d6eb08c99d706252e63 : FAILURE in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bc519277df349009f82f64f26216ab1 : SUCCESS in 10m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/34e361d9772f4e0f8431c8db36e73379 : SUCCESS in 32m 58s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/10ad6c73ce4b4c2096a8f646c05a4c71 : SUCCESS in 5m 41s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"0f4473faa581d875397802499960a4744a05d422","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-06 03:28:23.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"1e226eedfc95666f9fab224aaca745e6ce8a5386","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-06 03:34:28.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"1e564bb456794b69829383ec591f81a46d307ba7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 04:46:27.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7203511b4e424e84be14eafcdf80c79b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8e6b3eed3c2e49c382b8fab1cc9cd157 : SUCCESS in 51m 26s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ce544e8fa75348559ed291161f4d9224 : SUCCESS in 1h 11m 54s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/33067647444046668f9b299dade93408 : SUCCESS in 1h 04m 49s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/415dba3d8e424e1aba12ded6cc215267 : SUCCESS in 53m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/516b2b7ffb6f4e89bea0d8604b7442c0 : SUCCESS in 45m 27s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b561479740c54448b2b23274edb5a560 : SUCCESS in 1h 10m 53s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b86c82827d9c4316bdba7566aff50af4 : SUCCESS in 1h 10m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/61baa65dc40d4c46a2148d73ebe8c404 : SUCCESS in 51m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5a97179f20dc4b8aabcf2feddb285f86 : SUCCESS in 50m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45dcbd6e63ef4c5d8e5c5ce851b44f39 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/975ba4e7ab0e4955a8a8c241c5bbf8d2 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06249b9e7d3f4f9494c31c361d21a78c : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56bed1003ed84ad29b0f2db81a017989 : SUCCESS in 7m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1b3b0f7f52d4454b07efc7f57ae824f : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2ec3860ad354012aab6b18d31368433 : SUCCESS in 7m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bddd9eaa582b4601a87a59e700c7fe88 : SUCCESS in 36m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/cb9e06fb7039408db6103bf43def8200 : SUCCESS in 4m 49s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"3747e51bc94cef008d2ca055644549a80754cc1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 04:47:04.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"5e381c13e7eeae4c0bb369c34ad93530e70a9038","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 05:42:37.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db90f94f42494dbe892d3dde80f7bfec\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/03fadd987e5249bfa379adcdf1f82b41 : SUCCESS in 37m 45s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4e742365413c402598a89d0c296d8ba4 : SUCCESS in 35m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/85dc6ed568194872a7352d27e3156d88 : SUCCESS in 34m 27s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7998cad5a53f44d49ed49301b70ea2c1 : SUCCESS in 50m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/46c8b961aa354ec3990ff74353c760c7 : SUCCESS in 48m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f17b938510c8463582ea8e685c7b0aec : SUCCESS in 38m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/621eef861f844abfa2ae1aa2a0545bc7 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cb1431cbe504a2395c5b67ca69d3e2f : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e839f32a6997461ba06c0a4fdd48c799 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d43c9065d87435cae390dc4776c0cca : SUCCESS in 7m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7448bc7e3fbc46aca797cfee41992801 : SUCCESS in 38m 44s","accounts_in_message":[],"_revision_number":11},{"id":"f6ba408cbbb92f1e6204ee6bdd6ecd51396979d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 05:42:37.000000000","message":"Change has been successfully merged\n\n8 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":11},{"id":"e4edc57e9023ce34cc46331e0d4429ac73cc6c16","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 05:43:49.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8b4759073d14532bd90f8af353d85f1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/085619b0bd2f435ea97943f76b13e384 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"8beccabe4eb78f7da60748cec84df66946d423e7","revisions":{"0f95e80d43519f234f65c6410b7074afc1d4d9b0":{"kind":"REWORK","_number":1,"created":"2022-07-19 04:33:31.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/48/849848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/849848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/849848/1"}}},"commit":{"parents":[{"commit":"8abb3117afa5347e100887577eb29578371aac7e","subject":"Cleaned up adapter code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8abb3117afa5347e100887577eb29578371aac7e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:33:25.000000000","tz":-420},"subject":"Added additional test coverage for adapters","message":"Added additional test coverage for adapters\n\nChange-Id: I6e1a74fa3ecca0f82735ba36ac4336ab6cdeb56e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0f95e80d43519f234f65c6410b7074afc1d4d9b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0f95e80d43519f234f65c6410b7074afc1d4d9b0"}]},"branch":"refs/heads/master"},"6e33437d72cab261552fd6838f58db692c9a52b1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-07-20 01:52:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/48/849848/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/849848/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/849848/2"}}},"commit":{"parents":[{"commit":"d48eafe276a8617e8bd649997b10c2131d546466","subject":"Fixed issues with __repr__ and __str__ on objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d48eafe276a8617e8bd649997b10c2131d546466"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-20 01:52:07.000000000","tz":0},"subject":"Added additional test coverage for adapters","message":"Added additional test coverage for adapters\n\nChange-Id: I6e1a74fa3ecca0f82735ba36ac4336ab6cdeb56e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6e33437d72cab261552fd6838f58db692c9a52b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6e33437d72cab261552fd6838f58db692c9a52b1"}]},"branch":"refs/heads/master","description":"Rebase"},"9ece9b18bd82a81b101e1facf37c33cc3c5a46d1":{"kind":"REWORK","_number":3,"created":"2022-07-20 01:59:41.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/48/849848/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/849848/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/849848/3"}}},"commit":{"parents":[{"commit":"d48eafe276a8617e8bd649997b10c2131d546466","subject":"Fixed issues with __repr__ and __str__ on objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d48eafe276a8617e8bd649997b10c2131d546466"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-20 01:59:17.000000000","tz":-420},"subject":"Added additional test coverage for adapters","message":"Added additional test coverage for adapters\n\n- Removed uncessary log line in base adapter.\n\nChange-Id: I6e1a74fa3ecca0f82735ba36ac4336ab6cdeb56e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ece9b18bd82a81b101e1facf37c33cc3c5a46d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ece9b18bd82a81b101e1facf37c33cc3c5a46d1"}]},"branch":"refs/heads/master"},"fd78fd129a00056f3f3a578a2359cb96db640dbf":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-07-20 02:15:45.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/48/849848/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/849848/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/849848/4"}}},"commit":{"parents":[{"commit":"b02dc4d006499e3ebdc267abf8a2df0a7c195fe7","subject":"Fixed issues with __repr__ and __str__ on objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b02dc4d006499e3ebdc267abf8a2df0a7c195fe7"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-20 02:15:45.000000000","tz":0},"subject":"Added additional test coverage for adapters","message":"Added additional test coverage for adapters\n\n- Removed uncessary log line in base adapter.\n\nChange-Id: I6e1a74fa3ecca0f82735ba36ac4336ab6cdeb56e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fd78fd129a00056f3f3a578a2359cb96db640dbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fd78fd129a00056f3f3a578a2359cb96db640dbf"}]},"branch":"refs/heads/master","description":"Rebase"},"1e674dd8b690e93fc6778ab436fd5e9bf6a93c5b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-07-21 04:28:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/48/849848/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/849848/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/849848/5"}}},"commit":{"parents":[{"commit":"b4c6d0dbe7e5bdfa20f2c2b72ec242757b45922f","subject":"Fixed issues with __repr__ and __str__ on objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b4c6d0dbe7e5bdfa20f2c2b72ec242757b45922f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-21 04:28:59.000000000","tz":0},"subject":"Added additional test coverage for adapters","message":"Added additional test coverage for adapters\n\n- Removed uncessary log line in base adapter.\n\nChange-Id: I6e1a74fa3ecca0f82735ba36ac4336ab6cdeb56e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e674dd8b690e93fc6778ab436fd5e9bf6a93c5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e674dd8b690e93fc6778ab436fd5e9bf6a93c5b"}]},"branch":"refs/heads/master","description":"Rebase"},"9384a34566e70190f4b3c166640da4efcc984bb2":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-07-21 04:40:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/48/849848/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/849848/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/849848/6"}}},"commit":{"parents":[{"commit":"81d500ef6425d860fc2d3a3cbf3dbdf6628404d8","subject":"Fixed issues with __repr__ and __str__ on objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/81d500ef6425d860fc2d3a3cbf3dbdf6628404d8"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-21 04:40:59.000000000","tz":0},"subject":"Added additional test coverage for adapters","message":"Added additional test coverage for adapters\n\n- Removed uncessary log line in base adapter.\n\nChange-Id: I6e1a74fa3ecca0f82735ba36ac4336ab6cdeb56e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9384a34566e70190f4b3c166640da4efcc984bb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9384a34566e70190f4b3c166640da4efcc984bb2"}]},"branch":"refs/heads/master","description":"Rebase"},"80dfc617553773dcec5bba4c1b6f6c4c8a28bfc6":{"kind":"REWORK","_number":7,"created":"2022-07-23 04:43:05.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/48/849848/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/849848/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/849848/7"}}},"commit":{"parents":[{"commit":"81d500ef6425d860fc2d3a3cbf3dbdf6628404d8","subject":"Fixed issues with __repr__ and __str__ on objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/81d500ef6425d860fc2d3a3cbf3dbdf6628404d8"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-23 04:42:55.000000000","tz":-420},"subject":"Added additional test coverage for adapters","message":"Added additional test coverage for adapters\n\n- Removed uncessary log line in base adapter.\n\nChange-Id: I6e1a74fa3ecca0f82735ba36ac4336ab6cdeb56e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/80dfc617553773dcec5bba4c1b6f6c4c8a28bfc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/80dfc617553773dcec5bba4c1b6f6c4c8a28bfc6"}]},"branch":"refs/heads/master"},"e2ab7c1510e3959b76edf3f51e1fae5ed4106baf":{"kind":"REWORK","_number":8,"created":"2022-07-23 04:46:03.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/48/849848/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/849848/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/849848/8"}}},"commit":{"parents":[{"commit":"81d500ef6425d860fc2d3a3cbf3dbdf6628404d8","subject":"Fixed issues with __repr__ and __str__ on objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/81d500ef6425d860fc2d3a3cbf3dbdf6628404d8"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-23 04:45:56.000000000","tz":-420},"subject":"Added additional test coverage for adapters","message":"Added additional test coverage for adapters\n\n- Removed uncessary log line in base adapter.\n\nChange-Id: I6e1a74fa3ecca0f82735ba36ac4336ab6cdeb56e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e2ab7c1510e3959b76edf3f51e1fae5ed4106baf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e2ab7c1510e3959b76edf3f51e1fae5ed4106baf"}]},"branch":"refs/heads/master"},"ff6c70df9a9d75f4432e0ea4cd2ee88e016eee74":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2022-08-03 05:23:25.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/48/849848/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/849848/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/849848/9"}}},"commit":{"parents":[{"commit":"efe03d74314180443137bde3a298c01d2035a137","subject":"Fixed issues with __repr__ and __str__ on objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/efe03d74314180443137bde3a298c01d2035a137"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-03 05:23:25.000000000","tz":0},"subject":"Added additional test coverage for adapters","message":"Added additional test coverage for adapters\n\n- Removed uncessary log line in base adapter.\n\nChange-Id: I6e1a74fa3ecca0f82735ba36ac4336ab6cdeb56e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ff6c70df9a9d75f4432e0ea4cd2ee88e016eee74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ff6c70df9a9d75f4432e0ea4cd2ee88e016eee74"}]},"branch":"refs/heads/master","description":"Rebase"},"23b84a26d5bd879a78749fd0ae396f3e2104b9ed":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2022-08-04 02:17:49.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/48/849848/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/849848/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/849848/10"}}},"commit":{"parents":[{"commit":"bfb0c53a2125a95440743feb8d1876c69835cbfd","subject":"Fixed issues with __repr__ and __str__ on objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bfb0c53a2125a95440743feb8d1876c69835cbfd"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-04 02:17:49.000000000","tz":0},"subject":"Added additional test coverage for adapters","message":"Added additional test coverage for adapters\n\n- Removed uncessary log line in base adapter.\n\nChange-Id: I6e1a74fa3ecca0f82735ba36ac4336ab6cdeb56e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/23b84a26d5bd879a78749fd0ae396f3e2104b9ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/23b84a26d5bd879a78749fd0ae396f3e2104b9ed"}]},"branch":"refs/heads/master","description":"Rebase"},"8beccabe4eb78f7da60748cec84df66946d423e7":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-08-06 03:28:23.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/48/849848/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/849848/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/849848/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/849848/11"}}},"commit":{"parents":[{"commit":"ef187ddec8599c2833909049f77e53d90cf4eea4","subject":"Merge \"Re-factored central and rpc decorators\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ef187ddec8599c2833909049f77e53d90cf4eea4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-07-19 04:15:25.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-06 03:28:23.000000000","tz":0},"subject":"Added additional test coverage for adapters","message":"Added additional test coverage for adapters\n\n- Removed uncessary log line in base adapter.\n\nChange-Id: I6e1a74fa3ecca0f82735ba36ac4336ab6cdeb56e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8beccabe4eb78f7da60748cec84df66946d423e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8beccabe4eb78f7da60748cec84df66946d423e7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
