)]}'
{"id":"openstack%2Fdesignate~850333","triplet_id":"openstack%2Fdesignate~stable%2Fwallaby~Ic68808caef77f6c75f88c3b1297ff724469ad715","project":"openstack/designate","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2023-05-02 20:55:00.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic68808caef77f6c75f88c3b1297ff724469ad715","subject":"Change FIPS jobs to centos-9-stream","status":"ABANDONED","created":"2022-07-19 11:39:58.000000000","updated":"2023-05-02 20:55:00.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"28b66ca680aff5b95c1e9194d3806a1afb472bfd","_number":850333,"virtual_id_number":850333,"owner":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-19 13:12:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-19 13:12:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-20 19:28:28.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"}],"messages":[{"id":"ed8f4cee08c3bb46584f6df67e875b9cf293c77b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-07-19 11:39:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d875b14241d3e7f35460b7a621805ac610537b12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 13:12:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7c1701aefb54b8c999911617face59a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/df17ad16d8084d20bd5de20b87756a21 : SUCCESS in 54m 55s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d2d9b6ab87624425ba0aeac5648a67f5 : RETRY_LIMIT in 25m 30s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/c5b7820f8e3f4f309c220e49b6e38dc7 : SUCCESS in 1h 01m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5ec68c7785814e02b931932ce954ef90 : SUCCESS in 51m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/31bff0af89334247b2c0903121d118df : SUCCESS in 54m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/610f29eeeb06438e94e90d88229ec035 : SUCCESS in 58m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0319c08a06964969b0e5169d65bd73b5 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c4e092c7c9c430bb463e20e69b855d3 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e59079517cb84332b6382b16d4f7c6d8 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8582fefa5318457ebe24d7109a22ce4c : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/530c3bd279e44a26b1039315a5a38316 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf4bc136a6e4453ea3af0963ac7bd90a : SUCCESS in 8m 36s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/a6ea426adb2d4ccbafd8b0c2ee43096e : SUCCESS in 37m 20s","accounts_in_message":[],"_revision_number":1},{"id":"28ad2121ad776c7214919bb5787b8ecbfd54e44a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-20 19:28:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5700e86e9a1f620771a8822c758d1141b13449be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 20:36:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df4040bfac814d799991b3c1480f516b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/edeeea2ce2ae4b29a14b1ea352b5eac9 : SUCCESS in 50m 22s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/656262973faf4a20ae10ed021580fa4a : RETRY_LIMIT in 10m 37s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/82d92b43f0e64746b7b3b454d647c6a4 : SUCCESS in 43m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/be40c60c75254c84ab2a10a1c163dae3 : SUCCESS in 49m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/244760297ab841abb334f7698f16d235 : SUCCESS in 47m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/919535718a614727b2584efe3f48ef88 : SUCCESS in 50m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c40ab1248049412ab60b4376524f20be : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ada0d258cbd44033a5d8a6ae29ad2f30 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb69a8ba36944cddadc8946bcd9ddb09 : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/505d4056fd7a440799d69b757f6ff263 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f625b82031e40a9b87c593cb6ce52fb : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64e323f0ceda4d32904157ca208e60c8 : SUCCESS in 6m 59s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/f58d73127b0a4046908f766a691edac0 : SUCCESS in 47m 46s","accounts_in_message":[],"_revision_number":1},{"id":"28b66ca680aff5b95c1e9194d3806a1afb472bfd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-02 20:55:00.000000000","message":"Abandoned\n\nDuplicate to 875511 which has now merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11b60327dcc710ab4540afe88cb5a9b93b24873a","revisions":{"11b60327dcc710ab4540afe88cb5a9b93b24873a":{"kind":"REWORK","_number":1,"created":"2022-07-19 11:39:58.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/33/850333/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/850333/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/850333/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/850333/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/850333/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/850333/1"}}},"commit":{"parents":[{"commit":"8cef906600117b71903b54c4d7c1e072902f7a46","subject":"Minmum TTL value is zero","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8cef906600117b71903b54c4d7c1e072902f7a46"}]}],"author":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-07-19 11:41:04.000000000","tz":0},"committer":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-07-19 11:41:55.000000000","tz":0},"subject":"Change FIPS jobs to centos-9-stream","message":"Change FIPS jobs to centos-9-stream\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/848284\nChange-Id: Ic68808caef77f6c75f88c3b1297ff724469ad715\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/11b60327dcc710ab4540afe88cb5a9b93b24873a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/11b60327dcc710ab4540afe88cb5a9b93b24873a"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
