)]}'
{"id":"openstack%2Fdesignate~850563","triplet_id":"openstack%2Fdesignate~stable%2Fyoga~Ic6b3a3c6d2f5b648f00813191c30d03132e31503","project":"openstack/designate","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2022-08-16 12:10:28.000000000","reason":"Change was abandoned"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-08-16 12:10:28.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic6b3a3c6d2f5b648f00813191c30d03132e31503","subject":"Fix FIPS job - nslookup_target","status":"ABANDONED","created":"2022-07-22 13:29:43.000000000","updated":"2022-08-16 12:10:28.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"cherry_pick_of_change":847074,"cherry_pick_of_patch_set":2,"meta_rev_id":"6a661c20b23cb4100b6ec953ee82fa0aff1a490e","_number":850563,"virtual_id_number":850563,"owner":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-08-02 23:39:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-08-02 23:45:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-08-03 04:09:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-22 13:29:43.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2022-07-22 14:25:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-22 14:50:24.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2022-08-02 22:24:14.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-08-02 23:45:02.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2022-08-03 04:09:49.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"2efb5f3219ecfa10ec67f08931c16012a41bf74d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-07-22 13:29:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"020d9e87421c73f9c79674367c28ff31db85a7c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 14:25:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4421519910994f11a197d15afe4841fc\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bfdba4f1b13347ad8b5a532bfe0fdbae : SUCCESS in 42m 53s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/8529dcf9090947f8bf5b3d0d2041bb1a : FAILURE in 35m 09s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bea880403dd14eddacdb448b19ad8066 : SUCCESS in 45m 45s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/0e67089934954351a079e1045b4e06fd : SUCCESS in 43m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/24afe19a55f0432d82ca9757f93b374a : SUCCESS in 42m 42s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/19da38c5c7924fcb8c8953b173b09901 : SUCCESS in 54m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c51c4bde1b03488aa646dbfb50f1392b : SUCCESS in 41m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fd3b63f531924e8780384b2a542c5e54 : SUCCESS in 40m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02d09598693f461597e7391d89623622 : SUCCESS in 5m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bd26fdd6e594877bc76956550aba3fe : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b18d71d69c64ace8f3ff63546057b18 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af10a30197184624a8a28d59bb85a2f9 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66d1b4bc2b52481ea212941810efbcc5 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/782f390d232f48588d9b5868570d3ca5 : SUCCESS in 7m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3db3dd37fb264b328cc9fea2a496d537 : SUCCESS in 36m 20s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b37f6e8823194f6b811aed09e06d2bdf : SUCCESS in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8eb53902c9f614a05910b5595d08a6a4e8d60be","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-07-22 14:50:24.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"135b6309efe82a8633fe545c81aa7df6e2097ad3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-02 22:24:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c05fbd5d83eef60c5a1fa68e1904c14608a6176","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-02 22:24:39.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d522826e30493dd7a475cb36058ce252f7199cb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 23:39:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8ab2dafd310444e912760cbc7e9bcd7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/61254fff5719448a8fbcb9d48221f2bc : SUCCESS in 55m 31s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/acad0a36e73a427491604605d74607c9 : FAILURE in 49m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/df45642925074d90b00c92ee9fe70a04 : SUCCESS in 59m 41s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/6fcb6e2de16546978b06c6192339e091 : SUCCESS in 55m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8ca2e809f56540cb8405cfaf2f9c9e34 : SUCCESS in 48m 46s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0a6f474ebd224cf8afe550532cb93c63 : SUCCESS in 1h 07m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c97d0f47a1104b63973e160138839865 : SUCCESS in 55m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/943e8320b98c4088b9e11127f4083cc1 : SUCCESS in 57m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8df8d858cab24c7790b45ce4b8137d1c : SUCCESS in 6m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a08867a026704c6d8832dd954d9ee241 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f4f87117ee749e8b644973bc1fe8e10 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15cf3b8ce04544bda80b90fa1d04196b : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34a3263f66224a10b20cda481b060314 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea52079fd7c24300bdfed2bcab756780 : SUCCESS in 9m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dbebf883102a4d0895af9a34fd3bc7e3 : SUCCESS in 49m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/386e21cf8a2b49c3bd61935f1d0c2e3d : SUCCESS in 5m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ee86f0dc2e0d80bd4cedf8f07a13a8d85627aca2","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-08-02 23:45:02.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"970616af58907bc17fbdd6edae6c523737ff2e71","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-03 04:09:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6a661c20b23cb4100b6ec953ee82fa0aff1a490e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-08-16 12:10:28.000000000","message":"Abandoned\n\nAbandon in favor of this patch: https://review.opendev.org/c/openstack/designate/+/853279","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"58d1a3f4b08b561043e11ace221945bba26e02f5","revisions":{"dfedf402910dfb4cd398166e62bcb1061a95aa84":{"kind":"REWORK","_number":1,"created":"2022-07-22 13:29:43.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/63/850563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/850563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/850563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/850563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/850563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/850563/1"}}},"commit":{"parents":[{"commit":"2dc40a15894d2c2010b7176ff6f1dff0c8621503","subject":"Minmum TTL value is zero","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2dc40a15894d2c2010b7176ff6f1dff0c8621503"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-21 17:51:40.000000000","tz":0},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-07-22 13:29:43.000000000","tz":0},"subject":"Fix FIPS job - nslookup_target","message":"Fix FIPS job - nslookup_target\n\nThis patch sets a zuul variable \"nslookup_target\" to \u0027opendev.org\u0027 to fix the FIPS jobs due to a new zuul-job change[1].\n\n[1] https://review.opendev.org/c/zuul/zuul-jobs/+/844704\n\nChange-Id: Ic6b3a3c6d2f5b648f00813191c30d03132e31503\n(cherry picked from commit 9ac645806613710ab03da7e1841537d1d28a6e8f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dfedf402910dfb4cd398166e62bcb1061a95aa84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dfedf402910dfb4cd398166e62bcb1061a95aa84"}]},"branch":"refs/heads/stable/yoga"},"58d1a3f4b08b561043e11ace221945bba26e02f5":{"kind":"REWORK","_number":2,"created":"2022-08-02 22:24:39.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/63/850563/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/850563/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/850563/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/850563/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/850563/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/850563/2"}}},"commit":{"parents":[{"commit":"2dc40a15894d2c2010b7176ff6f1dff0c8621503","subject":"Minmum TTL value is zero","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2dc40a15894d2c2010b7176ff6f1dff0c8621503"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-21 17:51:40.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-08-02 22:24:39.000000000","tz":0},"subject":"Fix FIPS job - nslookup_target","message":"Fix FIPS job - nslookup_target\n\nThis patch sets a zuul variable \"nslookup_target\" to \u0027opendev.org\u0027 to fix the FIPS jobs due to a new zuul-job change[1].\n\n[1] https://review.opendev.org/c/zuul/zuul-jobs/+/844704\n\nChange-Id: Ic6b3a3c6d2f5b648f00813191c30d03132e31503\n(cherry picked from commit 9ac645806613710ab03da7e1841537d1d28a6e8f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58d1a3f4b08b561043e11ace221945bba26e02f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58d1a3f4b08b561043e11ace221945bba26e02f5"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
