)]}'
{"id":"openstack%2Fdesignate~850704","triplet_id":"openstack%2Fdesignate~master~I99d01a552d24123681e11f746c2e1195634efbfd","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-07-23 00:15:08.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I99d01a552d24123681e11f746c2e1195634efbfd","subject":"Fix sqlalchemy table_names DeprecationWarning","status":"MERGED","created":"2022-07-21 23:15:18.000000000","updated":"2022-07-23 01:33:29.000000000","submitted":"2022-07-23 01:31:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850704","meta_rev_id":"c05acd3d09bec4fa5aedb3ae1df6364a8d1d1682","_number":850704,"virtual_id_number":850704,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-23 01:31:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-07-22 14:06:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2022-07-23 00:15:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-22 10:40:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":2,"date":"2022-07-22 01:16:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2022-07-23 00:15:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-22 00:25:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-22 01:16:17.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-07-22 10:40:38.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-07-22 14:06:33.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"fedab41e81999f3d50c9bbe30f1811b8a1d91e92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-07-21 23:15:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa6f26cad3e680d9bc3e6e9bcfbf0a470ffd2b3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 00:25:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/339899fd745e4ecebc852ab36ff997b0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/25d913debc214efb8f736360339f5b83 : SUCCESS in 48m 57s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/110effb74a044b379bb939c82149404b : SUCCESS in 1h 00m 45s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/dad737b4b7fc4551960d7fbca6664302 : SUCCESS in 45m 02s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/a2c8d111184b4c2490722a571f239441 : SUCCESS in 45m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/659498af9c5b482b922969750a589d11 : SUCCESS in 41m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/53368a5bd0ce488e984fe5306a810c21 : SUCCESS in 1h 09m 04s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/77a409b7624e4499b314269648d403eb : SUCCESS in 1h 04m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8677625268704e9e892d6fe71a6a567e : SUCCESS in 43m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d47787bf3d1e4776832018286ba9452b : SUCCESS in 41m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8b3f6edf5d1430fb81b549d4d4781e9 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f64a034798244dd28eb3fc929e043363 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5149c2fdd464ca9a8a0d98a8e0bd98e : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81b5d86604e24a81b9f14f9c5187f9e0 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38f2bd9544c048e69d391a7f21232006 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cc8627905114894a475562387768090 : SUCCESS in 7m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f5a71c6eb4394901bde834885206d6c9 : SUCCESS in 47m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1eeb6a0bb5de42f38c64abcc56cdcbff : SUCCESS in 5m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"10e72fe1e80f1b7b16fd42e7a13dc48930f63d17","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-22 01:16:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0017540190c072d61e27ef6b25e796e9262ab751","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-07-22 10:40:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"31652edad25da3ccfd6a5b64daf382f08ab0ab3f","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-07-22 14:06:33.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d8632c605e40cdc07757a9f35e9ca4c77243e45b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-07-23 00:15:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8c3745b4e127cde308d9fbb5e5527c6662317692","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-23 00:15:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc5891cb2360003ae0bbfcf730d01a9e6f4f8842","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-23 01:31:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5161839fb31f4d5db6d5464e83ec6ef0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b8d5ae7e8be4453f90d85b94a0555471 : SUCCESS in 40m 02s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/82d44dee047049d180a2b2a603b2c1f2 : SUCCESS in 37m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8ee5b81a01b54ee5a4397309b0898dd6 : SUCCESS in 36m 50s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ebffc06b81dc4c788f5257ec00a3620b : SUCCESS in 53m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b7f3511d352d4ab181540b3234708e49 : SUCCESS in 34m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2707048cf3424a99b01f11dffa9dda39 : SUCCESS in 36m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d18df45ddf9482ab80749f288c0ced8 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/157ae1af4233459d8545e0cef844a7c9 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0522eda571a647ebac4c46464b8e38c2 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1214d677d6f044baa81be5f209d44e94 : SUCCESS in 8m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d6e14a59d8624647866f42866e8d3c04 : SUCCESS in 41m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a5854e171aa1288d5acb48b31943f0d1e9044a0f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-23 01:31:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c05acd3d09bec4fa5aedb3ae1df6364a8d1d1682","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-23 01:33:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c6c7da98276466f872f0c8a94566249\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f221380cc324990903df51aa0e4e0b1 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12e42df744807a5806e73726dec5993993fb41d5","revisions":{"12e42df744807a5806e73726dec5993993fb41d5":{"kind":"REWORK","_number":1,"created":"2022-07-21 23:15:18.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/04/850704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/04/850704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/04/850704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/04/850704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/04/850704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/04/850704/1"}}},"commit":{"parents":[{"commit":"d10056178ae13c25289f8cdd41c48062da2aab62","subject":"Cleaned up and fixed record objects and tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d10056178ae13c25289f8cdd41c48062da2aab62"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-07-21 23:12:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-07-21 23:15:17.000000000","tz":0},"subject":"Fix sqlalchemy table_names DeprecationWarning","message":"Fix sqlalchemy table_names DeprecationWarning\n\nThis patch updates Designate to use the new method of getting the table names from a sqlalchemy engine. It replaces the use of tables_names() with get_table_names from the sqlalchemy inspector.\nThe DeprecationWarning was:\nSADeprecationWarning: The Engine.table_names() method is deprecated and will be removed in a future release.  Please refer to Inspector.get_table_names(). (deprecated since: 1.4)\n  self.assertEqual(table_names, self.storage.engine.table_names())\n\nChange-Id: I99d01a552d24123681e11f746c2e1195634efbfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/12e42df744807a5806e73726dec5993993fb41d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/12e42df744807a5806e73726dec5993993fb41d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
