)]}'
{"id":"openstack%2Fdesignate~851826","triplet_id":"openstack%2Fdesignate~master~If9599c62abe05e4e7514e81acbc86318880a61ac","project":"openstack/designate","branch":"master","topic":"testing","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-08-03 06:38:40.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If9599c62abe05e4e7514e81acbc86318880a61ac","subject":"[DNM] Testing","status":"ABANDONED","created":"2022-08-03 04:31:41.000000000","updated":"2022-08-03 06:38:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8adca91537fd75b3e3620b2c7bda5bee8b1447ef","_number":851826,"virtual_id_number":851826,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-08-03 06:34:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-03 06:34:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fcbf49952509c8aa5032c7d5399340dc30338c47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-03 04:31:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8396429e4602df79419255290e5953092ed55c35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-03 05:07:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"90a8925e9a27aa9206993841af07e1061c3c7cf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-03 05:35:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"859a7dfa90dc9ef0424462cfade3af57478b2a9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 06:34:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/78c3158a32024cc39eeaed5eb6e6f200\n\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/19d7bc5eeb49424f9e311b5a7e7435ee : FAILURE in 54m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2922bf7162ba4289be114854f76cff8c : FAILURE in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/269dbcb19f4a4f74a5437769ab33047b : FAILURE in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/693833a675404160979bafcd124dd798 : FAILURE in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f127687a8b948eeb1d3e8bc76c465d1 : FAILURE in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44c2808db9f64f0a9d94fa1757ba295d : FAILURE in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/053452653c8948929bb0ae5617648e51 : SUCCESS in 6m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/00a69de212f1473facce17efb0aba53e : SUCCESS in 31m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/498d74201396487ca8d329a3057ae2dd : FAILURE in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8adca91537fd75b3e3620b2c7bda5bee8b1447ef","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-03 06:38:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"29c3305cdbd0b8cef76f9fd12102786e9f549f5e","revisions":{"83ea379bc5d47e6cdd0838d34bd6961008e8eebf":{"kind":"REWORK","_number":1,"created":"2022-08-03 04:31:41.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/26/851826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/851826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/851826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/851826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/851826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/851826/1"}}},"commit":{"parents":[{"commit":"64e3ac9c205b0330c42392ff8b2c2c047f00d50b","subject":"Switch Designate to Alembic database migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/64e3ac9c205b0330c42392ff8b2c2c047f00d50b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-03 04:31:24.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-03 04:31:24.000000000","tz":-420},"subject":"[DNM] Testing","message":"[DNM] Testing\n\nChange-Id: If9599c62abe05e4e7514e81acbc86318880a61ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/83ea379bc5d47e6cdd0838d34bd6961008e8eebf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/83ea379bc5d47e6cdd0838d34bd6961008e8eebf"}]},"branch":"refs/heads/master"},"f38e19f5e51e43a5d899ed1ac9e2e5a8efa886a8":{"kind":"REWORK","_number":2,"created":"2022-08-03 05:07:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/26/851826/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/851826/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/851826/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/851826/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/851826/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/851826/2"}}},"commit":{"parents":[{"commit":"64e3ac9c205b0330c42392ff8b2c2c047f00d50b","subject":"Switch Designate to Alembic database migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/64e3ac9c205b0330c42392ff8b2c2c047f00d50b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-03 04:31:24.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-03 05:06:59.000000000","tz":-420},"subject":"[DNM] Testing","message":"[DNM] Testing\n\nChange-Id: If9599c62abe05e4e7514e81acbc86318880a61ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f38e19f5e51e43a5d899ed1ac9e2e5a8efa886a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f38e19f5e51e43a5d899ed1ac9e2e5a8efa886a8"}]},"branch":"refs/heads/master"},"29c3305cdbd0b8cef76f9fd12102786e9f549f5e":{"kind":"REWORK","_number":3,"created":"2022-08-03 05:35:12.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/26/851826/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/851826/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/851826/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/851826/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/851826/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/851826/3"}}},"commit":{"parents":[{"commit":"64e3ac9c205b0330c42392ff8b2c2c047f00d50b","subject":"Switch Designate to Alembic database migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/64e3ac9c205b0330c42392ff8b2c2c047f00d50b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-03 04:31:24.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-03 05:35:03.000000000","tz":-420},"subject":"[DNM] Testing","message":"[DNM] Testing\n\nChange-Id: If9599c62abe05e4e7514e81acbc86318880a61ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/29c3305cdbd0b8cef76f9fd12102786e9f549f5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/29c3305cdbd0b8cef76f9fd12102786e9f549f5e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
