)]}'
{"id":"openstack%2Fdesignate~852612","triplet_id":"openstack%2Fdesignate~master~I5d78939eb43476e6c93235e131086bc8537b0e03","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-08-13 00:54:30.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I5d78939eb43476e6c93235e131086bc8537b0e03","subject":"Simplified backend unit tests","status":"MERGED","created":"2022-08-11 05:43:57.000000000","updated":"2022-08-13 02:15:15.000000000","submitted":"2022-08-13 02:12:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852612","meta_rev_id":"d0572c2d6b79dd083067157fd7167f33d5d32b7a","_number":852612,"virtual_id_number":852612,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-13 02:12:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-08-11 16:22:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-13 00:54:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-11 06:49:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-11 16:22:43.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"6464b694a84084eba8dd829830abfd0176862b2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-11 05:43:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3f3a24a315b047d1e5ebbf0d483a8ab3718fdf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-11 05:46:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b9f2f1cbb862b42c25b2075cb1f2aeca2321e276","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-11 05:49:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ddcd880fcb33acb369c63e416775ced7a0c48309","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-11 05:49:28.000000000","message":"Topic cleanup removed","accounts_in_message":[],"_revision_number":3},{"id":"f8ff551f42006a3b26a8ccc813de34ef3f42d32a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 06:49:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd27df5d420341a6923e0a68b9a174c0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/892da4a038c646988ede374fa1923d0d : SUCCESS in 40m 45s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1ab51f66c175498e9a4501f641e898fa : FAILURE in 25m 43s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a48daf389984d2c90eed0f8e47bd008 : FAILURE in 11m 36s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1fcaf4206f1f455aa7881c9dcd5b1dfb : SUCCESS in 36m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0b80a8760f8a41a78159ac0969a7828d : SUCCESS in 37m 40s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ecfa7dfab71f4fcbb594bcd3557f115a : SUCCESS in 58m 51s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/37db373c5c9f4ecdbbc4082adb1a891f : SUCCESS in 58m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f3acc87e1cc340c5aa0052e7bbe9551d : SUCCESS in 39m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7ba9308d30e14ac5bbb20f3938eb43da : SUCCESS in 56m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e419c5ad4f9040389f01b21ecc5ad814 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f958cbaa19b249efb54403683bbf16b5 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c48c426d4b74aadaa988db091b9d2a8 : SUCCESS in 7m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cad9bb5d5e21451a807ca9f037e400ad : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64b14831e23a4284bd4253accac433dc : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83f81a0c38aa47379ab89471760968c6 : SUCCESS in 7m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fbf3bd88b52f47b09959ba223c4e99f6 : SUCCESS in 37m 53s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/aee46ca43a644a0a9afd611d204fe572 : SUCCESS in 5m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3bbb7763b49da03ba05c922ac79e1a927bbd8e1d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-11 16:22:43.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aea9898514784cc13b290d3284cd82361629cc0f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-13 00:54:30.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6b1972576de0dafe304a03c32d486f218d878e55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-13 00:55:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b666ecaef02b453549d375a844b02e5bd4e3d79b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-13 02:12:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a028c159888459992a150daacff0f5f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cc8a61d3479f469e9732d9d030d33bed : SUCCESS in 47m 32s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/82c56da30cc14c4a872cdb2f470cec6e : SUCCESS in 46m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7875a1f94c4d42aaa426997ff6b2cd10 : SUCCESS in 45m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/10d7b99e5ada43dc90b1854b762a48f4 : SUCCESS in 1h 11m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d11d7242448345fdab9641193f27064a : SUCCESS in 46m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4410c09206984d5bbf7a8442e1615414 : SUCCESS in 46m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/404727e27422472f864e8fb60a6437ad : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b09de1453f744a18a84e97d30e2fdeaa : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/145b9036bfb54c249e911f2db3bb1a12 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fe516f04cab4dc69c65774388be863c : SUCCESS in 6m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a05c552c93d64ccb8ce047b69d2da077 : SUCCESS in 30m 42s","accounts_in_message":[],"_revision_number":3},{"id":"c89835ef3683614347c7c244c376909c294babfa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-13 02:12:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d0572c2d6b79dd083067157fd7167f33d5d32b7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-13 02:15:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46741adebc2c41f1b7dad6fbe6411ec3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7756a8e5c9d14361b6deafa7fd066e6e : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"43428e48d0470f1e6fbd96cb51ccb97c20e12e14","revisions":{"30a7d6b2b4034561d718275b187ddbe5db9ff214":{"kind":"REWORK","_number":1,"created":"2022-08-11 05:43:57.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/12/852612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/12/852612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/852612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/852612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/852612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/12/852612/1"}}},"commit":{"parents":[{"commit":"f3c29e8bceea832e1cb787740fa676b10a410848","subject":"Merge \"Fixed issues with __repr__ and __str__ on objects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f3c29e8bceea832e1cb787740fa676b10a410848"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-11 05:40:16.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-11 05:43:30.000000000","tz":-420},"subject":"Simplified backend unit tests","message":"Simplified backend unit tests\n\nThese tests have no dependencies on the Designate Test class\nand by moving to the oslo test base class instead we reduce the\nruntime of these tests from ~5s to ~150ms.\n\n- Added some basic tests for the fake backend (aka noop)\n- Switched from designate.tests.TestCase to oslotest.base.BaseTestCase\n\nChange-Id: I5d78939eb43476e6c93235e131086bc8537b0e03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/30a7d6b2b4034561d718275b187ddbe5db9ff214"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/30a7d6b2b4034561d718275b187ddbe5db9ff214"}]},"branch":"refs/heads/master"},"97d15d45797365dba61bc3721c9aed1315d86b68":{"kind":"REWORK","_number":2,"created":"2022-08-11 05:46:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/12/852612/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/12/852612/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/852612/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/852612/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/852612/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/12/852612/2"}}},"commit":{"parents":[{"commit":"f3c29e8bceea832e1cb787740fa676b10a410848","subject":"Merge \"Fixed issues with __repr__ and __str__ on objects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f3c29e8bceea832e1cb787740fa676b10a410848"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-11 05:40:16.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-11 05:46:01.000000000","tz":-420},"subject":"Simplified backend unit tests","message":"Simplified backend unit tests\n\nThese tests have no dependencies on the Designate Test class\nand by moving to the oslo test base class instead we reduce the\nruntime of these tests from ~5s to ~150ms.\n\n- Added some basic tests for the fake backend (aka noop)\n- Switched from designate.tests.TestCase to oslotest.base.BaseTestCase\n\nChange-Id: I5d78939eb43476e6c93235e131086bc8537b0e03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/97d15d45797365dba61bc3721c9aed1315d86b68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/97d15d45797365dba61bc3721c9aed1315d86b68"}]},"branch":"refs/heads/master"},"43428e48d0470f1e6fbd96cb51ccb97c20e12e14":{"kind":"REWORK","_number":3,"created":"2022-08-11 05:49:19.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/12/852612/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/12/852612/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/852612/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/852612/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/852612/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/12/852612/3"}}},"commit":{"parents":[{"commit":"f3c29e8bceea832e1cb787740fa676b10a410848","subject":"Merge \"Fixed issues with __repr__ and __str__ on objects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f3c29e8bceea832e1cb787740fa676b10a410848"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-11 05:40:16.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-11 05:49:13.000000000","tz":-420},"subject":"Simplified backend unit tests","message":"Simplified backend unit tests\n\nThese tests have no dependencies on the Designate Test class\nand by moving to the oslo test base class instead we reduce the\nruntime of these tests from ~5s to ~150ms.\n\n- Added some basic tests for the fake backend (aka noop)\n- Switched from designate.tests.TestCase to oslotest.base.BaseTestCase\n\nChange-Id: I5d78939eb43476e6c93235e131086bc8537b0e03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/43428e48d0470f1e6fbd96cb51ccb97c20e12e14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/43428e48d0470f1e6fbd96cb51ccb97c20e12e14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
