)]}'
{"id":"openstack%2Fdesignate~853037","triplet_id":"openstack%2Fdesignate~master~I44f79dd26421b469692869841d73056839285598","project":"openstack/designate","branch":"master","hashtags":[],"change_id":"I44f79dd26421b469692869841d73056839285598","subject":"remove unicode prefix from code","status":"ABANDONED","created":"2022-08-15 00:34:28.000000000","updated":"2022-08-15 22:18:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dd0a82631cad646e81102728727a181bafd1e53c","_number":853037,"virtual_id_number":853037,"owner":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-08-15 00:48:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-15 00:48:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a449ab8713891f715a87dc4c97ee421358febac3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-15 00:34:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4813e78ffc7fab5ec03a94198d01f63c776a2ab4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-15 00:35:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0b546190d57bf6a9de7f5c8edd60b16f620f22d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 00:48:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6cef1c7c3ae4a698e87dc6c1394b346\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad58e2694878442f90bb7c625d7a52b2 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27eb29ae1bd542a0b4ae069a92c97489 : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbf4c4e67beb40b5a03434d69928404d : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":2},{"id":"dd0a82631cad646e81102728727a181bafd1e53c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-15 22:18:59.000000000","message":"Abandoned\n\nDuplicate to https://review.opendev.org/c/openstack/designate/+/826415","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"62a1cdf4dd5d55c7957ae5e543df4247bc36b379","revisions":{"ae8f0729b363e9b239a8975a96bdf7ec5447ca33":{"kind":"REWORK","_number":1,"created":"2022-08-15 00:34:28.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/37/853037/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/37/853037/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/37/853037/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/37/853037/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/37/853037/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/37/853037/1"}}},"commit":{"parents":[{"commit":"63563a50983c85df852654e908963eeb7331e3c8","subject":"Merge \"Simplified backend unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/63563a50983c85df852654e908963eeb7331e3c8"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-15 00:33:57.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-15 00:34:03.000000000","tz":480},"subject":"remove unicode prefix from code","message":"remove unicode prefix from code\n\nChange-Id: I44f79dd26421b469692869841d73056839285598\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ae8f0729b363e9b239a8975a96bdf7ec5447ca33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ae8f0729b363e9b239a8975a96bdf7ec5447ca33"}]},"branch":"refs/heads/master"},"62a1cdf4dd5d55c7957ae5e543df4247bc36b379":{"kind":"REWORK","_number":2,"created":"2022-08-15 00:35:51.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/37/853037/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/37/853037/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/37/853037/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/37/853037/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/37/853037/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/37/853037/2"}}},"commit":{"parents":[{"commit":"63563a50983c85df852654e908963eeb7331e3c8","subject":"Merge \"Simplified backend unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/63563a50983c85df852654e908963eeb7331e3c8"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-15 00:33:57.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-15 00:35:32.000000000","tz":480},"subject":"remove unicode prefix from code","message":"remove unicode prefix from code\n\nChange-Id: I44f79dd26421b469692869841d73056839285598\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/62a1cdf4dd5d55c7957ae5e543df4247bc36b379"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/62a1cdf4dd5d55c7957ae5e543df4247bc36b379"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
