)]}'
{"id":"openstack%2Fdesignate~853123","triplet_id":"openstack%2Fdesignate~master~I5dca5495e1d4510682b6244d2e7ca707308927d8","project":"openstack/designate","branch":"master","topic":"tested","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-08-23 04:03:40.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I5dca5495e1d4510682b6244d2e7ca707308927d8","subject":"Add additional backend tests","status":"MERGED","created":"2022-08-16 05:28:40.000000000","updated":"2022-08-23 05:12:40.000000000","submitted":"2022-08-23 05:11:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853123-tested","meta_rev_id":"4937a054d63ead698e84271cd0ff7d448117ff1e","_number":853123,"virtual_id_number":853123,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-23 05:11:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-08-17 21:22:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2022-08-23 03:06:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-23 04:03:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-16 06:49:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-17 21:22:55.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2022-08-23 03:06:38.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"a3fc2c13e42b7135c51cd61a261e7c69fad14244","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-16 05:28:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a813930693d6fa314b5bf2886c4c56874b58562","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-16 05:35:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f4a8cdaa038b2caf06534231d71cfe95c6f00935","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 06:49:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc5549506d0648f4a76a19589a7e22af\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c9ef3b117bf7456a98dca68f1fc9a3dd : SUCCESS in 48m 35s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d3cde55359244042be6c83d050c3e524 : SUCCESS in 1h 02m 13s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/27a4ab6926dc4470be5645a7fa850a51 : SUCCESS in 1h 04m 22s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4426932d1b544446933e5b7e3c92f448 : SUCCESS in 47m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9af3997b320a4eb09346f4ea1ce64d1c : SUCCESS in 49m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/94f61128fc104299b9168f9438f7e0fe : SUCCESS in 54m 17s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b7a5eba87b9749618c5b0b957aa8c314 : SUCCESS in 1h 08m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/baa68e049ec34242af33174a2eef2538 : SUCCESS in 52m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4860d290e31344fea9dd42fe1a0548a2 : SUCCESS in 44m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bda4654baa5f48dd81b1ebe1525e31c2 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b302326afe63435fa620c5d2c3a604bf : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4794f8039e3e468d93fe838876fb3fe1 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74c54ebfe6ae49389615f489671ab2f3 : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2be42ad4db0f43d49c5297f64ba5c58f : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25c9d739009c451f8384c4f6d0771d7a : SUCCESS in 6m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a4319042f5104d96abf5a0cccb20f84e : SUCCESS in 34m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c9ac39abf6dc474e8fb1eaf8f1d2e7d8 : SUCCESS in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"280ce10272d7d248237afcf9af31b91d563848de","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-08-17 21:22:55.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c032d949d568f4f85feb58fb80ad9630c2012ee","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-23 03:06:38.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"17a04c87ad63dfad92ac3a5df151b3edd01d1e49","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-23 04:03:40.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5964a162a8354859b7bf0a9968d4459098d25bd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 04:04:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d5d107028a1fc884da3b5770a0cc80da7d0b250c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 05:11:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79b2e4ff458545c0a774832473113a16\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d204ebb5d0c24674a66cc76f21207254 : SUCCESS in 50m 32s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/0754af51d3b540ba9caff50d8981cbd3 : SUCCESS in 48m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bfc2ace10fea445d92142040ac410acc : SUCCESS in 45m 56s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/703dfadcd4084884a158fc5b9951a56b : SUCCESS in 1h 02m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/52a11afb30b64efba7a3a00aad4d1bb1 : SUCCESS in 47m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e9f4c9864c1d4c9599102bc7fddf9460 : SUCCESS in 47m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1917ffd7f5e4df4977c1220d9e6c652 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d2c9f37f9e94954914346c9adb18b1f : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdee0bf8cb5241599f5a33fce3166336 : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7246912bd584f469569d5914dd511dc : SUCCESS in 7m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/155d0ce85ca84987a85949ba9c111a9e : SUCCESS in 32m 56s","accounts_in_message":[],"_revision_number":2},{"id":"287e8863d3bec35eff49554cd870887d5baf0d21","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 05:11:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4937a054d63ead698e84271cd0ff7d448117ff1e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 05:12:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e199068c3093406da96509e673e7cd22\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da6b0ec69811499eb0f6016282b74984 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f237b0a63c76ef3a3558b446c61382785d1fe41a","revisions":{"c3338ba486141627fa2266a9bac154857fd50022":{"kind":"REWORK","_number":1,"created":"2022-08-16 05:28:40.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/23/853123/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/853123/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/853123/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/853123/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/853123/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/853123/1"}}},"commit":{"parents":[{"commit":"927e34ee5a53771530e78bcbbd74549c632b539a","subject":"Merge \"Add basic get backend test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/927e34ee5a53771530e78bcbbd74549c632b539a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-16 05:28:02.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-16 05:28:02.000000000","tz":-420},"subject":"Add additional backend tests","message":"Add additional backend tests\n\nChange-Id: I5dca5495e1d4510682b6244d2e7ca707308927d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c3338ba486141627fa2266a9bac154857fd50022"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c3338ba486141627fa2266a9bac154857fd50022"}]},"branch":"refs/heads/master"},"f237b0a63c76ef3a3558b446c61382785d1fe41a":{"kind":"REWORK","_number":2,"created":"2022-08-16 05:35:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/23/853123/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/853123/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/853123/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/853123/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/853123/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/853123/2"}}},"commit":{"parents":[{"commit":"927e34ee5a53771530e78bcbbd74549c632b539a","subject":"Merge \"Add basic get backend test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/927e34ee5a53771530e78bcbbd74549c632b539a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-16 05:28:02.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-16 05:35:42.000000000","tz":-420},"subject":"Add additional backend tests","message":"Add additional backend tests\n\nChange-Id: I5dca5495e1d4510682b6244d2e7ca707308927d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f237b0a63c76ef3a3558b446c61382785d1fe41a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f237b0a63c76ef3a3558b446c61382785d1fe41a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
