)]}'
{"id":"openstack%2Fdesignate~853638","triplet_id":"openstack%2Fdesignate~stable%2Fyoga~I7345d7ef505420767209ba037e3c8930a282d03f","project":"openstack/designate","branch":"stable/yoga","topic":"bug/1986733-stable/yoga","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-08-23 21:08:23.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I7345d7ef505420767209ba037e3c8930a282d03f","subject":"Fix designate-manage pool update MissingProjectID","status":"MERGED","created":"2022-08-18 22:35:35.000000000","updated":"2022-08-23 22:23:04.000000000","submitted":"2022-08-23 22:21:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853638-bug/1986733-stable/yoga","cherry_pick_of_change":853370,"cherry_pick_of_patch_set":1,"meta_rev_id":"5aa8dc3a85480bf5658640a2b2c3216add111fbb","_number":853638,"virtual_id_number":853638,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-23 22:21:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-08-19 12:50:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2022-08-23 21:08:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-19 03:04:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2022-08-23 21:08:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-18 23:51:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-19 03:04:48.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-08-19 12:50:34.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"4e57dfeb40bd2cd043ba44a86971f4976360096e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-18 22:35:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"eb37c029def7ba8c41d53c632a9aa65820486f9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 23:51:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/70530341e99a4a9ca9ba0c89d0e62305\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e4bf1989542c4ca09d73b6d66181124c : POST_FAILURE in 50m 25s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/b81311a1b8f04be4b34f8d9fe767b231 : RETRY_LIMIT in 22m 09s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c94d152dbb494147b21e86dd38db2d6e : SUCCESS in 58m 42s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/cede4822c9f149eeb2741ee5c999ec6b : SUCCESS in 53m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5c6a92ef32bf412cb3c31e6d64a30685 : SUCCESS in 38m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1f25e74dcea8462ba1b6623afdcbba69 : SUCCESS in 1h 10m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/142f0bb0d7924a0585212cc660fad774 : SUCCESS in 35m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6aa6699bf1af4f3f93f39132ed655e70 : SUCCESS in 54m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e95083069a6845c08808d135c98b1810 : SUCCESS in 5m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd4a1e86e5bc494194d7e9104ee283cf : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58c1ff6eace349a88878bafe223bcd95 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4693c6e67e62410bbc63cb8db62be01e : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f5b238748214afc84f99497347b41b2 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7fba6fb489b45e78619c2141051fd73 : SUCCESS in 11m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3f6eb9ecd04461ca73852f2e2911fda : SUCCESS in 4m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a7e769ea06cd4d69bdea4e350f5b3a9b : SUCCESS in 52m 10s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/96a8ff8e9a2f49f486ae290d125e1466 : SUCCESS in 4m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"887b0cd4eb038c37fc14dd1edc6240078c98df35","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-19 03:04:48.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0409f4f783883bf90c2bf160313c2d85ce9edf1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 04:14:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0ab1e475b2046d18493d37450dbb412\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b0a894880b2748f1b2e875ebeee58919 : SUCCESS in 59m 01s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/f316c3d20488468d8e23d1b7d1b3af86 : RETRY_LIMIT in 12m 13s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/84da86614a93449cb1252731492d4d42 : SUCCESS in 1h 06m 44s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/bd6fb11262824ed49bd933efbb1bfd76 : FAILURE in 59m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7cd80c4fdfa74194b89a2797a9be8f0e : SUCCESS in 54m 43s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/24a90f93c49247c4a140856199bb129b : FAILURE in 44m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6621e25c3c2043cda8004d7e9a919b06 : SUCCESS in 40m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c82f8dd9900b4869803dbddccb37c844 : SUCCESS in 1h 06m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9d41a52f2b34acbafc62852629cc1d5 : SUCCESS in 5m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4cd422d48f84d89a27762a1401e0aa7 : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8db893c0b29848b982e0fb953556e7eb : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1070a7c003fb4c0eac5352f372dc464c : SUCCESS in 7m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3598120048c4424085c825c3d59d0a0e : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b9ff2086aa04d8fb18bfa14c574f7ad : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8dcc70a48694a64ba06d07fe55f0da9 : SUCCESS in 4m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/086bbef0f87f40959a4b4efdaa2c20f4 : SUCCESS in 49m 09s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8cc0fadfe9724b7facdfbf91556c1257 : SUCCESS in 5m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c791c66fc99165c024c228f3c19bce6d79feccd","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-19 04:33:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f53b94b488679070060a9fffc7b59978eab390b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 05:47:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1e8541e5b2c4f3788ff6eab3e65d52a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a384abdbfab04eeb9155552b55289b7e : SUCCESS in 56m 46s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/bc15306671094173aa6dd8c22d576562 : RETRY_LIMIT in 23m 48s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1e3236516fb42509c3c921ef578d0d3 : SUCCESS in 59m 36s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/6036ba5423e64bb59c03d32a1f973c72 : SUCCESS in 52m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2434bd4b23fe434cbcdd3e4daf7a536d : SUCCESS in 53m 46s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/75563e5360584a54bf92a25e6f58afe1 : SUCCESS in 1h 09m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8cd3104f3e0d4e04b5a65911af725131 : SUCCESS in 56m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cfb980ee516648759e559b2f6de0d927 : SUCCESS in 1h 01m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b04701449b324f3782fae5892f33c296 : SUCCESS in 6m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/274201478eb44c2e8b93fda0f030906a : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cc2c1ee5d4c4dbc85adebc83ba68568 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dea9e7f230b8444897a3fc533d606302 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6614beba7c6b4684808783e35b1bcba7 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43eabc6d5f9b4c9999c51e5b1f17cad4 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbb3501e69664f05bcdb065d0f17d923 : SUCCESS in 4m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9b7f570c542e4c17831792aa45d412fc : SUCCESS in 35m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/57ddd73dcc0e48b7ad607527649f93c1 : SUCCESS in 4m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f6cb8ef3481c984ea3c55ad332e2d22adc34a910","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-08-19 12:50:34.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"15f2f0744ceba159a677a60e4c20c68256f20ae8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-23 21:08:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d8c882a1f0745a35f5644400d74c3dba1b988f99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 21:09:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f989a855684d4b2d76b38640e481f922d5445fb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 22:21:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e118df2be34456e9a08abc810a7268b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f732f1dbe6ba40c2884ebfdf3b58debf : SUCCESS in 1h 05m 24s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ddc6e0c6386542a5b922a96138b0cad9 : SUCCESS in 56m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/265692df0e484b90b300fe90dbbd356d : SUCCESS in 51m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d459157d916a46a28aa17da777f856f2 : SUCCESS in 1h 11m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c70afd73d186487597bb59675027be00 : SUCCESS in 55m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1cc8ea0f5d5b4d1597731eb66fb843e7 : SUCCESS in 56m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25922c5516f34398991e3f35dccf53c7 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bccafa8c659e4b06bc92a2564ec91763 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2aeda07ecdcd4768b7ae41d7a9a8391c : SUCCESS in 9m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11e78f245468427d8c48eafd21959840 : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/077f9de62ed14470ad1ca16fba62f606 : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73c8bdce035a4208b9554ffde30c5f87 : SUCCESS in 6m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7b730b14eda048e2950787a3edd0e42e : SUCCESS in 52m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0df29aa12c22ed365df70a0ac44fd1b847ad4bf7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 22:21:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5aa8dc3a85480bf5658640a2b2c3216add111fbb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 22:23:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c3fceff80bb47bd90436315bc90079a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e81d68570e7d43ea87355a78c0ffa378 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3e3367fac1f4652a8d939a26748191a : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9d96cc843aa54582db52b394ad8223a1939ac05e","revisions":{"9d96cc843aa54582db52b394ad8223a1939ac05e":{"kind":"REWORK","_number":1,"created":"2022-08-18 22:35:35.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/38/853638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/853638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/853638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/853638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/853638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/853638/1"}}},"commit":{"parents":[{"commit":"2dc40a15894d2c2010b7176ff6f1dff0c8621503","subject":"Minmum TTL value is zero","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2dc40a15894d2c2010b7176ff6f1dff0c8621503"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-08-16 22:17:39.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-08-18 22:35:35.000000000","tz":0},"subject":"Fix designate-manage pool update MissingProjectID","message":"Fix designate-manage pool update MissingProjectID\n\nThis patch fixes a bug where adding an additional pool to designate using the \"designate-manage pool update\" command may fail with an exception:\n\ndesignate.exceptions.MissingProjectID: A project ID must be specified when not using a project scoped token.\n\nThere was an extra check added as part of the scoped token work that a project ID must be provided when creating pools. This was incorrect as pools are still valid with a None project ID as they are a system resource and not tied to a specific project.\n\nThis patch removes that check, but retains the RBAC check for and \"admin\" token.\n\nCloses-Bug: #1986733\nChange-Id: I7345d7ef505420767209ba037e3c8930a282d03f\n(cherry picked from commit a03c4657c2bf957fef035807ef7df732f7a1c994)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9d96cc843aa54582db52b394ad8223a1939ac05e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9d96cc843aa54582db52b394ad8223a1939ac05e"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
