)]}'
{"id":"openstack%2Fdesignate~853781","triplet_id":"openstack%2Fdesignate~master~I6bfe074ec9db9bd95021771b8ffe988f1f23e2be","project":"openstack/designate","branch":"master","topic":"clean_up_pools","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-08-23 04:20:43.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-08-23 06:09:25.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I6bfe074ec9db9bd95021771b8ffe988f1f23e2be","subject":"Clean up manage pools and add additional testing","status":"MERGED","created":"2022-08-21 11:04:10.000000000","updated":"2022-08-23 07:28:40.000000000","submitted":"2022-08-23 07:27:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853781-clean_up_pools","meta_rev_id":"61fdbd4779e2be9df67e455a50bdd72e902f4c6a","_number":853781,"virtual_id_number":853781,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-23 07:27:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-23 04:20:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-08-23 06:09:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-21 12:31:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-23 03:25:14.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"c5e6019eb9afc38cab785791e85523b626150f09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-21 11:04:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88a3c39081cd64e014b6cf1d7a63b3f284f1db0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-21 11:09:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2b6eef5eb6a2cc36365702ca3d5e5909891630b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-21 11:11:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f95a3eaacde0a519591d7dd46cdf3dc38815e8af","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-21 12:28:22.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f5bd4a2180b389ff8f8624bda0ef95dd58840f6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-21 12:31:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7088b9d5df31402cb328e23c1b4dc6f8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3bdb717481ce46488e5dee5070d17ad7 : SUCCESS in 55m 44s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/76a28fc5e8b442b0b14660c4001fc456 : SUCCESS in 1h 03m 06s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2f3abb29ee54212af9c1d3981af82af : SUCCESS in 54m 53s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1dc7330bd0434b3092c2ee152fe048c3 : SUCCESS in 56m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/296fcc4857ab4f5b86b61af9a104f40b : SUCCESS in 47m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ac6c8b559dd148f5b9eb889d16a7c40d : SUCCESS in 1h 15m 30s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ddb4d9a7cc4d4a8ca9c35cb4be939fa7 : SUCCESS in 1h 08m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/da0e84b87a544bd3b5004ed2999abf46 : SUCCESS in 52m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/78e60111aa2643d4a525d191ae0b13eb : SUCCESS in 49m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0462a3727094120a572fc7bb1dddc71 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16223cbe87314f80957a3cf6ef51f8ec : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fce79b22775840bfb1ac6d00591fdb5e : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2bbff1ea9c04dae82faff1ce2dec993 : SUCCESS in 7m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30a70583401e4f77a0ce314b458268e6 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3ff692535f14255a7140c71f96d6e89 : SUCCESS in 9m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7c3616e66d644d7ab889bf7c2c25f8f3 : SUCCESS in 44m 58s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0deb8c55a6d54e3d981f47fc64fbb95b : SUCCESS in 4m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"42e302c2adb308ae24d4e1ed881d1a3344b6f226","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-21 22:42:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4cda79f31be20de57acac912e7cd13a9df33de81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 00:08:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53899fd4d26a40c1ac9d25df196f29f5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/12bb372d80394ff5adb332eea2da0192 : SUCCESS in 50m 31s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/16b26c39286d42f7ac17402f4db65c0d : SUCCESS in 1h 13m 19s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/169b5ca97a3448399e05054f3dbdfc12 : SUCCESS in 57m 48s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/222046dd072d41b48a757785e4e3a8b4 : SUCCESS in 58m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6d451db9f7ad41baaec33f33b7bc4110 : SUCCESS in 50m 41s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/97c5d701b8154a7faf4ad65b82e532a2 : SUCCESS in 1h 15m 41s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/20324dcfd52644b0a928978472183d90 : SUCCESS in 1h 13m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c3a153beb2834951bf520f8c6fa1a850 : SUCCESS in 57m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a15363976065495b8eb40cfbf25427d2 : SUCCESS in 50m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3a1009052504deb919040697ce1c3df : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95d0bed3c1b147efa6a0584069c91795 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bf65498f977429888e58d0a3fde8f76 : SUCCESS in 7m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52aa0bd9ce1f44fe8b016c5d8e0e9adb : SUCCESS in 7m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c9565e2c35a4480a84b88f0aa31a4f3 : SUCCESS in 9m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98d7f9e575ff42239696c5200eb2aa6a : SUCCESS in 11m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a90f636d7b2b48c4b33fc00e450c2740 : SUCCESS in 36m 40s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/bd3f8ac99e624bae939d7b1f9b7e4951 : SUCCESS in 6m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4d4284f7af80482043152c29625f8fa4c9c788ba","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-23 03:25:14.000000000","message":"Patch Set 4: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4825058e50dc5dd003f95bf93868a92d8f04e4cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-23 04:19:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f280b537ce5dc8a61e390fd45c0792a08e822641","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-23 04:19:53.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"aa4d6cfc4e499d7d4d37e754e889f399961fe560","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-23 04:20:43.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b360e5eb8fc1e9fb737e08ff419fa400dc2a2176","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 05:27:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/381b58c683f24107bc8f7b7fb093beb3\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/86ad4b0eae9540c1badb1cf2ae6b6fc5 : SUCCESS in 37m 08s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2d15f8f580744f54981fc1754cc91af9 : SUCCESS in 49m 13s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/25fab8dda8c6491087cb1fbfca9e3a1a : SUCCESS in 44m 03s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/9d0a452b55e84b04a64080d38787b836 : SUCCESS in 51m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d79fd4e1397b43d8b8b50927e716a210 : SUCCESS in 42m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/234f89fc5e00459ea448fc7707826d42 : SUCCESS in 1h 06m 56s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/223f7766a2844db181ad2c3c28052a06 : SUCCESS in 1h 01m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e463650664034bf99048fcccf2d2da89 : SUCCESS in 46m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ca6371d0f2994619803566469a4e4276 : SUCCESS in 49m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23479551a6644d2996efe6d2e093c682 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ce4335be60843b2bdc874c42381ff69 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38e8a7c1de76432e8eb52143e290dfcc : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed6056b009b94758be3b2b8e85603b81 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1591ee111b064728ab9dec710d3f06a1 : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21a77be7a62642e2ba81d95448c984ce : SUCCESS in 6m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8d719ecf97fc42faa5cdc9db7aecc71e : SUCCESS in 42m 30s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7b162947c6a14100911f3a672113bcd3 : SUCCESS in 3m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2a368cea623fb6eac87d7bfd22e0c68960d12dd1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-23 06:09:25.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"76f2a8d2560f899d315e058002c20549a4471804","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 06:10:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b46c4c7bc41c4331ec72ce2c82cf22db0064f1ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 07:27:38.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5276534fd2424d9397a98bb0032b4d69\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0392aaad33ed48a4bc5522345d848df6 : SUCCESS in 54m 50s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/363727c651b246798edfa9bbc459d387 : SUCCESS in 55m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/614b91857b494693a0c4f88690130b2e : SUCCESS in 51m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/116e64f06d144017a23d59724b309dea : SUCCESS in 1h 11m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c136d520df2f4983b08c3d30e26683b6 : SUCCESS in 50m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f4a2d948c5ad4154a6a5a81913f42e92 : SUCCESS in 51m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2e696a5c8154f2b9355f8e9b73db397 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1933ec426b54a4d8f7ac72320070e24 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/225cd1b6acab40d18038b4d521336b8a : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46f2a90d3c8249ecbc477b74d2071e8e : SUCCESS in 8m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d900b1a6d21a4207be9e3b1a4a8cbc41 : SUCCESS in 33m 09s","accounts_in_message":[],"_revision_number":5},{"id":"fc365c16e310ebb8f43bbedbfe2638739375662c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 07:27:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"61fdbd4779e2be9df67e455a50bdd72e902f4c6a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 07:28:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47407bf045fe472c91a783ed3a2b9a43\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfdf14fab9304b46ad596b42321d5c6d : SUCCESS in 40s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e65b32a6e6755cc0c9f286ab58ae20742d6d18bc","revisions":{"16f56897f15d09b16057e9a5e72495f17f712e7e":{"kind":"REWORK","_number":1,"created":"2022-08-21 11:04:10.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/81/853781/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/81/853781/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/853781/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/853781/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/853781/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/81/853781/1"}}},"commit":{"parents":[{"commit":"a03c4657c2bf957fef035807ef7df732f7a1c994","subject":"Fix designate-manage pool update MissingProjectID","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a03c4657c2bf957fef035807ef7df732f7a1c994"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-21 11:03:54.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-21 11:03:54.000000000","tz":-420},"subject":"Clean up manage pools and add proper testing","message":"Clean up manage pools and add proper testing\n\nChange-Id: I6bfe074ec9db9bd95021771b8ffe988f1f23e2be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/16f56897f15d09b16057e9a5e72495f17f712e7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/16f56897f15d09b16057e9a5e72495f17f712e7e"}]},"branch":"refs/heads/master"},"948adcaad13c3c6b5d3f658c0f0958969424b730":{"kind":"REWORK","_number":2,"created":"2022-08-21 11:09:23.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/81/853781/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/81/853781/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/853781/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/853781/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/853781/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/81/853781/2"}}},"commit":{"parents":[{"commit":"a03c4657c2bf957fef035807ef7df732f7a1c994","subject":"Fix designate-manage pool update MissingProjectID","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a03c4657c2bf957fef035807ef7df732f7a1c994"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-21 11:03:54.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-21 11:09:06.000000000","tz":-420},"subject":"Clean up manage pools and add proper testing","message":"Clean up manage pools and add proper testing\n\nChange-Id: I6bfe074ec9db9bd95021771b8ffe988f1f23e2be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/948adcaad13c3c6b5d3f658c0f0958969424b730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/948adcaad13c3c6b5d3f658c0f0958969424b730"}]},"branch":"refs/heads/master"},"3f50b7616b1ef678df8d9e50c4785fa909cd9fbc":{"kind":"REWORK","_number":3,"created":"2022-08-21 11:11:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/81/853781/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/81/853781/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/853781/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/853781/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/853781/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/81/853781/3"}}},"commit":{"parents":[{"commit":"a03c4657c2bf957fef035807ef7df732f7a1c994","subject":"Fix designate-manage pool update MissingProjectID","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a03c4657c2bf957fef035807ef7df732f7a1c994"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-21 11:03:54.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-21 11:11:32.000000000","tz":-420},"subject":"Clean up manage pools and add additional testing","message":"Clean up manage pools and add additional testing\n\nChange-Id: I6bfe074ec9db9bd95021771b8ffe988f1f23e2be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3f50b7616b1ef678df8d9e50c4785fa909cd9fbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3f50b7616b1ef678df8d9e50c4785fa909cd9fbc"}]},"branch":"refs/heads/master"},"db655753bc259274163f796d8beb0a053d0778fd":{"kind":"REWORK","_number":4,"created":"2022-08-21 22:42:52.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/81/853781/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/81/853781/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/853781/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/853781/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/853781/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/81/853781/4"}}},"commit":{"parents":[{"commit":"a03c4657c2bf957fef035807ef7df732f7a1c994","subject":"Fix designate-manage pool update MissingProjectID","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a03c4657c2bf957fef035807ef7df732f7a1c994"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-21 11:03:54.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-21 22:42:36.000000000","tz":-420},"subject":"Clean up manage pools and add additional testing","message":"Clean up manage pools and add additional testing\n\nChange-Id: I6bfe074ec9db9bd95021771b8ffe988f1f23e2be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/db655753bc259274163f796d8beb0a053d0778fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/db655753bc259274163f796d8beb0a053d0778fd"}]},"branch":"refs/heads/master"},"e65b32a6e6755cc0c9f286ab58ae20742d6d18bc":{"kind":"REWORK","_number":5,"created":"2022-08-23 04:19:22.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/81/853781/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/81/853781/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/853781/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/853781/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/853781/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/81/853781/5"}}},"commit":{"parents":[{"commit":"a03c4657c2bf957fef035807ef7df732f7a1c994","subject":"Fix designate-manage pool update MissingProjectID","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a03c4657c2bf957fef035807ef7df732f7a1c994"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-21 11:03:54.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-23 04:19:13.000000000","tz":-420},"subject":"Clean up manage pools and add additional testing","message":"Clean up manage pools and add additional testing\n\nChange-Id: I6bfe074ec9db9bd95021771b8ffe988f1f23e2be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e65b32a6e6755cc0c9f286ab58ae20742d6d18bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e65b32a6e6755cc0c9f286ab58ae20742d6d18bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
