)]}'
{"id":"openstack%2Fdesignate~857486","triplet_id":"openstack%2Fdesignate~stable%2Fyoga~I5bd1c10d863212683752e05bb450e6f531ff7e72","project":"openstack/designate","branch":"stable/yoga","topic":"bug/1989020-stable/yoga","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-09-16 05:38:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5bd1c10d863212683752e05bb450e6f531ff7e72","subject":"Fix Redis connection over TLS","status":"MERGED","created":"2022-09-14 16:11:24.000000000","updated":"2022-09-16 05:39:31.000000000","submitted":"2022-09-16 05:38:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"857486-bug/1989020-stable/yoga","cherry_pick_of_change":856313,"cherry_pick_of_patch_set":1,"meta_rev_id":"55517fe853eabbae3684d2bdbeefb8cbbc4d77b9","_number":857486,"virtual_id_number":857486,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-16 05:38:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2022-09-16 03:17:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2022-09-16 03:17:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-16 03:20:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-16 03:20:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-14 17:03:16.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2022-09-14 17:26:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-14 19:40:54.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2022-09-16 03:20:00.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"f9542aeff1bb125799c7430dabe6364ab928bded","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-09-14 16:11:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"57a33ecd8767a9805480198373582d2ddc05836b","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2022-09-14 17:03:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dba022c8fa95761a797775f267f095d56f50353b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 17:26:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffc82f0824394597aa8419d5776c5af3\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a30515865ab24dcf898ba33e2682d9bf : SUCCESS in 52m 42s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/fd87590a57084dc8803ce77d030e7e21 : RETRY_LIMIT in 11m 25s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/99cd16a3718e470b84498d2571b3dd00 : SUCCESS in 59m 06s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f54da1dc0dfa40969f86b80cdfa51069 : SUCCESS in 58m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/80f0d8dab2bf4572b6ca72a6a4bc24de : SUCCESS in 58m 02s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b3f28e49be6c48e58c8cbccff96b7144 : SUCCESS in 1h 08m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/81fcff8041c0402a92cea7424bd73788 : SUCCESS in 52m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6e4dbd40b8e14660a2d3c897cf01555b : SUCCESS in 54m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/264d6f75ee87425fbb75d20352d4d786 : SUCCESS in 6m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f36efdb3ef1242738f089c9ca25e6ad7 : FAILURE in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a280b359cafc4cfe8578139a3fac10ab : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47e56ecdaefd40a59f37976c4c9b0362 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0834beb7acae447e8efb0680fd123cda : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60cc16f8c75f4eb58ee126754912406d : SUCCESS in 7m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/97ce0a6ad67a42c58972577411acff07 : SUCCESS in 30m 02s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/99da886101b34251a6857265a06f354f : SUCCESS in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d7bd7b20094cff1c2b1ddae55a7cd4562d17e7dc","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-09-14 19:40:54.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dbc8a0c63c73bd40a2cf1a6792756588924d5353","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-09-16 03:17:17.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"387b595ab9d9ea16c165f8c903a9feaba46433dc","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-09-16 03:20:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aefbc48539f51c848b6da23a1a6bb9ccc440d009","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 04:33:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a25741ae3a744f85b35cf8cca6d7e360\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2796d0e6d3dd4223922695a765f6ad37 : SUCCESS in 54m 16s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/3ec0834a8c364b5b8406262b16b84f45 : RETRY_LIMIT in 23m 31s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa8c63940e27467d825136803b3cd348 : SUCCESS in 1h 02m 19s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f3e4e7173b7e440fadd7807f674621c0 : SUCCESS in 54m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b80c7c202eb046b3ae7c3d261cec8966 : SUCCESS in 51m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/95389df1b95b43a286d220195fe17a12 : SUCCESS in 1h 07m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/65252151e628414e9cf097feba2a84d0 : SUCCESS in 54m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cc8f0e2ed65a4316b0078afbd8113584 : SUCCESS in 52m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1409f6d045fc4f29ad791cd77f78ae45 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/851cf4ecfc4d4940b281d0c34e14774f : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8434d7cae04c43deaffb47606f7a56e4 : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/629bc0e019a34c4aa96a423dca170156 : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f8104f480484f0a969d38857d8cc872 : SUCCESS in 10m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c5d61fa7b5f94800a6b76417e8799152 : SUCCESS in 30m 27s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/abe67e38684c4462a19735623a68774b : SUCCESS in 5m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8021e76c32685d930b2bd1917a7c33a9f7c9a0e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 04:34:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e449f0e7b72057b1a333e6cd5fa1a6658f7ccac0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 05:38:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/466b15bc89cf42e8b03e7933b8d37912\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dcbd4f7c0ea245799379ebb7aea7588d : SUCCESS in 51m 11s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f50cbc9cfe8a4f6bbe546b57068e93b9 : SUCCESS in 48m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/57c60fa589c3475592af567dec131955 : SUCCESS in 48m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b1cd7c808c3b4de2bb23cc8518ec9878 : SUCCESS in 59m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ee063ba1ed31486d88ed052fa41178dc : SUCCESS in 47m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7dff1e0cb8e24135bdb9f05d6ea94079 : SUCCESS in 41m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aea915fa1f14c098777d1a3a25a53a0 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/def60994874f4495a4ab0afd9f22b463 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/887361002fc24fc0b4cd6b84da80e89b : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c292bb7ae03347bdae0f90c5065ed685 : SUCCESS in 7m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d5921dd2c126497ab7e7bd0ac26a7c25 : SUCCESS in 32m 15s","accounts_in_message":[],"_revision_number":2},{"id":"54a94ce56809278d107e69e803ba076efe54eb23","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 05:38:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"55517fe853eabbae3684d2bdbeefb8cbbc4d77b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 05:39:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bde6fe1f92da45f79a677d4e180814c2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4abc483b15f84503833dc780e7d3e636 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"eaab811acfe70ff72f0f3457931a1211753beb1c","revisions":{"85ee83c15c90d74932002fa568d122ebae3a0791":{"kind":"REWORK","_number":1,"created":"2022-09-14 16:11:24.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/86/857486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/86/857486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/86/857486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/86/857486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/86/857486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/86/857486/1"}}},"commit":{"parents":[{"commit":"9d96cc843aa54582db52b394ad8223a1939ac05e","subject":"Fix designate-manage pool update MissingProjectID","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9d96cc843aa54582db52b394ad8223a1939ac05e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-09-07 18:40:24.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-09-14 16:11:24.000000000","tz":0},"subject":"Fix Redis connection over TLS","message":"Fix Redis connection over TLS\n\nWhen Designate is configured to use Redis for coordination over a TLS connection, it will fail to connect with \"ssl.SSLError: (\u0027timed out\u0027,)\".\nThis is caused by eventlet raising ssl.SSLError instead of the expected socket timeout the core libraries return.\nThis patch monkey-patches eventlet to return the proper exception.\n\nCloses-Bug: #1989020\nChange-Id: I5bd1c10d863212683752e05bb450e6f531ff7e72\n(cherry picked from commit 93dee6a3ff44fb7470b3008e8fbbaf99822bbe82)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/85ee83c15c90d74932002fa568d122ebae3a0791"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/85ee83c15c90d74932002fa568d122ebae3a0791"}]},"branch":"refs/heads/stable/yoga"},"eaab811acfe70ff72f0f3457931a1211753beb1c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-09-16 03:17:17.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/86/857486/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/86/857486/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/86/857486/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/86/857486/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/86/857486/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/86/857486/2"}}},"commit":{"parents":[{"commit":"91395df0bdc4547bf48d5ff856e8b4988fae7d31","subject":"Remove lower-constraint job from stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/91395df0bdc4547bf48d5ff856e8b4988fae7d31"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-09-07 18:40:24.000000000","tz":0},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-09-16 03:17:17.000000000","tz":0},"subject":"Fix Redis connection over TLS","message":"Fix Redis connection over TLS\n\nWhen Designate is configured to use Redis for coordination over a TLS connection, it will fail to connect with \"ssl.SSLError: (\u0027timed out\u0027,)\".\nThis is caused by eventlet raising ssl.SSLError instead of the expected socket timeout the core libraries return.\nThis patch monkey-patches eventlet to return the proper exception.\n\nCloses-Bug: #1989020\nChange-Id: I5bd1c10d863212683752e05bb450e6f531ff7e72\n(cherry picked from commit 93dee6a3ff44fb7470b3008e8fbbaf99822bbe82)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eaab811acfe70ff72f0f3457931a1211753beb1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eaab811acfe70ff72f0f3457931a1211753beb1c"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
