)]}'
{"id":"openstack%2Fdesignate~857978","triplet_id":"openstack%2Fdesignate~master~Icf73e730b31ab26b8be65347239636c0137ab4bd","project":"openstack/designate","branch":"master","topic":"add_tempest_plugin_devstack","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2024-08-28 17:43:12.000000000","reason":"Change was submitted"},"5572":{"account":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"last_update":"2024-08-28 17:43:12.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-08-19 19:17:42.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-08-13 22:21:17.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-08-28 17:43:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icf73e730b31ab26b8be65347239636c0137ab4bd","subject":"Multi-pools implementation","status":"MERGED","created":"2022-09-15 20:31:06.000000000","updated":"2024-08-28 17:44:11.000000000","submitted":"2024-08-28 17:43:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":121,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"857978-add_tempest_plugin_devstack","meta_rev_id":"dc955f870d7a377252d162e9656e8088a6f9d1c3","_number":857978,"virtual_id_number":857978,"owner":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-28 17:43:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-13 22:21:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-27 23:34:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"CC":[{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-15 21:34:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-19 20:13:22.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-11-23 15:49:40.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2023-01-19 20:26:56.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"CC"},{"updated":"2024-04-25 14:31:49.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2024-08-19 16:09:41.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"d199986e72a7c7751dffa7e063835aa940049751","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-09-15 20:31:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bacfb33b5d4d08ad958dc1d8c319918cd1a2670b","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-09-15 20:31:31.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"76d2d256db1b4262223d960b1242c37b5d5596dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 21:34:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9cf544118d1469d87bd22639956fc0a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/47fdd9c600f84e479a3e5aefa59625b8 : FAILURE in 16m 41s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/64e98716169a4e46b25254ebc2cfc8b5 : FAILURE in 39m 36s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e65e7876a44046579ad330cf08b2ca8d : FAILURE in 25m 39s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/84121713a9554c949d042e5b24b0a760 : FAILURE in 15m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/285489ac376a4a9c9f9dd781a42133fc : FAILURE in 28m 42s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/39f6535a025b472f904cdc4ec9d6a87e : SUCCESS in 51m 34s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/13e9110a42f440f9bf6395eb643ced5e : FAILURE in 1h 00m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d30a2b0dd5434b1ca5f0ab65a9aabdfe : FAILURE in 33m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/aaaa7b28cdb1409dad741089fa31c4c0 : FAILURE in 19m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52e14cb48dc046f1b1fc6b233a3c7c3a : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f2692a4b4d34a5d835468d9bab41239 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ddb5a91d3b14055b0ff62413b1b0618 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e49521ff65c343a09c896134acd98cd3 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4379b1311b9411da6596a1fbac71242 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6bb7ed0365940659882f5839b891cb2 : SUCCESS in 8m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ea06953e41fd48c0845091c904de7562 : FAILURE in 35m 04s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/43c1ad6f0f4d4674aec3b7eb3d863b1a : SUCCESS in 5m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0989f34d683d7260e91d45a831e94a8bc3a9f8f2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-09-19 20:13:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"70839236b5ca288ac918dd0822b28163074fef75","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-10-24 21:47:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2075b6482596074b6c9ad7890f758e110ba35896","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 22:47:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b7ffad960e04c908a7d590988b6cbd9\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0e2bbe0a69f54da198298f4f47935b4a : FAILURE in 33m 14s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/562b736bf6334c1895ce0f3ed45a9f49 : FAILURE in 44m 28s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3891620b4f0d426f9024dcb7489cac63 : FAILURE in 30m 59s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ed70cdc89a134845ac19770f44738a76 : FAILURE in 27m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7e9974eb5f1c4fcbb6ca851c82540e0f : FAILURE in 30m 54s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/cd55f9c8969a4e64ac3c2ea38a0e085e : SUCCESS in 58m 27s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f19af2241ef74b129426939642908d65 : SUCCESS in 56m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9f7415dbe6b54ae29cb93a5275ff1732 : FAILURE in 31m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d031026bb8c04473b54a3f197499149e : FAILURE in 28m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bbfd34b5d6f4c7fb1a95e7db5c7701e : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/259e7ee392c144fdb531a75b1b414909 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a68ae9f8c7a64eb5a2b2e0d4a861ffcf : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b196ac14280b407fb7b3a3041b68c93f : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cc426e735d548bf9274999f6b9fc004 : SUCCESS in 8m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4bd10abf8ee04bff8fadddf15489daa7 : FAILURE in 32m 05s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2a6eb73a14134f2398571aa94a2a5a81 : SUCCESS in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7f173983cfb6baffe48088f1e835ad8ba820b062","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-10-25 21:21:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c66366bc865df141e263c030168045007d99c228","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 22:17:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2d494da16be4c28b6e03af82e371fd5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1b2d9f0b60f3400e9218b54812a72e56 : SUCCESS in 37m 32s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/718d12a7ab464391a92a74238a8a4f1f : SUCCESS in 54m 05s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/17527ee36abd48a2af3342d9e8cd154d : SUCCESS in 46m 38s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/c4df40fce2e14535baf99b4b1c2c0ec1 : SUCCESS in 43m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/356a6e0662ae4023b26e134d68f2d579 : SUCCESS in 44m 38s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6c9a1eb78bec4d3fb12453096c240ce3 : SUCCESS in 54m 18s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cc1f728cc20e4364ab9fa560cacb6ec2 : SUCCESS in 53m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3d0791d5d05a402796992c96dbc81005 : SUCCESS in 39m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f68dc79e33824297912f872570477d2e : SUCCESS in 42m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e197129f4ea94b7da2759e92f3ef95bd : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/677ff7a486874cddac590ac0e8d67db9 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa4295773c01434c9d8e888a16cf98df : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6e9553f2bda4b3d8d0fdb4e3b9343ce : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a8bb2108e7143119396b6d0aa26817f : SUCCESS in 7m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d1e1e4f05bfc4e60838eaa7d0116e475 : SUCCESS in 33m 29s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e7c1b2789e3c43e58504f2030a4f1ac9 : SUCCESS in 4m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9274792581b9130f3730d4d43191c1fd8c0c1a41","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-10-25 22:45:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7e2930ba333d74aff746ef078ee8892a694b9542","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 23:42:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba99223bac3549228e4e8de197ddd7e0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/516a6188a0a64320b4ec4215747b3d1a : SUCCESS in 39m 11s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a19da6fcfb8343d684891ddc4ed890bf : SUCCESS in 56m 09s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a171ee42c6b4c4bb30d64b0119bade0 : SUCCESS in 40m 20s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e0efe91dd08349ca894254e41a6a3cf7 : SUCCESS in 31m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/182661e7b6594c838b0287ce151ce323 : SUCCESS in 39m 13s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3f6f2aca7090479290625390c91f912f : SUCCESS in 41m 47s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b8a8092ea54a4b60ba31f5c127083f62 : SUCCESS in 53m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/faa001bb20834088b4a039d3fe95597a : SUCCESS in 29m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1ab441af21824c15b07833c64e970792 : SUCCESS in 40m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9531cc70cd334079bb2bda4b8ac06fba : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eb99b2951254c4ea8ebe2ea29fbc2c8 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c431b2f4165c41e18a8f9da7e4419af2 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86f49ebe6cd04c36a188acfdc6cef0e4 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a40708f981ce47058ac1260b37b08dfb : SUCCESS in 5m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/269558cd06094f5ab09ad0618224bf0a : SUCCESS in 27m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c0d57be12ff5416faa55397d0e9e87e5 : SUCCESS in 5m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ce6c03a19e7c44c4d9d4fc5d0c61a4469fc4dd2d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-11-18 22:50:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9b22d1dc9bc2dc07747956e34ae990223f22f466","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 00:08:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a984729c2456493e81c93004c4f06680\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/64d6f4d6f6854ef7b2ad3d4b93fc9520 : SUCCESS in 44m 25s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/32a5675d93cf4cdd93c7b7f215430772 : SUCCESS in 57m 12s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/15b16028737b407089981927e20f5afa : SUCCESS in 43m 22s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2477537d39e34a609579421ce047a853 : SUCCESS in 46m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a1f9d6e63131436b8bef71c9af0da490 : SUCCESS in 41m 38s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3b205104cf8c42959ce02d3e53d294ec : SUCCESS in 56m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3791c6c9ed4b492d8a7038cd941fa211 : SUCCESS in 58m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6f3c6f50c51a4979a3205e9475637c56 : SUCCESS in 35m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b6da9f13e0954b91a3fafa36aaa23ad4 : SUCCESS in 36m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0c8064af7aa43e0814aef2ab1eb6c0b : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dbbd53a5a974c2c8bb2f11a10273cc8 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c445aaad4fe4489bf519c3757a34b79 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9420d83ac5aa489d9ea4bfbeaf88e6a8 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cd12ed8a9f24dea82f6090563fd42aa : SUCCESS in 7m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7eb2fddb4b7e4e6bbf3bb0fbea020b93 : SUCCESS in 25m 15s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c63dedd0d8df495ea881009ab28ab1c9 : SUCCESS in 4m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f0e675df565765e3902854f772b14d3c2b428153","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-11-19 02:44:14.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":5},{"id":"1661e2cb2496d00d962bcbe94101bea92bc6489d","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-11-21 15:31:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"daedb7aa257e8716bb7cd9c0ac0cb1c49c31eefc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-12-01 22:59:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"de27017e91e0c58189b08c257846567ff1a02c94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 00:15:42.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9a6355d0eed45649a30e3d613353f85\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9182755263b4407b9e046bde76e5c9b7 : SUCCESS in 33m 16s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5d1fef8452b141c991a9ac14ac981422 : SUCCESS in 57m 23s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5324dd97a834a1386a48fe1eeb9945c : SUCCESS in 43m 00s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/643a644acf9b4e0fa2cc4ccc74b2e39b : SUCCESS in 40m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ae574b38c7074faea1d3a81618dd6f2d : SUCCESS in 37m 12s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8ede259249dd4319bf4920bc7466afaf : SUCCESS in 49m 16s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bb8860c321df4c88a469d512aeeec142 : SUCCESS in 1h 02m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9e2dacb1c72641899887120b39e76941 : SUCCESS in 46m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e86e1a51360f45b2a218d09408bd85b0 : SUCCESS in 36m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42d424876c6f4c2eb10fbff270935da4 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fcf3c648d154b7d82f03a0fa65aac36 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/485e3c1a30e6437e8453f3ce9323831d : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11870687a6314a7c94f11aed8d088ac8 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9921cd15400e4f40a85d4308805fccb3 : SUCCESS in 7m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/15e1f4aa2be14cb79829e58a25de9f9d : SUCCESS in 27m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a45fff19a44f44d2934682d7d2dbb1c7 : SUCCESS in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d46d06e8a1b732c8195b10393cc73b933a7adb95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-01-17 20:16:53.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"eb01eac9ab9578c11e39576021a1415ea6913191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 21:36:40.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3fd1b9f41b54b25bd7c56470d26bcf4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/15a676473f5342c9b3d4ef4228ca9b6b : SUCCESS in 40m 04s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8ce5d78238a2491b9d13ba055bbd0e4a : SUCCESS in 1h 02m 26s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/63b6a20807ec4d35bf1e835fc53a988c : SUCCESS in 50m 47s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/7188b4b3a0434b55b0c8710b9df4dc73 : SUCCESS in 36m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fa74211e19d3409e8ae9e5c72dc41bc0 : SUCCESS in 53m 39s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3de5316a61af4ab397ef40486f7259dd : SUCCESS in 1h 12m 06s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/15868c3795c3403cb7f4fa56b3ab316f : SUCCESS in 59m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c240b12c8ddf4671b93b977e255b02bd : SUCCESS in 49m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/373a6379f47a43c9b3bbd1a696b8c199 : SUCCESS in 41m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/108debe0b22f48a49038cf329c7d7b6d : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd06c3a50cda4fc09e20be3bc5293729 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78fdd6aeac684ec7860fb363e0f0d4bf : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9083435492d8449ca8e097d1812e3d9c : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8bfb34f91cc46b2b4d91d6f85123ad1 : SUCCESS in 7m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/66742e57faef48408bfa8e04fd538903 : SUCCESS in 35m 29s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a23dddb3fa6244b5aeec1250a5c1e5e2 : FAILURE in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0bea06f8f9e6cea9b45efe5e1f90b6bb8d9b9a2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-01-18 22:26:41.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6b755acd42f9a131467677a6345cb088bd864427","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 23:25:07.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d79322680504ede984181e6e97c42ac\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/38af865ee94a4d16aff148a66074c5a3 : SUCCESS in 47m 06s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2ccd79916904428297f4f3592f453b6e : SUCCESS in 48m 41s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c7453ed81014c7585ff14460a6df841 : SUCCESS in 42m 39s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/7df7c0b976af472aa0779f0a59ec9221 : SUCCESS in 39m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cd8deb5c25414401899019781a352b5b : SUCCESS in 40m 13s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2877e38d94b7439d91ba4d13fd33eb8e : SUCCESS in 54m 03s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a89d8ae9e7b4427c820d65def9f9b693 : SUCCESS in 54m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/827282a81633477c9c3c7cc3b9976534 : SUCCESS in 51m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7a434add0f044cbab424273d0f838048 : SUCCESS in 34m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5318b89c1424df89768d500613527e3 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13f81dd70f2047dbb2b33494a6aa0e68 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ace7bf9d6f3f4fb7ab794bb8fa4c825b : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f393b613ac14cbcaa7ccb41b4a45079 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5cacc8ff4ff497f9e86079edd6016b1 : SUCCESS in 7m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9086bd5859ba4980a2305451fa204f89 : SUCCESS in 45m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d45db33e202c495c8e876690ad321df0 : SUCCESS in 3m 54s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7547afa75ebded1a572412ccc265060fdfea1fd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-01-30 21:22:28.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"fbf4c833f74aecacad1d68c68773ea5bd8cf0206","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 22:55:21.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c9a323c4f1e46038854b63f1238a917\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c4aceabcad2b4ec68b81043f67452f48 : SUCCESS in 48m 51s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/093b235308ff478eb88f12f8c2423b5e : SUCCESS in 1h 14m 28s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7c12a3c4b584c03a003e7bcbfe153a8 : SUCCESS in 1h 03m 27s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/d3f3431901e04f2f8326a743dd24a485 : SUCCESS in 50m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fdb5f595175f41b5a102a6888f6a73bb : SUCCESS in 55m 09s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b29b9bb9ebb8457cbfa80f3a9d0cbbe9 : SUCCESS in 1h 32m 08s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f267d5b4fb614d4b8fadf4e1f2981816 : FAILURE in 58m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ba4ca20e3dce45e48eba436ee2ff45c5 : SUCCESS in 43m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/82c3c14c378546a3943eaf7c36db3d53 : SUCCESS in 40m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/885eb44af45e49759f93aaefd91b9966 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32fd98e5da9142c096db1b0d7521dc13 : FAILURE in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5112aaa13334307b31ac6f7bfbcfb2c : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b84781ba6364dd8b9e08a6c4f014f4b : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86b263fea0f64ddcbc40d8545789d569 : SUCCESS in 6m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4e211b0ed2a8425c9202f22f859cf2a8 : SUCCESS in 29m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8ae14cba21a3441e8bf63bcaeae88f2a : SUCCESS in 4m 39s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"fb027fd6bb91bbaa995911494ad069f28dbb129f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-01-30 23:35:36.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"ef70bc1c2fdc32d541e10a32cb1616674ecdb2e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 00:42:58.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70d7112e58554a98898dae0c2e70f735\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/06768aceb6cc49e68d729289a63bb8ae : SUCCESS in 49m 06s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/87ba807f93f54c8c9ca200475ced1e8b : SUCCESS in 55m 52s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/76518df878fc41ebaea3b76e6eec682f : SUCCESS in 42m 32s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/7dac28cc6fae4e58a5761faaf8eb3193 : SUCCESS in 37m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6759760991c746018c8c276813e56523 : SUCCESS in 34m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6ecbeead03af4eee9c00936942b6da2c : SUCCESS in 54m 14s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/21ba8f9573ad4b1cb09a98e473fdab7f : SUCCESS in 58m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/779e0529d5114650bf751af845cbc1f4 : SUCCESS in 47m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/10f2d0a906494fb38df90fe5a6b22a0b : SUCCESS in 40m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a71b3a7356e43e0863f893d0fb3169c : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8934b619c86443d6aeade137f1b21ee7 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b79c4f6634d46779ff2ff45a095f869 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00b78410c87b400b97cc99fc4c1f9971 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfe3543e7c3642a4a2dcbc9de78b95c7 : SUCCESS in 7m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0e7b79469a574130bbcdc9d35f127a0c : SUCCESS in 27m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b1c1a450793146bd9b0cfd9114da6622 : SUCCESS in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d3ffc837861433e0767e3b950c55dbeea600dd96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-07 15:58:02.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"changekind:NO_CHANGE\")\n","accounts_in_message":[],"_revision_number":11},{"id":"988a33a99474bdf36d0201e928ea14b3e9d8c017","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-07 15:58:48.000000000","message":"Patch Set 10: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":10},{"id":"90e9bd8985cfef3d0331a2b4adc00c015f08d6d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 17:16:43.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dcff2c6db8184d719dc5fb1d07c34f5f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/636044df54114186bc2cba2d858fdf02 : SUCCESS in 46m 47s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b18dbf141c5540499ac46b6f355f6051 : SUCCESS in 46m 02s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f17a8f141f74b99ab06e8742139d73f : SUCCESS in 42m 59s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4d55a7088a6c4b94805f76c43a9314d1 : SUCCESS in 48m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/de3f68f0116e42e985a2389b3fbe15b6 : SUCCESS in 45m 13s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e3d90f21eca84634857b1357938df2c8 : SUCCESS in 1h 09m 46s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8f09215fb3c048efa79b07d1a925c63a : SUCCESS in 50m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/df400340be244a7aa12220c35d690c10 : SUCCESS in 34m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9f692b5aa7eb447688fe6b613278c885 : SUCCESS in 52m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d49813cf4def43afbff96cd2c26afe54 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c9a9637e9b948d28b9804b8a20c5e88 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbfc663760504e59a96bd6437314e08a : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e15e09ff6ba7437eb495916be9349c69 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1d9042980414510afa04fd523780126 : SUCCESS in 5m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5677fa6c9e8f422fae91c403ab65d737 : FAILURE in 14m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9d69f1383c334dd69e78f2b0bd0fdf73 : SUCCESS in 3m 58s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"aa8ed1252175fc7b1e3084e343bab79cd0a3f62d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-04-25 14:24:13.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"e67d643ae8596f5c706eb0cbda05c535c2b3a85c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-04-25 14:26:50.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"2c7f4fa7142a24727f158a870fe2b64bfc58cf62","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-04-25 14:31:49.000000000","message":"Patch Set 13:\n\n(14 comments)","accounts_in_message":[],"_revision_number":13},{"id":"9c395e53df6ba5582b28c3fcd4aa2dca35af730e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 15:33:53.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/616f3899038745f982589e35dfcedd5d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/02001efa0a564a32880c2eddfb548f48 : SUCCESS in 41m 39s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/baa76fd34887424f9cd27e8579af5a13 : SUCCESS in 49m 50s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/cda8a7284f1348b495c948d492b7a12d : SUCCESS in 50m 50s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f5e1ff8e874147eea09250c2220058e7 : SUCCESS in 44m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/58733f736e20463ca28c1ad511c3d787 : SUCCESS in 44m 42s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/dc5aed18a0364a3fa4197e4a644adc74 : SUCCESS in 59m 41s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d8c85c76c42c41c884dcdb625e4fde89 : SUCCESS in 1h 01m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b0df8d44cf394f1b8c47ddccfff8d301 : SUCCESS in 44m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3a8c5b442e8d4ad3aa332a380759c5e4 : SUCCESS in 48m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e500aaacd7fc4ca5a9d775114c491e76 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8eb6ea98e9d474d922015b03efdf00f : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f36a7e2c5a124fc7a4eb3561042aadd1 : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/06c4ba54350b432197cf7c7dd9d28f32 : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/48f162daceb045ba95e991a4891b77da : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd0ed9ac0ecb4b8c9b385d9141960cbd : SUCCESS in 7m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1750da2449a8408f96e8d470e615c6a3 : SUCCESS in 30m 59s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/524d531a85a740b7ba9d72d8db46f885 : SUCCESS in 7m 40s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5a02bc1d479b49daa37bb764ced58535 : FAILURE in 33m 40s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3dd81363c2b34397a737306745cee6be : SUCCESS in 5m 03s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/83066566dded4e708982830cf61e5789 : FAILURE in 48m 36s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"a878c30b634314db010875556b913867782ee39b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-04-29 11:23:51.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"e93608148e5158e3564a3d2a276d799f06e89351","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-04-29 11:25:53.000000000","message":"Patch Set 14:\n\n(9 comments)","accounts_in_message":[],"_revision_number":14},{"id":"070975309d1f383b368ae83ab826cc3bf2a8ee36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 12:27:45.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a030559035f40b2b08bbb25ed5fe26d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e147c20436e64bbd95b269c0a9a681c0 : SUCCESS in 46m 08s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4f73c4136e7642668802e01059c81cb7 : SUCCESS in 56m 26s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b15baad1f77e41bf944be91c8e60e463 : SUCCESS in 50m 22s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5c91415f03864a009ee0224c31f61b73 : SUCCESS in 44m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2b652ac923174068bf0054eb7dc00e5a : SUCCESS in 44m 23s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/84d225cecc62419d85030d8708fbfb2b : SUCCESS in 59m 37s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ab05de6069f74126aa5474a9a9de3c10 : SUCCESS in 48m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e707c1d6924d472d81afe37ab15ae943 : SUCCESS in 42m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f1d600a4192e404783b2fb155c356407 : FAILURE in 42m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12906717561c4e029ea343a83fa87860 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b9b21373c204aae8e6b38aee13fb958 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e69a9ea5feaf4e0ba761c20229d37edb : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/13641c42ea6b4581915c0674919c1b53 : SUCCESS in 6m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/83aa300a8bcb4fd18c0165878c73e2ad : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb9b2eddaa944567861ca402534e34b5 : SUCCESS in 7m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ffaf1ba76a5849e9b2f83588b9c9238d : SUCCESS in 32m 42s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/65932d4a8ff74b9d82c21e95023097a7 : SUCCESS in 7m 02s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7fb8bf95f09a4814a2e9fecc786bcb86 : SUCCESS in 39m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/939deb61ba6943979bfd8573599d845e : SUCCESS in 4m 36s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/d4054c3137654417babc48f1b32eded9 : FAILURE in 43m 45s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"076d47352d0e7b8e3efd806a06fdd60c75064ebd","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-04-29 12:39:20.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"ff4fe668ed1e274d9df34b0d86e0266831104b8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 13:53:00.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7eeb862b6824ce1ac8c9186908300ba\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/feb4d4d7b98f4d6185fb2fcc3754fa74 : SUCCESS in 45m 11s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/00e3b0913e984a42ac1abd15941ba308 : SUCCESS in 42m 57s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f685df18ab234e499e9a1eb98a3beea7 : SUCCESS in 39m 32s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/42270ce7f4884c119d2d68093b0ad607 : SUCCESS in 37m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d7b2c4a0e13a4891a67db8b6bba71cbc : SUCCESS in 34m 27s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/65a8f7bcab6d4829a970ac70c741060c : SUCCESS in 54m 44s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e60b46aa44b6425aa4a6db42f207a7b7 : SUCCESS in 53m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/efd23d6a72f147f2b15d80c306669352 : SUCCESS in 40m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d1fd6233488c458ab3d0d3654565f546 : SUCCESS in 44m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8a00ec5b43648128b1c4cd4ccd144c3 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a14ff53905e84e6883b2b3fbc6ac295f : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fed50826cc554eb18ceea9505fcb90b1 : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f4a96a53bb84432391e82c5409876edd : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/96a6adc044124698a6265ba4ef619da8 : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37e009e8e54646a9889caf27f0b221e7 : SUCCESS in 5m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7173ff7a03db43f088efdbfbcac476c8 : SUCCESS in 48m 58s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9b2d09f2e8ac4e4f8dfe5c138ece401d : SUCCESS in 7m 24s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/30cda1c78b544ae3b0dc28dd4f16bc07 : SUCCESS in 53m 53s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c19dc5217f484034b37ecf47653df801 : SUCCESS in 4m 30s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/f4b68284123c43ffb58223ff82226483 : SUCCESS in 35m 13s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"9cd27a89a1efded62663e87b415e2b2783c4e1d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-04-30 13:27:01.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"6512715f0119d0dfe71af1d159f090ce33f79d77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 14:34:47.000000000","message":"Patch Set 15: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0059051f6f46425ab23e3f7b7c908249\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e1ff36665951425fbd9f882164c434d3 : SUCCESS in 42m 00s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/597f81f7dd2d4d039efc03156815b08f : SUCCESS in 45m 02s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a61f89e2e51e42c7b30e229121d5bd94 : SUCCESS in 55m 39s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1bb1ebf6ca8c40ff9700334b07dd364e : SUCCESS in 35m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2b5cc1a61e3147c290d4e272e0e4b275 : SUCCESS in 52m 12s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3a8ffa0db0b74a3f84263a4ca761ee1d : SUCCESS in 55m 23s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1f51b9a991bd4cadbaad6d117e3a6e2b : SUCCESS in 56m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1686bfa082724a55b32935a3e3796ceb : SUCCESS in 43m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1e93d4e09ccc4052a662fc0492fcc28d : SUCCESS in 37m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43192771b62e45bf99f1296cff5d365b : FAILURE in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5302667fd7654ac8baa7cc8417722388 : FAILURE in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbbc75c92a514012ad74f60796361d24 : FAILURE in 5m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7091dccd02a643bf9b57b24c25038911 : FAILURE in 5m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/edcaf75f5b5845478bce1cdb3932b3dd : FAILURE in 10m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aae3697908b9462dadbf991448ead8b9 : SUCCESS in 7m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/114b59aabdee45e3b54207827d58d452 : SUCCESS in 35m 57s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d0a52dd7e698443385c11471115a6200 : FAILURE in 6m 57s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ecf30802aea94858a532e2f4ef596115 : SUCCESS in 48m 39s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/79336993185b428291b593d2182696f4 : FAILURE in 5m 22s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/3c3ef2dafb764f999aaa74baf3be8efb : SUCCESS in 47m 05s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"b06c9b738da0e882a22459d296f211f15b427c64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-04-30 14:46:26.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"784feec872f6950e6bbe4ab618109bd385108550","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-04-30 14:48:51.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"ea6a7adbf41506ba277d8ac837db0dc6c17cdee5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 15:49:29.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cff8511b802b4e12a9c412c7534e046a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/19299d22a44b489abc004f63acb48e2e : SUCCESS in 48m 22s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/50dee9fdbceb414ca1d8705000a75dd6 : SUCCESS in 57m 25s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c6463311d6644549c747a5f0173f8e9 : SUCCESS in 52m 02s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ccaa0871da434812b0dd9dd0fb5bfb7a : SUCCESS in 45m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9880f846b5004ecea7a2118126cce633 : SUCCESS in 40m 26s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5d4258e017ae45bbbf0a52b17d43be11 : SUCCESS in 59m 49s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c7741241d2574d2b8c425e39139d7b40 : SUCCESS in 1h 01m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/555eb70670c64946892e8650bb0bc0e4 : SUCCESS in 35m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/64020201d06d4dbb9693f541ca4df49d : SUCCESS in 34m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3a8635d4d9346429d0c2c2a20c8b63f : FAILURE in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d7ba034b6cf4043a7349c8a5b1f2cc3 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b6cbbcf437e464b826c0b134b13e3c8 : FAILURE in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c7b17ceab23b4d968cd4c8f81989ba7c : FAILURE in 5m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f648beadf9c94e038b0b020296bc7c89 : FAILURE in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5a18ca2c1044ab690bb64ce4269b591 : SUCCESS in 7m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/05e8f4d95a8e49ea85130d12b14b6cbb : SUCCESS in 32m 48s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8756f002cb6b4953a8e06672845b6fd2 : FAILURE in 6m 58s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/55b036cdc9404961963184223a549588 : SUCCESS in 43m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/65c69441a6ed4e998c9c4de4cd4cce4c : FAILURE in 4m 55s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/472dd4934c5748ec9b699dd511d396f1 : SUCCESS in 49m 41s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"19f7d3abb3a485b434020632901d497afbe692fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-03 09:49:56.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"1cd6d1c3db5b9852c094d5fad975f2f9f9093181","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 11:06:25.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67c369a06ed44bc7967decc3d4650156\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/65be6cea705649fdab0f97f3b0b0256f : SUCCESS in 40m 25s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3e006107ada442c3987788abe44a6e09 : SUCCESS in 42m 22s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c08eccdfa7ad4ab19e2c9404f5d4e6e4 : SUCCESS in 36m 43s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/dc1af44d69374ec9bfc050f59e87fbb7 : SUCCESS in 37m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1e70c5b993ac4c20854e9906d0c9715b : SUCCESS in 49m 07s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b1aa567d97334b95af993320fb42c39f : SUCCESS in 51m 48s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0a87e8148e7c4b2baae3012fb35de8c9 : SUCCESS in 1h 05m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a5a417c5f206402eac29099745c984bd : SUCCESS in 43m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ec1677d109bc476eb9739f287fa5d967 : SUCCESS in 37m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99d2e49565304ae48bd503cba68c851a : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b663b1760d7b489b83cd49392dd9d5d5 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e88eeef25944dd8886260ca1fcbdf3f : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/256a12af0eee4caea606bafb18634665 : SUCCESS in 5m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f44cf86406484074b1b3671c7103b6af : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ec9018d3cb24588a50ea98976ed1bb0 : SUCCESS in 8m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/54e80fb549b64271a69e949e15db501b : SUCCESS in 43m 43s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/695e8824a74540a4a8cc9ea1a48dc88e : SUCCESS in 6m 32s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ae48690b903840358bc82cc950391c2b : SUCCESS in 33m 20s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/afdda4d30e8c45b39ad6f23e09f5186c : SUCCESS in 4m 23s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/9f3ef51af32b48fc9850bfd9e7f10fac : SUCCESS in 48m 27s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"5ee1a6315c090bd82e6bdaa8e7c8723756001a52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-08 09:20:18.000000000","message":"Patch Set 18: Patch Set 17 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"12676a8af3f24b0c74108f39bd32b90f7a137aec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-08 09:50:27.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"e1549881dab9a169a4d3ddbeab61accfef49fb52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 10:59:55.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3e148dc7e734c99b12202a2470142f4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7a73bc4486484290a211f6d7fae600f8 : FAILURE in 45m 07s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b41738b2a4b4472c99587cbc64099213 : SUCCESS in 42m 36s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/77b4eae56ba64d6fbce50c4f14ebcc13 : FAILURE in 6m 37s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/833b3100a5fe49149f09eec1d4820c57 : SUCCESS in 44m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ee2c4c79df8842e2ac69c1129961fee8 : SUCCESS in 43m 25s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/42cd0f74c64940e5bd3d212e0caf1fdf : SUCCESS in 1h 03m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3abeb038f0fc4c16937ceb94483dd05f : SUCCESS in 59m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8ae704ccf3e84ad9959874d03aab2932 : SUCCESS in 47m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cbed68920203444eb93e3f9a67a7df42 : SUCCESS in 43m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7bb24f14f97441aaf9ffd29276fb477 : FAILURE in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f28642df62d24c7a866a6ee16d99aa60 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11f35f7f0def458b8859dc3cc823cd90 : FAILURE in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/781578c9cbde4ed0afe990dd422661eb : FAILURE in 5m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fdd1885c4878407c904e5cbfddbf3ca6 : FAILURE in 8m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f1545e1c31e4cefbd50115010eee206 : SUCCESS in 7m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/12445646720f4e2cb6579e6336180720 : SUCCESS in 37m 39s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/96b87f627d4143fca8e26f526bbc35bc : FAILURE in 6m 02s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8b92044728e24d67a34348d16d4e9b72 : SUCCESS in 40m 30s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e2f258db943f436f804ae7bca0cbafb8 : FAILURE in 4m 27s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/b7cf9a0abcb3478086ee05f5e3ae3dc2 : FAILURE in 46m 00s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"eb18c4c4cbc0f14ce63124d24ed8b22060afe9a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-08 12:38:15.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"b1ecf0194dd53dc89475a23a209ef795ec58cfde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 13:47:56.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3788be7c45bf40968d4bbcb426fd73eb\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0417b988157b45348b85445ddd208f9c : SUCCESS in 44m 16s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/791722c3245c436b8a0bcd237f5bbdbc : FAILURE in 12m 22s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2a2d4224c4f430ca69ed14d403b29d0 : FAILURE in 7m 38s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/503bbe2544154e2691714a7bfc7b20c1 : SUCCESS in 45m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ffbc352fbdf94d6f850b50919a57979a : SUCCESS in 48m 28s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0135641bc32c47bca754adf7ce3375b3 : SUCCESS in 1h 05m 13s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4e534c8ee0e247f28944b9770f671121 : SUCCESS in 49m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/64bd63418182409dbb02cc6f8f4481ad : SUCCESS in 42m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3fb65aa61a1648448059fdb321af1ea8 : SUCCESS in 47m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/355e17f32d9149a39aa67aedc0e466e5 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/488a70f3559f403396485c0b3433d8f7 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa6aaf8d5ba344dd95585e8ee9e1006b : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f3b5b6534b94ad5aab1a00fd34dca0a : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/da5c2c1c6c6f4a03acb462c492abd460 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a68eee1bedc3497f9fc40b024de5eed3 : SUCCESS in 8m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/766fceca463647b8b3c169f8309067d1 : SUCCESS in 39m 25s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9a77a10845e04c56b64727dd8b4f96b3 : SUCCESS in 6m 45s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ae0a3a84cd294c42b44aa23abc51e1f3 : SUCCESS in 46m 40s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b8827573be3a41958bd1275b818f6394 : SUCCESS in 5m 31s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/f74711159b6f46829f30a940ba69e2e9 : SUCCESS in 44m 12s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"37c197150ec1ea94889e84af87bb544493e3af6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-10 11:23:29.000000000","message":"Patch Set 21: Patch Set 20 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"725b80a1291207b422183a14da6168a846ee7103","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-10 12:32:36.000000000","message":"Patch Set 21:\n\n(3 comments)","accounts_in_message":[],"_revision_number":21},{"id":"b6ea80f21031e361bac086e297a33f7744317b3a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-10 12:33:35.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"0f233238031e497ea51f7f65c6d8cda5ef40bfca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 12:36:20.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0284d691a872402daf1af6a7c2f3cd3b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/62c2f27030f64694ad7e99f9b212410b : SUCCESS in 31m 49s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/919b33f64234432ca24b600e7e8960ad : SUCCESS in 57m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/76fa500be79446b79aed2be150a35a45 : SUCCESS in 41m 04s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/cd23849bc4684dfb8bf6d4ee48b6b3e0 : SUCCESS in 33m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7292c5c9b6184628878738c25f24e2d5 : SUCCESS in 40m 08s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c82d1d6bf0fd48c59ec0e47a6af8bd80 : SUCCESS in 58m 15s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f2f6915c125341d3bebf6214ac5473ae : SUCCESS in 1h 06m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/538da1bb51a84110a1fd0fe3ae38c0e3 : SUCCESS in 46m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7670e98bc47343dc95017e892155bc3d : SUCCESS in 43m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b476e131d96a43508c4dbc7200bd4918 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33f65f6ede8341f38ed2dfd1d42659f5 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/844da5f345d2408db25997214452c9ad : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3dabc28c88f24c18af23acd91b9abbb8 : SUCCESS in 6m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3fc6602ce62a4cba924810a5e2d9082d : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4205c14f85c49fab2510c41594de83d : SUCCESS in 7m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5119f1a741c3413aaa61ba4d63d79ef8 : SUCCESS in 39m 27s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2fdadb3bd28e4f5badda25b3a0209651 : SUCCESS in 6m 30s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9a7bc6f414ee44d086285254e4f39032 : SUCCESS in 47m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b4313e245c4b47b8b680a53335f7782f : SUCCESS in 6m 14s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/74b246bb354e4de5bc705ae5471e54aa : SUCCESS in 42m 36s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"743f200c7b4c41557d7a070c262dfa423306c0a9","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-14 12:01:03.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"15c60a63af6f6a3c662e9fc0365da742534df117","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-20 17:35:48.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"f0975c178e163cb35d9da4d4fd87f32fe45fd0a3","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-20 17:41:55.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"25498573c27009f608dc4d2b6504b26aca77ea65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 18:58:26.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/29df95ba5f2f4ec5b927ad4281cfd6ee\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2dfcb3c5f9db4b879cfa3e0b6ebb2da4 : SUCCESS in 35m 45s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/26166fc62ebc4d1598edd7e76a9151d3 : SUCCESS in 41m 15s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/88609ab9c22244548243ae7543b16065 : SUCCESS in 37m 44s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e54dd67ddd6d40d281bf392165b82b34 : FAILURE in 37m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b9ce5444e8e84f4caa74f2094d1caf4c : SUCCESS in 40m 47s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/cfd72cb5e64b402999c4608364d46510 : SUCCESS in 55m 39s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c4e831a262c34daf957b09400eb13fda : SUCCESS in 51m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f89208b82c6e4994a8349e77aadb59a1 : SUCCESS in 33m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a4a7910083c342c2bb71ded3991b45ea : SUCCESS in 35m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3bd9db5cc454fb2a47056c5f3df1a40 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a1a559e06034ea0ade9c92645cd5efc : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d067b6c95564e858f7e4ca693bbc37b : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2055177644734f109e87670fd22444be : SUCCESS in 6m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dc03d00c21294ddbbf63af55c9161d13 : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34f4d407fc6f43f6b929dbed24b4cd58 : SUCCESS in 7m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/41f2b5b796ab4dddbfcf129b027bed86 : SUCCESS in 31m 50s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/af6814b48b9d4b38a2f4e8f92a2c7dbb : SUCCESS in 6m 44s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/04c72b870fdc4b5d85a444aa891b9ff3 : SUCCESS in 47m 26s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4a51d66faade46349654a42a2a2429cc : SUCCESS in 4m 45s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/f1ceb924b1ff4df08d0f8694c8061f7d : FAILURE in 46m 50s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"a3d931a1656af121143bb969d51e6fd26044cde6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-21 08:34:02.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"c59ce33354fc8953452d7ab9f4c97e49fa3bf35f","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-21 08:35:52.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"d72a5ebbcd472aff930fc0114c1af7fe52a9494a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 09:37:33.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf86d3ef95304de5bf598c357334d2c2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bbfef88461ac42ae9800757152b0d2fa : SUCCESS in 46m 39s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/91f75768b5524e40a4edc7a6c4086884 : SUCCESS in 57m 52s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bcb5eefece874fdfa21a24f708db02d0 : SUCCESS in 46m 47s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/14872493f9c247c5a103415f1382eef2 : SUCCESS in 44m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3f848e3fb4a7443888f6e5129817e062 : SUCCESS in 50m 41s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e1bf9f3aa1174147913a91bd0c29c339 : SUCCESS in 57m 27s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/839e61f24de1499b821278c1962e3564 : SUCCESS in 1h 00m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/769038066ba341909a189c554093878e : SUCCESS in 48m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/518baed40bc1430abd68aac9116c5c38 : SUCCESS in 39m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/141e9dfe3a994315bc8259a6c1496581 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5123240babe440c1bd7923a7e0e5c613 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44588c17a6df45ec8d68b08b332091a2 : SUCCESS in 5m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c87b0ed80caa461c955cc4a2dcd8e61f : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/455b97e6aa7f43939c1f7daf3705b0a7 : SUCCESS in 10m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b708fa88f7248c2977e9d6e6d578dc4 : SUCCESS in 8m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9275f770aa4b41efaccfd0a70ad5b2f3 : SUCCESS in 45m 39s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d15fd6f36a9c43c486b3a5138b81b477 : SUCCESS in 6m 25s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4be0a05715bb402aadd3706d5a34f67e : SUCCESS in 48m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c53d9e659a854b53aa27e7e464a75eb9 : SUCCESS in 4m 56s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/cdd99419591f4ee2a73dd14286d58abf : SUCCESS in 42m 27s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"6d7348ed17872e8bad82a678c6915bb2dc3d19a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-21 13:47:38.000000000","message":"Patch Set 24: Published edit on patch set 23.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"45f17ffe5fbd5b2f67186331277701a09507d355","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-21 13:48:03.000000000","message":"Patch Set 24:\n\n(2 comments)","accounts_in_message":[],"_revision_number":24},{"id":"6f460a62c9cbd15ea2f5bd0a5a8cb0025df1b697","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-21 14:37:28.000000000","message":"Patch Set 24: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"af4b92a437c39b5bad57c3ed5b095941986d1756","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 15:05:15.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bb8bdcca88f4ac586f29dbf600a90ed\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/75f9f06628fd420897b638f0a8c08586 : SUCCESS in 46m 38s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/010f44c7b84b4d8fafd2ae076be5e836 : SUCCESS in 1h 10m 45s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8187e3fe99ff438095912b2eef04463c : SUCCESS in 52m 53s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0c49d25283a44989affcbf28edaf399e : SUCCESS in 45m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c45a7a3f73184d6fb9b31d3d2980e6bc : SUCCESS in 47m 24s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/fada9f69a9534e1f9cc241f2adc68a99 : SUCCESS in 1h 03m 45s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ea6a33fe642d42f59c269a57e1a8824e : SUCCESS in 1h 01m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/012ed8c3a36b41bcbf19d70a916f9e24 : SUCCESS in 49m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/25a3bab7f7fc43bbb88a879ad4c12d0d : SUCCESS in 43m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e4785b2543749c3bd2958693241efdd : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5becf0fe0a546d991b767a42000aee2 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60e75953a33a4be7b18fd456b25fccdc : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c9e6631b91042cabac1afd3ebca96d5 : SUCCESS in 5m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bdc845398e2449f5bceab07391e5e330 : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27f1d46eaff945c9b8033095c5d861e3 : SUCCESS in 6m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ed47156935684c57aae61ffe2b2b54d5 : SUCCESS in 32m 51s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/dfa8f8a7a84b4a47bea6e8974c68a552 : SUCCESS in 7m 01s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/60c2272d24604cacbaf0a3a94f9209b7 : SUCCESS in 49m 24s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6d19b012a5a540f7832033b78258420c : SUCCESS in 4m 56s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/f2ba3123d93b435381e6bc04613ae9ab : FAILURE in 33m 10s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"52c2466dc816d5bd90c7f541f904ffc5a396b262","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-28 11:58:36.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"b073e8c00a670776bb16404bfcb40a01dc1a1278","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-28 12:11:58.000000000","message":"Patch Set 25: Patch Set 24 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"a3d33c0c2217859f7ea26d5a49ef778d409926a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 13:25:48.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6baecae1b21d47b7b10ae12f712f6666\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0acac3cb7fd2423cb8091f6629847feb : SUCCESS in 43m 04s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4cd96f51bbff43d888d4fac8af4e9ae4 : SUCCESS in 1h 02m 32s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a881a9a23674d5abd9549c7dd057c1a : SUCCESS in 1h 02m 07s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/61bb39cc3c474722950d7584d8e9e7a4 : SUCCESS in 41m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/eaa18c7e2c17419dbcef88569c1e51ce : SUCCESS in 41m 37s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/74b42ee4b2824fd69c2185959090ed43 : SUCCESS in 1h 04m 27s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/90c5b5fd860f4e54a0253c1202e48135 : SUCCESS in 1h 08m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bb5e5d83700647f5b6aa6546b8a8c1ca : SUCCESS in 38m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7df08d7812064cfab5dfdfb95dcc8f3a : SUCCESS in 37m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ceb943b2f454a67bc5a4c20895998c5 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c4b0a37cf1646cabd23e0bf4974b149 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07cfe393b13a4d418528e0cb029ad20a : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/31f527856bd34e10a882330bb17fcc3a : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/39c499db40c94b61a90c4633429ab1ab : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3e01a9ca78342fa8dfda989a516278f : SUCCESS in 7m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/96bd4468ac56406da65220441d287bcd : SUCCESS in 32m 00s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/379778dcc29c43e297da71f9d29dc577 : SUCCESS in 5m 48s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/eacbbb8927bd4750b60687996ae29d81 : SUCCESS in 41m 12s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/aa29e2179a704a05945f9d0370821808 : SUCCESS in 5m 33s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/6298ee3d161e4679961455d93eb5c8cc : SUCCESS in 52m 00s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"659d9e2982ce51e658653586aa85cff4cc781f2d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-28 19:02:59.000000000","message":"Patch Set 25: Code-Review-1\n\n(28 comments)","accounts_in_message":[],"_revision_number":25},{"id":"ef76777a95c4c0c3d85fca6eab766d8f5b98091d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-29 17:22:37.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"a243fc78dfa9347a2ebb7e9fb33e46c34bf8f198","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-29 17:24:01.000000000","message":"Patch Set 26:\n\n(30 comments)","accounts_in_message":[],"_revision_number":26},{"id":"337468c842f627938563bf6aee74fa37030b0fa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 18:41:54.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2389c0ee0206462d98eb0cb124be2eb9\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a9dc83ec84bb47efaad0c7e5cdb1e2f3 : SUCCESS in 53m 10s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/851b1b217a204c1da8b81a819cf28d6e : SUCCESS in 48m 35s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d4ae5dc03774e5ebfa92d423dd8c01e : SUCCESS in 59m 50s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9c20c1901b1e46ffb210ac402a341315 : SUCCESS in 46m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/33bb3a49fbda44bcbcee1fe1250a5092 : SUCCESS in 56m 09s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/cb19fa48d42f4bd2b3cf46017cb1ecaf : SUCCESS in 1h 13m 41s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/913733b4c1de4f7a87eb3836a5c0891e : SUCCESS in 51m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6e321409f28a4707baeaf5082e291e95 : SUCCESS in 51m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/aa57e20faf99439391f6d0d1da75a80b : SUCCESS in 50m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab31af3f0dd3454cb5038a5995ee0d8f : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/146b5e5f30bc40fdae4c520352872b25 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d8ce92e5805402ebee624b9df108319 : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/486c829d63894ddebf2b72438ce6a551 : SUCCESS in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/13b7a4b2727b477ea2252ed8aff2c307 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8ae83ebb87747f3992d86fb28079874 : SUCCESS in 7m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ba31235206ca4596836e27c6a88791de : SUCCESS in 36m 01s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/cf9f040560f44813b2dbc7ad7c0802e6 : SUCCESS in 6m 48s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7235f703c2db45c0ac01e122206f5698 : SUCCESS in 46m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0998c458120b453ebcd2c7a36cc377f3 : SUCCESS in 5m 04s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/156d0d0893fa4c88ac7cd9eddd691a3b : SUCCESS in 37m 03s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"1cd8d45d2047df0fd491c25ab5bb6c01961c0578","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-30 15:39:00.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"c67328f789f442c5f2ba292f00181b874537bba5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 16:59:00.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3caba0667a6a4bbc977dcc2a1579e278\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c7e507a9d434468bb97b848ccd34ce0a : SUCCESS in 37m 40s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2feaaf2ec3d34c528b816a0f975cb55c : SUCCESS in 51m 40s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/124cae1979e6418b8735a58083cec6bb : SUCCESS in 45m 43s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/581937864f77474bba2638ffe607a2fd : SUCCESS in 43m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/034bcc76b07541d7b1fa34fc5ff24a65 : SUCCESS in 48m 16s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/60a6bf4ba31842fdb12253f3c9429d8e : SUCCESS in 1h 08m 39s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3447b67bffb6456abc21d46c22020980 : SUCCESS in 1h 08m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3bd9fafd6d13437b8708d77b092a586b : SUCCESS in 44m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9ed06adad7cd4d1bad048b2291dc779e : SUCCESS in 41m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57525951d06b4662a1b3bad465f34e68 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c9d2e6ed79c44348afc19b1c146132e : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61fff72224104e6d932d36e685a79072 : SUCCESS in 6m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1af528d8fc524b0999f46c727bbf2df3 : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6e143aaed4f4d148ff89c3d198180b5 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cab4890be62b4cfbb04ed2cefcd05d74 : SUCCESS in 8m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/efe5247ca7e94c46b31a26bb180dd0cc : SUCCESS in 34m 42s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/985a7136ca944685b530f768b123d460 : SUCCESS in 6m 50s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/09028ea1395a4ec89335f7c272fc0ebc : SUCCESS in 48m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/07ad004059d44bf6996811e088c87222 : SUCCESS in 4m 53s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/f1543feac2cb493ebf992dfc1e43f688 : SUCCESS in 49m 32s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"0fa954719679d3e8b910adfc881e9e1a959275e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-06-04 11:30:46.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"8c1c15a5ae394cd37b518e9cc0dfd4c9bcbe062b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 12:56:18.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6713bdf13e714216a22f0bbba06bcd75\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fe41857a80ab40e598b8754792fe6b3a : SUCCESS in 52m 19s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9efe20949d444934b67dae88d6803a0c : SUCCESS in 1h 07m 03s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8ac5ad0365243d59779c041661aa606 : SUCCESS in 1h 04m 41s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1509ef9f8f9d42bcb348af5cd14e749b : SUCCESS in 56m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/abb78fbe335642ac9de77d77f74a25e6 : SUCCESS in 48m 15s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a918fd2f724c4f94ad9e4e703a32e4a3 : SUCCESS in 1h 19m 10s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7ba9b2df2e7b4cc3a0563e022f976061 : SUCCESS in 58m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fe85b81d8c7d4b4dbff89f4090022ff5 : SUCCESS in 51m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/af81ff4d5d454c62af7c6d4eeaa28466 : SUCCESS in 55m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2475084d362048dea64624b13eb041b1 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c996d94e6d214c60b1e41303ab4003c8 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1c3a9bda63b4c08a3d52ce132336744 : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea928ede073d4a739b3eaaf943da5443 : SUCCESS in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ad0e04f9d1e94871b297e5b584e8c7c1 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c107ce5fa1c4406bb644e13e328cfc91 : SUCCESS in 6m 15s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/995fb2b7ff1b4e09beddc4d408c86796 : SUCCESS in 8m 29s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6891cbe66d8b49288e1788e114022971 : SUCCESS in 39m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8710b71640f04467a2f3eb321af470a7 : SUCCESS in 5m 56s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/4f483c95fc4e461c864bb600bfef69eb : FAILURE in 45m 15s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"ad8556c4d15210c0c5891d6caf7327d1192d749d","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-06-04 13:12:04.000000000","message":"Patch Set 28:\n\n(3 comments)","accounts_in_message":[],"_revision_number":28},{"id":"5e0e5381366ee00fa777ef4ce8cab82fd41cf76f","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-06-04 13:15:11.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"62b320d59811992f046912859d316f8b1de2e8f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-06-04 15:06:09.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":29},{"id":"36cec414048b83b6452e198ef3c980d62af11907","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-06-04 15:07:08.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"5fc25a474ce3f0f1d2c96b8112f694d52c85b6d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 16:32:09.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5effd1ee567e48b488728a34d1db04f0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3c630cb1bd3c459bbd2369f0f3617b89 : SUCCESS in 41m 18s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7df4fdf8ed234d52adf628b126fab83d : SUCCESS in 54m 55s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc9d2136eb1249d79f7574d33a46ae4b : SUCCESS in 53m 03s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b5940616fe4646b48242befad7c690e5 : SUCCESS in 49m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b96ab7eff45f41019b8e9d32fdb2e5b9 : SUCCESS in 46m 31s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8117955cfc09443d89589aaf026824d4 : SUCCESS in 1h 05m 38s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ca3e3b0fdc4f4bff9f003e50132d71a2 : SUCCESS in 1h 15m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f16bde89fbab4b449c16da7e50b5f5d9 : SUCCESS in 48m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/940a3c99cd9048e09c591140dd0491cd : SUCCESS in 50m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/266fad9c821f4cfb98802159458474f1 : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf3f8a64a5b444c38e3299cd7fa0cfe4 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a270b371addd44988927dbae6cede531 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/442764ccdd694e35921d90f82a548f42 : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/054eda0ae5cf4feba43212a741956949 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7209fcc512c40149595d7334687de56 : SUCCESS in 9m 21s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3f8b5aa459cc42c384b59eeb0de864b5 : SUCCESS in 6m 51s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/12e5eb668625422190a394854f772ebb : SUCCESS in 59m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ef4a2567e379482e95de09f15ed35de4 : SUCCESS in 4m 13s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/cae482094dc74bf2823c8714b69ba138 : SUCCESS in 40m 16s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"2ba6a25141e35ed0117d4d9e2392d6bfc873605f","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-06-13 10:59:54.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"a108778f1859ef0159227f999116e870e8b97078","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-06-28 12:13:46.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":30},{"id":"053e737cedc1f8e7e083a791a2381d33e3151001","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 13:36:45.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45115ad82d034f69b22147a34025e03a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4d97b57268614be7bd5e1e6693bbae3f : SUCCESS in 43m 27s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1268ad60cbc843fb97fbf2fb2e5422cc : SUCCESS in 58m 42s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/71c68c87122f4582b8ee1216b10f80ff : SUCCESS in 58m 00s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/00c783f42deb436d9be9d2396bb630d7 : SUCCESS in 50m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/20cdef1753114d8c9382e255a9f893c1 : SUCCESS in 47m 34s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/837707a1cf1c42d4b5a56ad936769a41 : SUCCESS in 1h 10m 28s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a030c077dec24a698fe0ff35b422c6ad : SUCCESS in 1h 16m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/89dda6cc26f94a768e8158244b717ae1 : SUCCESS in 49m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/26acde070884425dafe960351ca7a92f : SUCCESS in 43m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9770856cb43f4c38be5560422375b68e : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f28dc92a146f4b8e855cfd0e979a96f0 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1c6ce9569424f928854e775eb9c84e1 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac837799f791442ca528668ba728d619 : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/861dad9cc676483f8f6893dbc1366012 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99dd9cc479ee49a391af17f335975235 : SUCCESS in 7m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d55f2e6cf4e24aabadd482dc72d7d55c : SUCCESS in 34m 54s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/43935f6012cd458d82b5b63d31be1a7c : SUCCESS in 6m 15s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e604c75a0ec0401db23b95f42b3ae1b2 : FAILURE in 37m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/deb0127fdf944d1988ac1f2ea0eac7e4 : SUCCESS in 5m 12s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/41fa05be7a4a47a4b8145e05d6210f5e : SUCCESS in 52m 01s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"7dc0275376b684f7a66dc8ba0b52c093676d97b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-06-28 15:35:00.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":31},{"id":"84915e16e43f92e57336173f8045f6fc94205921","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 16:58:57.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/009d1a8d531e45429bbed21119fb556e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4e9fb016716240c98712a5ff974c7ad7 : SUCCESS in 39m 07s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0c7e423ef6c241a8ae5e633a64ef57da : SUCCESS in 54m 55s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/61228e879885464ea5d883d59735336e : SUCCESS in 45m 43s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b6b702c1df4a426ea14da5307545979c : SUCCESS in 39m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b3f8ddab641445c39922edde161e36a8 : SUCCESS in 41m 30s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/78f8328b0f33401f8e345582fbea0d0c : SUCCESS in 1h 13m 11s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2578b61cbd6340e7867a7d0ee56acbf3 : SUCCESS in 1h 18m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/831aac31b17e4129b9b539167f383903 : SUCCESS in 50m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b2bd86eb10324d0095606fd227a30b98 : SUCCESS in 52m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ce916735aed4cd0a8e2cef6f47917fd : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2164e239e3a24272936b0689dcf827d5 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a2bec728c6d42bbbad083dd9ecf0359 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d8e054ed114b43c8b16c482879c268c9 : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/339774d5b4d44237ab08064dba8847ef : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1e25d2a18024615809b4b71f4ed8507 : SUCCESS in 9m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/61226b6b211f4d46aadec3f0438fd226 : SUCCESS in 31m 59s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/07b3a0c4fb3d49faa13b5a8168f73fa0 : SUCCESS in 9m 03s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9b666284abf94d40879715db7cb68656 : SUCCESS in 44m 52s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8b6ba238109b48e28afd158b1146fc16 : SUCCESS in 4m 42s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/2715321a0e85493a9817c10ecf7f754b : SUCCESS in 41m 41s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"24bd1ac0ca96796228409ddab14dafbd18ac9960","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-06-28 22:49:13.000000000","message":"Patch Set 31: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":31},{"id":"6ec99dc903578f1bfcd33ad71f2cc0c20cdf8172","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-07-01 12:04:50.000000000","message":"Patch Set 31:\n\n(3 comments)","accounts_in_message":[],"_revision_number":31},{"id":"073e0a743e27692ac4e147992f5c0e345c16e2bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-07-01 12:14:46.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"b773d676c0ccd72be22f4827fe4616e066b22b43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 13:34:53.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72a431573e974bfbaf9b655c70fff544\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b54e9e22a44d4271bfe934436014da24 : SUCCESS in 46m 47s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/65ab538603e243139b8cee950f87b7ec : SUCCESS in 1h 00m 54s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/48f639fd6818446c834c9a3e2d8c191d : SUCCESS in 57m 55s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b4a7303aedfc4570b78eea40d8cbf8fb : SUCCESS in 41m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9d5e1e35b208475ba0269703f2f40c4b : SUCCESS in 44m 54s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c6efff44367242898adb1a99662ee9ff : SUCCESS in 54m 51s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5f94f393ef2a401b981c0f80d6c6ac37 : SUCCESS in 1h 13m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a202aa6a1d5d40ef817e601a5abf6288 : SUCCESS in 46m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/46b7bafc4d8844b4bdff4e13639fb9f3 : SUCCESS in 52m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0de2bb20e24846fdabb450293aa3d00d : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e6eb046edcd4eb980d90218cce3301f : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7cccc9e0c85342c5bbd22fa58686666d : SUCCESS in 5m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/29b7220c80074b7bbc34006e3a9ac294 : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/93c5186dc96c4d38b6007b11e3f7df7e : SUCCESS in 18m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08cd18138edd41a0b55e9e9614371eba : SUCCESS in 6m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0fcb8f7d048f44e7bd4f69519ef60306 : SUCCESS in 29m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1c2a62a221a24c699e40ef05cdc5e8e2 : SUCCESS in 5m 56s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/996f797c7e8e427f816f6ddeb69d5663 : FAILURE in 38m 35s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"eddac3a87124dccf9443d605b4869d418844a772","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-07-11 17:40:18.000000000","message":"Patch Set 32: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":32},{"id":"585ee2a506b1544d3ee8017c15638d81c3a9c74a","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-07-12 11:32:23.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"560f3187ddaa2cf1be3be62c9cd8b47c0cf95d58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-07-12 11:42:20.000000000","message":"Uploaded patch set 33.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":33},{"id":"7a8268b4048b4b28586931800f6d497525df9c33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 14:46:22.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3ce2117c64e4cf087800e15eadf2e5c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6270de71633b4e3087fae42411e7c92b : TIMED_OUT in 1h 13m 43s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ca1e86c4ae2d4200a84b39c9af260b95 : SUCCESS in 44m 31s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/03f45df7c68445159621e34b89eaf5fa : TIMED_OUT in 1h 13m 57s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/08b86fc1ced9435badee8c80f96cbb0b : TIMED_OUT in 1h 13m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b7bc08153cbb46dfa712fb0082d656b6 : TIMED_OUT in 1h 12m 38s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5661f7ad7fcb4666becc217c218ecd50 : TIMED_OUT in 3h 02m 58s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b8d3fa94a0ca40b28d33aecaa09e05c5 : SUCCESS in 1h 14m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1d63545009a34536bb6dcf83dc3fe793 : TIMED_OUT in 2h 02m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e87597a4bd5349b58d5125c5547e3330 : TIMED_OUT in 2h 02m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b626d5d5fee4a6caa67265bf037bf07 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e21eae4f447f450c97b7e4f8323c3996 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c6a706ce6fe4c4da10413ab70883e68 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/791b160e434449189d404b929469f4f1 : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c0436d5fc61446a498cc4a570c81971f : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/139efba599c843bc8b87a2f675c5ac10 : SUCCESS in 6m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/90f464b9c82a45d381369613c0ddd0bf : SUCCESS in 34m 12s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3a380a7cd9ee4e589a9e7844e775731d : SUCCESS in 4m 26s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/dba2ada4145b4b7da3759b6c2a83f462 : TIMED_OUT in 1h 12m 50s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"fb07fcc1474651545baa848404d669ef9249d8d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-07-23 13:53:22.000000000","message":"Uploaded patch set 34.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":34},{"id":"ccc771507d8588551dd6ed0be5b81bfa1fb7ef84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 15:15:59.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3d5883984bc4065a51af494926c87be\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0f70882a619d4796a2adc71417f56783 : SUCCESS in 41m 23s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a94f11901feb4809aace113eeabeb583 : SUCCESS in 1h 07m 35s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1574068eca14453a8e770e3bb8da5d7a : SUCCESS in 58m 40s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/aaff897db1604dc4a30f15df69bc770f : SUCCESS in 50m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7658be0250b6465aa6a4003b95cad53d : SUCCESS in 45m 42s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/14fea76a2ff54be78685bad93e6413e0 : SUCCESS in 1h 12m 46s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/86ea3d4b67744238849056a85a6434b8 : SUCCESS in 1h 15m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/948eb408b31a45f9825bbadf431733ad : SUCCESS in 48m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/819d370424f145438408c43f8a637e27 : SUCCESS in 48m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a335c804378487ab409b8a78bdf1139 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08c8403a928a4553bf69bd228f2c98e9 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9445f51e049d423fbdfc31751c809ef1 : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/750f2d0196fe41d29e716097ff5f5b4b : SUCCESS in 5m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c60bb838573b4909a076e941ef97c308 : SUCCESS in 9m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20f19bd8feb14625ad2825a858789090 : SUCCESS in 6m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d5bb124034844481a5f797df109ec57c : SUCCESS in 30m 20s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9f9273f2602141a5a0738e13dfbc2492 : SUCCESS in 4m 55s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/c0f9d1d5ccdd410d977ef573c3365325 : FAILURE in 40m 49s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"2b97281f6176963f632aac21c49d31f6d035e045","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-08-05 09:13:21.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":35},{"id":"f2a694a3e5de40554b34174d850363b464a3b97f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 10:24:47.000000000","message":"Patch Set 35: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5df3731cbb847d1a73ec3fac1c349d4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/61f292da72434449b53e3465bd7d2c49 : SUCCESS in 47m 40s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b34a6cabbddd4886afb0cbe2ecad0145 : FAILURE in 23m 35s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/85fd2a363fce49fa86eedd2af9a5b6dd : FAILURE in 8m 25s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5a8c25c87a384487b303db22f3ba469f : SUCCESS in 48m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8a346d8e4878475a80dafa80ca0e01bf : SUCCESS in 37m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/1ec7c484b84e4f1fab5e805614ee9163 : SUCCESS in 1h 02m 54s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/91dd7749f6e644ab90bd081104273eab : SUCCESS in 1h 10m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6c89439ef7364a8faa14630b358cbd92 : SUCCESS in 40m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/45c51250060f4d6985381baefee9db95 : SUCCESS in 47m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/881c62e2be6c4514ac1c767e36a02b4e : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b7f9d3ca39746ea9fab2db6e30fbbb7 : FAILURE in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21baa5b765dc46b6ae6044e0d897fe88 : SUCCESS in 5m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/76ed47e0012e49258b16032565823e08 : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f9914a19ac394cfbaf3947c9af6f9cef : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8591810fe7494b1a90663edc055e49ec : SUCCESS in 6m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e1bf65d4ae2a4ee2a0c6f819b5ff9e10 : SUCCESS in 37m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e054bef149e84726bb1d10f36ed05465 : SUCCESS in 4m 07s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/23751c8e580644b785a6f742805c9e74 : SUCCESS in 41m 02s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"0d2cb5310ef8b980c97aaf55774565b66bd209da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-08-05 13:25:43.000000000","message":"Uploaded patch set 36.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":36},{"id":"e62feab37997b8fdd6f7c9d28e1b7f72205d3d2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-08-05 13:41:21.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"8a30ee2dc4e70caf4ed10102a13c6170c4e156e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 15:08:34.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ddaea8e9c3d4a538628431b2ac4b803\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/844979fd42d04a51918a47f52b86624f : SUCCESS in 47m 19s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0296129b248646a1a3427f7d4903bf2f : SUCCESS in 58m 34s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d787dd34987a4e5bbdc3acdcfa3a1029 : SUCCESS in 58m 37s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/eabf02a31312454dbb1258c4a671cf2b : SUCCESS in 48m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d6934b1ea6a347a8b7d04e9c56acdb92 : SUCCESS in 48m 23s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/51f808cac1b6483483a2e0eb48157bdd : SUCCESS in 1h 14m 12s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4ea88f610f6949d89da6fd174674dbbc : SUCCESS in 1h 14m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ec5499e8d10a44c69ee7508404b23155 : SUCCESS in 48m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6fb0ee3a3f8b4d76b1b9969d5504ff87 : SUCCESS in 48m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2542dd4daeb64bdfaa7108413293705e : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/597a8164d6fa46c28b39b0a628f5ad5a : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd903f3a735945bdbbd5d1326cc4ad81 : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1ecceff587743a4b12ece28b21b730a : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d3ace1e0fd94f5fa7efee266ab3e4eb : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76e85a86b05b40d59eb5c48254dece5e : SUCCESS in 7m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1adea7b344a44bd1b829f8b85f4a0dd2 : SUCCESS in 31m 38s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ade214f898744074a6e7ccb7d6938f59 : SUCCESS in 4m 09s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/1112280421bc4b338eeb5c3db6e31199 : SUCCESS in 39m 14s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"1495cdb1bd4da915767401d855e55d7d4b7daf64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-08-05 17:19:08.000000000","message":"Uploaded patch set 38.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":38},{"id":"c19b98bcabfb8df48142c39cd19e3c3e288694c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 18:43:14.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cbd150fcbe142b2b008768b726a2242\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9c865fb6dd2b492f82bef4d2badcb6ba : SUCCESS in 45m 22s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7539f5eb2f5c4cf08ec3bb4699fd4721 : SUCCESS in 47m 09s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/03144d8dac3442ae9fe42d2a5c9a9a49 : SUCCESS in 43m 43s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9e3cdca3653b48a09be3ef8c274683ec : SUCCESS in 38m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/22e80476cc894886ab6404e8d6a8de97 : SUCCESS in 35m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b5792dbbae37446b85d1f2f951d7a2b5 : SUCCESS in 58m 59s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f991220423c2477db92e84e13ce04f7c : SUCCESS in 1h 09m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cc29e1588ed64269976d5a0f80132ab8 : SUCCESS in 35m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/dc3d67fc100b47d3b43e3d30f3725c2c : SUCCESS in 39m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/445ac1ff343b48498e8b8585805280d9 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b3320407100459088411b82a02bc426 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5af70c94b3704ddc8d15ffaf059c79ae : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f1f28596631d4971ac6cd08f08685458 : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7eca5ebca7864d48b157869b7f22e008 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5c2c37cd955490eaab598a770dbbec5 : SUCCESS in 7m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/35dfef565a1448d884a68381b6fe2d3c : SUCCESS in 37m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/674489ffefdc41039555526e6c70ff9c : SUCCESS in 4m 26s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/6ac16d06f5784197806e29012597fe36 : SUCCESS in 47m 41s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"8e1735beafffd4e9dad26ac0a424d5186fe59cc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-08-09 16:07:38.000000000","message":"Uploaded patch set 39.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":39},{"id":"52c074370d6a54cd3ff8138febe6d30b3a464ba6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 17:23:20.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3863c5bd162c4d57966b89d48e9d6f0b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/98e37e1dd6304beda790fce6817c9b79 : SUCCESS in 42m 38s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/da732620feb2489c9683b23430e63f16 : SUCCESS in 46m 48s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/381cb2be72784b9793f5aa8fcb2042f7 : SUCCESS in 45m 36s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d744c7dbefb74e36889800c22c4011aa : SUCCESS in 39m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e49e56339a744b28ac5ea30253e38364 : SUCCESS in 36m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d71ae99499954b0fa1f69dcba5502233 : SUCCESS in 1h 09m 18s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/214c483a78654ddea7580f02a3cdf141 : SUCCESS in 1h 02m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/be1a9359bce54ebc9ee0beb01895bc81 : SUCCESS in 38m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/76878059fdd94a13b8eb12dc9f6118cf : SUCCESS in 51m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2e83d85bfb64f84a519bef668c11496 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a31480c3928445ab0c56af63d47466f : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/669188e9b5f14956b5114aaca85d2199 : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc70e00d5d574363a5bae66eaf897641 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2af184fa2edf4530aa8aa3c2bb5db947 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae3ab16e8bbf4635b6f7841f16144bf4 : SUCCESS in 6m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/89df7c03c19e4377a00d3e290f4d91cd : SUCCESS in 33m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/140d32d8947449dfb60f6fb85eb05c08 : SUCCESS in 5m 01s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/5385a58398e04cb88b083f7be582ecb1 : SUCCESS in 54m 09s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"5112d04a7263940c0ff5680339ba343f65b40d37","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-13 22:21:17.000000000","message":"Patch Set 39: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"893a029be9365634f123fe6bcd427bebb8d2f99c","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-08-14 10:31:38.000000000","message":"Patch Set 39:\n\n(2 comments)","accounts_in_message":[],"_revision_number":39},{"id":"2e92d01709ef698183d8283814b539e113ce90db","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-27 23:34:25.000000000","message":"Patch Set 39: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"b307e427e676cb74ca50721ee20aeb4b60b199d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 16:26:29.000000000","message":"Patch Set 39: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":39},{"id":"9574e98bed822996b406b715db18a5e505404d19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 17:43:10.000000000","message":"Patch Set 39: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d78a1e1e92d64f2fbddd5f2b6f04a0ef\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/21d1a09daad34318908036f68accdae5 : SUCCESS in 52m 00s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8d46a695d76d4b0abbb8bad79b8569f7 : SUCCESS in 45m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c6b1bc034beb4599bc87e7e63e757126 : SUCCESS in 50m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a55222f5848b481999b30325f43ec279 : SUCCESS in 1h 10m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4c59551cba6c4c3dbf1bdff0323c0271 : SUCCESS in 47m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bfa5872ab3144acc84dc7d5c8ea35d64 : SUCCESS in 49m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61846e3923184d2fac215db08e5b55c3 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1be4a7bb26c42a18133e2484d5bf1bd : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/320783b3bb9e471cbfb56aac589fc2d2 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7923899643045b7a59e161f682c6cc7 : SUCCESS in 6m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/223b8aaa0999415a8bb094b3b901b885 : SUCCESS in 30m 44s","accounts_in_message":[],"_revision_number":39},{"id":"e8f7ebfcc7aaeecc415a5af5787519902c05590f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 17:43:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":39},{"id":"dc955f870d7a377252d162e9656e8088a6f9d1c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 17:44:11.000000000","message":"Patch Set 39:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cf4432c3cef44a681533a154e9f56b7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/020ccc9bdd57452ea50ebe66c5adb449 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":39}],"current_revision_number":39,"current_revision":"0e9294aca19d1909209a8ef730fda60f9914df0f","revisions":{"2da1d7822cbe44c977fd745cc8bf2a81e638c5fd":{"kind":"REWORK","_number":1,"created":"2022-09-15 20:31:06.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/78/857978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/1"}}},"commit":{"parents":[{"commit":"93dee6a3ff44fb7470b3008e8fbbaf99822bbe82","subject":"Fix Redis connection over TLS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/93dee6a3ff44fb7470b3008e8fbbaf99822bbe82"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:31:05.000000000","tz":0},"subject":"WIP Multi-pools implementation","message":"WIP Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2da1d7822cbe44c977fd745cc8bf2a81e638c5fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2da1d7822cbe44c977fd745cc8bf2a81e638c5fd"}]},"branch":"refs/heads/master"},"28198b4a85d0d4b595c0020490a2c1c5d04e193f":{"kind":"REWORK","_number":2,"created":"2022-10-24 21:47:10.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/78/857978/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/2"}}},"commit":{"parents":[{"commit":"06b297eaf39b084a7f999e32c41f16eefc3ad859","subject":"Merge \"CI: Move queue configuration into project-template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06b297eaf39b084a7f999e32c41f16eefc3ad859"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"dkehn","email":"dekehn@gmail.com","date":"2022-10-24 21:27:04.000000000","tz":0},"subject":"WIP Multi-pools implementation","message":"WIP Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28198b4a85d0d4b595c0020490a2c1c5d04e193f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28198b4a85d0d4b595c0020490a2c1c5d04e193f"}]},"branch":"refs/heads/master"},"d6132df169e43eb207e98764c1d9d5d98cac4382":{"kind":"REWORK","_number":3,"created":"2022-10-25 21:21:47.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/78/857978/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/3"}}},"commit":{"parents":[{"commit":"06b297eaf39b084a7f999e32c41f16eefc3ad859","subject":"Merge \"CI: Move queue configuration into project-template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06b297eaf39b084a7f999e32c41f16eefc3ad859"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"dkehn","email":"dekehn@gmail.com","date":"2022-10-25 21:20:43.000000000","tz":0},"subject":"WIP Multi-pools implementation","message":"WIP Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d6132df169e43eb207e98764c1d9d5d98cac4382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d6132df169e43eb207e98764c1d9d5d98cac4382"}]},"branch":"refs/heads/master"},"67617c536a6e40375101677ccf49eb8ed36ca3dd":{"kind":"REWORK","_number":4,"created":"2022-10-25 22:45:25.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/78/857978/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/4"}}},"commit":{"parents":[{"commit":"06b297eaf39b084a7f999e32c41f16eefc3ad859","subject":"Merge \"CI: Move queue configuration into project-template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06b297eaf39b084a7f999e32c41f16eefc3ad859"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"dkehn","email":"dekehn@gmail.com","date":"2022-10-25 22:44:44.000000000","tz":0},"subject":"WIP Multi-pools implementation","message":"WIP Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/67617c536a6e40375101677ccf49eb8ed36ca3dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/67617c536a6e40375101677ccf49eb8ed36ca3dd"}]},"branch":"refs/heads/master"},"c47336e1595a30d4ddbeaaad51780bb9ec1a6794":{"kind":"REWORK","_number":5,"created":"2022-11-18 22:50:41.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/78/857978/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/5"}}},"commit":{"parents":[{"commit":"06b297eaf39b084a7f999e32c41f16eefc3ad859","subject":"Merge \"CI: Move queue configuration into project-template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06b297eaf39b084a7f999e32c41f16eefc3ad859"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"dkehn","email":"dekehn@gmail.com","date":"2022-11-18 22:48:34.000000000","tz":0},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c47336e1595a30d4ddbeaaad51780bb9ec1a6794"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c47336e1595a30d4ddbeaaad51780bb9ec1a6794"}]},"branch":"refs/heads/master"},"aaccab50a1ce811a3948dd9563a5d265e71a113e":{"kind":"REWORK","_number":6,"created":"2022-12-01 22:59:54.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/78/857978/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/6"}}},"commit":{"parents":[{"commit":"06b297eaf39b084a7f999e32c41f16eefc3ad859","subject":"Merge \"CI: Move queue configuration into project-template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06b297eaf39b084a7f999e32c41f16eefc3ad859"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"dkehn","email":"dekehn@gmail.com","date":"2022-12-01 22:57:27.000000000","tz":0},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aaccab50a1ce811a3948dd9563a5d265e71a113e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aaccab50a1ce811a3948dd9563a5d265e71a113e"}]},"branch":"refs/heads/master"},"492cd8a5fad5a744db52123e3edb602faaa427d7":{"kind":"REWORK","_number":7,"created":"2023-01-17 20:16:53.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/78/857978/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/7"}}},"commit":{"parents":[{"commit":"de000517fb6435526d808c137512a21ea5561580","subject":"Fix missing space IllegalChildZone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/de000517fb6435526d808c137512a21ea5561580"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"dkehn","email":"dekehn@gmail.com","date":"2023-01-17 20:00:42.000000000","tz":0},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/492cd8a5fad5a744db52123e3edb602faaa427d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/492cd8a5fad5a744db52123e3edb602faaa427d7"}]},"branch":"refs/heads/master"},"4900780f8e901a4059a8c5b837cc85258092905d":{"kind":"REWORK","_number":8,"created":"2023-01-18 22:26:41.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/78/857978/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/8"}}},"commit":{"parents":[{"commit":"de000517fb6435526d808c137512a21ea5561580","subject":"Fix missing space IllegalChildZone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/de000517fb6435526d808c137512a21ea5561580"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"dkehn","email":"dekehn@gmail.com","date":"2023-01-18 22:24:05.000000000","tz":0},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4900780f8e901a4059a8c5b837cc85258092905d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4900780f8e901a4059a8c5b837cc85258092905d"}]},"branch":"refs/heads/master"},"b4fb2f68a09e32931b3ce2f25b5cd88ae819fc1f":{"kind":"REWORK","_number":9,"created":"2023-01-30 21:22:28.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/78/857978/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/9"}}},"commit":{"parents":[{"commit":"de000517fb6435526d808c137512a21ea5561580","subject":"Fix missing space IllegalChildZone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/de000517fb6435526d808c137512a21ea5561580"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"dkehn","email":"dekehn@gmail.com","date":"2023-01-30 21:16:14.000000000","tz":0},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b4fb2f68a09e32931b3ce2f25b5cd88ae819fc1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b4fb2f68a09e32931b3ce2f25b5cd88ae819fc1f"}]},"branch":"refs/heads/master"},"eaf3637fa622becad5e2a4393ea0c0e132c7f919":{"kind":"REWORK","_number":10,"created":"2023-01-30 23:35:36.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/78/857978/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/10"}}},"commit":{"parents":[{"commit":"a9cf887da11f18dbda7c6c1332d430680a9edc14","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a9cf887da11f18dbda7c6c1332d430680a9edc14"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"dkehn","email":"dekehn@gmail.com","date":"2023-01-30 22:21:27.000000000","tz":0},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eaf3637fa622becad5e2a4393ea0c0e132c7f919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eaf3637fa622becad5e2a4393ea0c0e132c7f919"}]},"branch":"refs/heads/master"},"f2c91dafd269d04bcf3b24dfede20646f0b65ee1":{"kind":"REWORK","_number":11,"created":"2023-04-07 15:58:02.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/78/857978/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/11"}}},"commit":{"parents":[{"commit":"f4ce71c8f80778b555eb1c0a29b67b8c32ed5f97","subject":"Merge \"Move to a batch model for incrementing serial\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f4ce71c8f80778b555eb1c0a29b67b8c32ed5f97"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-07 15:57:57.000000000","tz":0},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f2c91dafd269d04bcf3b24dfede20646f0b65ee1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f2c91dafd269d04bcf3b24dfede20646f0b65ee1"}]},"branch":"refs/heads/master"},"eebc2ffb50698952287293dd712fac0a4ef2040e":{"kind":"REWORK","_number":12,"created":"2024-04-25 14:24:13.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/12"}}},"commit":{"parents":[{"commit":"bbd1c106f1d5f68aee989fb8ab9c442a92f832b0","subject":"Merge \"Cleanup deprecated and ineffective api options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bbd1c106f1d5f68aee989fb8ab9c442a92f832b0"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-04-25 14:22:32.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eebc2ffb50698952287293dd712fac0a4ef2040e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eebc2ffb50698952287293dd712fac0a4ef2040e"}]},"branch":"refs/heads/master"},"7ed51793bfb642379ce549f40b6bed351a58e0c9":{"kind":"REWORK","_number":13,"created":"2024-04-25 14:26:50.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/13"}}},"commit":{"parents":[{"commit":"bbd1c106f1d5f68aee989fb8ab9c442a92f832b0","subject":"Merge \"Cleanup deprecated and ineffective api options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bbd1c106f1d5f68aee989fb8ab9c442a92f832b0"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-04-25 14:26:38.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7ed51793bfb642379ce549f40b6bed351a58e0c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7ed51793bfb642379ce549f40b6bed351a58e0c9"}]},"branch":"refs/heads/master"},"6d8759e90e9e3ca32cd2e08580c057dda08cdf09":{"kind":"REWORK","_number":14,"created":"2024-04-29 11:23:51.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/14"}}},"commit":{"parents":[{"commit":"bbd1c106f1d5f68aee989fb8ab9c442a92f832b0","subject":"Merge \"Cleanup deprecated and ineffective api options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bbd1c106f1d5f68aee989fb8ab9c442a92f832b0"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-04-29 11:23:45.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6d8759e90e9e3ca32cd2e08580c057dda08cdf09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6d8759e90e9e3ca32cd2e08580c057dda08cdf09"}]},"branch":"refs/heads/master"},"7467825ec7afb040b8ee3667cc9d9458b87e2b7e":{"kind":"REWORK","_number":15,"created":"2024-04-30 13:27:01.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/15"}}},"commit":{"parents":[{"commit":"bbd1c106f1d5f68aee989fb8ab9c442a92f832b0","subject":"Merge \"Cleanup deprecated and ineffective api options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bbd1c106f1d5f68aee989fb8ab9c442a92f832b0"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-04-30 13:26:51.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7467825ec7afb040b8ee3667cc9d9458b87e2b7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7467825ec7afb040b8ee3667cc9d9458b87e2b7e"}]},"branch":"refs/heads/master"},"708441c52e5cb34269c7249c85e8cb9021e90dce":{"kind":"REWORK","_number":16,"created":"2024-04-30 14:46:26.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/16"}}},"commit":{"parents":[{"commit":"bbd1c106f1d5f68aee989fb8ab9c442a92f832b0","subject":"Merge \"Cleanup deprecated and ineffective api options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bbd1c106f1d5f68aee989fb8ab9c442a92f832b0"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-04-30 14:46:19.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/708441c52e5cb34269c7249c85e8cb9021e90dce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/708441c52e5cb34269c7249c85e8cb9021e90dce"}]},"branch":"refs/heads/master"},"c4c9d9f0c35c40f9caacd0a40dd7f8e3bc53d1f7":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2024-05-03 09:49:56.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/17"}}},"commit":{"parents":[{"commit":"4229f25a19456cc9107bc3740946373503717837","subject":"Fix YAMLAdapter when optional attributes are missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4229f25a19456cc9107bc3740946373503717837"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 09:44:25.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c4c9d9f0c35c40f9caacd0a40dd7f8e3bc53d1f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c4c9d9f0c35c40f9caacd0a40dd7f8e3bc53d1f7"}]},"branch":"refs/heads/master"},"d7b1877ec1d514e2db60b2eaeba7c928347e3390":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2024-05-08 09:20:18.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/18"}}},"commit":{"parents":[{"commit":"e7d890ed20f7396b1a934ec1f99441f2fe8b5d7e","subject":"Fix YAMLAdapter when optional attributes are missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e7d890ed20f7396b1a934ec1f99441f2fe8b5d7e"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2024-05-08 09:20:18.000000000","tz":0},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d7b1877ec1d514e2db60b2eaeba7c928347e3390"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d7b1877ec1d514e2db60b2eaeba7c928347e3390"}]},"branch":"refs/heads/master","description":"Rebase"},"363e7c228d8deb011543373ba0502b72b708b4fa":{"kind":"REWORK","_number":19,"created":"2024-05-08 09:50:27.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/19"}}},"commit":{"parents":[{"commit":"e7d890ed20f7396b1a934ec1f99441f2fe8b5d7e","subject":"Fix YAMLAdapter when optional attributes are missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e7d890ed20f7396b1a934ec1f99441f2fe8b5d7e"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-08 09:48:59.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/363e7c228d8deb011543373ba0502b72b708b4fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/363e7c228d8deb011543373ba0502b72b708b4fa"}]},"branch":"refs/heads/master"},"99ed40422d96ce1a5b82c5ceac610689f7940f72":{"kind":"REWORK","_number":20,"created":"2024-05-08 12:38:15.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/20"}}},"commit":{"parents":[{"commit":"e7d890ed20f7396b1a934ec1f99441f2fe8b5d7e","subject":"Fix YAMLAdapter when optional attributes are missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e7d890ed20f7396b1a934ec1f99441f2fe8b5d7e"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-08 12:38:06.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/99ed40422d96ce1a5b82c5ceac610689f7940f72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/99ed40422d96ce1a5b82c5ceac610689f7940f72"}]},"branch":"refs/heads/master"},"7f66e63348d52bfbc131c7e5e437aedee22a7487":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2024-05-10 11:23:29.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/21"}}},"commit":{"parents":[{"commit":"baa88827091b8c77ad766ccadb13fc2dd073e666","subject":"Fix YAMLAdapter when optional attributes are missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/baa88827091b8c77ad766ccadb13fc2dd073e666"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2024-05-10 11:23:29.000000000","tz":0},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7f66e63348d52bfbc131c7e5e437aedee22a7487"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7f66e63348d52bfbc131c7e5e437aedee22a7487"}]},"branch":"refs/heads/master","description":"Rebase"},"2cac494273299a736e616a1db4b64949aed0df86":{"kind":"REWORK","_number":22,"created":"2024-05-20 17:35:48.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/22"}}},"commit":{"parents":[{"commit":"ad561565fdae85e386a70a60b033693444f39bbd","subject":"Fix YAMLAdapter when optional attributes are missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ad561565fdae85e386a70a60b033693444f39bbd"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-20 17:35:26.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2cac494273299a736e616a1db4b64949aed0df86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2cac494273299a736e616a1db4b64949aed0df86"}]},"branch":"refs/heads/master"},"5c352c9681150c452b0f00b448a7ff53a053ea8d":{"kind":"REWORK","_number":23,"created":"2024-05-21 08:34:02.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/23"}}},"commit":{"parents":[{"commit":"6dd9aaf13b693c831aabbfac5406757b8ea96fc6","subject":"Fix YAMLAdapter when optional attributes are missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6dd9aaf13b693c831aabbfac5406757b8ea96fc6"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-21 08:33:51.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5c352c9681150c452b0f00b448a7ff53a053ea8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5c352c9681150c452b0f00b448a7ff53a053ea8d"}]},"branch":"refs/heads/master"},"bd19b7ff91110d99a1fc49ecb47d0c786a490d9d":{"kind":"REWORK","_number":24,"created":"2024-05-21 13:47:38.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/24"}}},"commit":{"parents":[{"commit":"6dd9aaf13b693c831aabbfac5406757b8ea96fc6","subject":"Fix YAMLAdapter when optional attributes are missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6dd9aaf13b693c831aabbfac5406757b8ea96fc6"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2024-05-21 13:47:38.000000000","tz":0},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bd19b7ff91110d99a1fc49ecb47d0c786a490d9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bd19b7ff91110d99a1fc49ecb47d0c786a490d9d"}]},"branch":"refs/heads/master"},"3dc94ef48553c110b723cbf476c7d7dddb0fdac6":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2024-05-28 12:11:58.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/25"}}},"commit":{"parents":[{"commit":"6cc0a7d8876291f874edaac2ff86b7d7a84a738a","subject":"Fix YAMLAdapter when optional attributes are missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6cc0a7d8876291f874edaac2ff86b7d7a84a738a"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2024-05-28 12:11:58.000000000","tz":0},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3dc94ef48553c110b723cbf476c7d7dddb0fdac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3dc94ef48553c110b723cbf476c7d7dddb0fdac6"}]},"branch":"refs/heads/master","description":"Rebase"},"1d861cbca468c6cd777c8fbd4ee615f34a067543":{"kind":"REWORK","_number":26,"created":"2024-05-29 17:22:37.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/26"}}},"commit":{"parents":[{"commit":"6cc0a7d8876291f874edaac2ff86b7d7a84a738a","subject":"Fix YAMLAdapter when optional attributes are missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6cc0a7d8876291f874edaac2ff86b7d7a84a738a"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-29 17:22:17.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nCo-Authored-By: Omer Schwartz \u003coschwart@redhat.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1d861cbca468c6cd777c8fbd4ee615f34a067543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1d861cbca468c6cd777c8fbd4ee615f34a067543"}]},"branch":"refs/heads/master"},"af9cdadb0197943cb7629712bde981c80b97d7f1":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2024-05-30 15:39:00.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/27"}}},"commit":{"parents":[{"commit":"dee0fac35de1f799a3525b602feefe257b808f65","subject":"Add designate-tempest-plugin under devstack_plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dee0fac35de1f799a3525b602feefe257b808f65"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-30 15:38:33.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nCo-Authored-By: Omer Schwartz \u003coschwart@redhat.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af9cdadb0197943cb7629712bde981c80b97d7f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af9cdadb0197943cb7629712bde981c80b97d7f1"}]},"branch":"refs/heads/master"},"1f22935212c407db011d62286ab5a22127f09a27":{"kind":"REWORK","_number":28,"created":"2024-06-04 11:30:46.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/28"}}},"commit":{"parents":[{"commit":"dee0fac35de1f799a3525b602feefe257b808f65","subject":"Add designate-tempest-plugin under devstack_plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dee0fac35de1f799a3525b602feefe257b808f65"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-06-04 11:30:29.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nCo-Authored-By: Omer Schwartz \u003coschwart@redhat.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1f22935212c407db011d62286ab5a22127f09a27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1f22935212c407db011d62286ab5a22127f09a27"}]},"branch":"refs/heads/master"},"1d7f5e598915421d834737c3757e3b727d5fce55":{"kind":"REWORK","_number":29,"created":"2024-06-04 15:06:09.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/29"}}},"commit":{"parents":[{"commit":"dee0fac35de1f799a3525b602feefe257b808f65","subject":"Add designate-tempest-plugin under devstack_plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dee0fac35de1f799a3525b602feefe257b808f65"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-06-04 14:40:29.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nCo-Authored-By: Omer Schwartz \u003coschwart@redhat.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1d7f5e598915421d834737c3757e3b727d5fce55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1d7f5e598915421d834737c3757e3b727d5fce55"}]},"branch":"refs/heads/master"},"b219aa27ed2fd99c59915f8a16743159bfbe489a":{"kind":"REWORK","_number":30,"created":"2024-06-28 12:13:46.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/30"}}},"commit":{"parents":[{"commit":"dee0fac35de1f799a3525b602feefe257b808f65","subject":"Add designate-tempest-plugin under devstack_plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dee0fac35de1f799a3525b602feefe257b808f65"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-06-28 12:13:33.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nCo-Authored-By: Omer Schwartz \u003coschwart@redhat.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b219aa27ed2fd99c59915f8a16743159bfbe489a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b219aa27ed2fd99c59915f8a16743159bfbe489a"}]},"branch":"refs/heads/master"},"8cd36485172ea2f8714a1aac48b311c14ad1d4cc":{"kind":"REWORK","_number":31,"created":"2024-06-28 15:35:00.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/31"}}},"commit":{"parents":[{"commit":"dee0fac35de1f799a3525b602feefe257b808f65","subject":"Add designate-tempest-plugin under devstack_plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dee0fac35de1f799a3525b602feefe257b808f65"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-06-28 15:30:28.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nCo-Authored-By: Omer Schwartz \u003coschwart@redhat.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8cd36485172ea2f8714a1aac48b311c14ad1d4cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8cd36485172ea2f8714a1aac48b311c14ad1d4cc"}]},"branch":"refs/heads/master"},"fac644888ae33ae5046bf6b204ae3e8a4b6fda33":{"kind":"REWORK","_number":32,"created":"2024-07-01 12:14:46.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/32"}}},"commit":{"parents":[{"commit":"706db9be728ceae0f4f3f670da029e3bcea892fb","subject":"Add designate-tempest-plugin under devstack_plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/706db9be728ceae0f4f3f670da029e3bcea892fb"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-07-01 12:14:18.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nCo-Authored-By: Omer Schwartz \u003coschwart@redhat.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fac644888ae33ae5046bf6b204ae3e8a4b6fda33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fac644888ae33ae5046bf6b204ae3e8a4b6fda33"}]},"branch":"refs/heads/master"},"524ec36191da92cc75f2d507d0a8036651a0fd4b":{"kind":"REWORK","_number":33,"created":"2024-07-12 11:42:20.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/33"}}},"commit":{"parents":[{"commit":"706db9be728ceae0f4f3f670da029e3bcea892fb","subject":"Add designate-tempest-plugin under devstack_plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/706db9be728ceae0f4f3f670da029e3bcea892fb"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-07-12 11:42:06.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nCo-Authored-By: Omer Schwartz \u003coschwart@redhat.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/524ec36191da92cc75f2d507d0a8036651a0fd4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/524ec36191da92cc75f2d507d0a8036651a0fd4b"}]},"branch":"refs/heads/master"},"98b70d3a0bb244a7b862f0de0a2a1b9f7f41d6a3":{"kind":"REWORK","_number":34,"created":"2024-07-23 13:53:22.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/34"}}},"commit":{"parents":[{"commit":"706db9be728ceae0f4f3f670da029e3bcea892fb","subject":"Add designate-tempest-plugin under devstack_plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/706db9be728ceae0f4f3f670da029e3bcea892fb"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-07-23 13:53:10.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nCo-Authored-By: Omer Schwartz \u003coschwart@redhat.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/98b70d3a0bb244a7b862f0de0a2a1b9f7f41d6a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/98b70d3a0bb244a7b862f0de0a2a1b9f7f41d6a3"}]},"branch":"refs/heads/master"},"499415bb26208e665265daaee5801cd2eeabfa51":{"kind":"REWORK","_number":35,"created":"2024-08-05 09:13:21.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/35"}}},"commit":{"parents":[{"commit":"706db9be728ceae0f4f3f670da029e3bcea892fb","subject":"Add designate-tempest-plugin under devstack_plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/706db9be728ceae0f4f3f670da029e3bcea892fb"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-08-05 09:13:06.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nCo-Authored-By: Omer Schwartz \u003coschwart@redhat.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/499415bb26208e665265daaee5801cd2eeabfa51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/499415bb26208e665265daaee5801cd2eeabfa51"}]},"branch":"refs/heads/master"},"ebe5c4047a8aef6ac6a8a2b7eb7525209d8a20b7":{"kind":"REWORK","_number":36,"created":"2024-08-05 13:25:43.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/36"}}},"commit":{"parents":[{"commit":"706db9be728ceae0f4f3f670da029e3bcea892fb","subject":"Add designate-tempest-plugin under devstack_plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/706db9be728ceae0f4f3f670da029e3bcea892fb"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-08-05 13:25:34.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nCo-Authored-By: Omer Schwartz \u003coschwart@redhat.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ebe5c4047a8aef6ac6a8a2b7eb7525209d8a20b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ebe5c4047a8aef6ac6a8a2b7eb7525209d8a20b7"}]},"branch":"refs/heads/master"},"9b29f05dad8cb8b953550abef1600fa5ee9e0805":{"kind":"REWORK","_number":37,"created":"2024-08-05 13:41:21.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/37"}}},"commit":{"parents":[{"commit":"706db9be728ceae0f4f3f670da029e3bcea892fb","subject":"Add designate-tempest-plugin under devstack_plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/706db9be728ceae0f4f3f670da029e3bcea892fb"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-08-05 13:41:12.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nCo-Authored-By: Omer Schwartz \u003coschwart@redhat.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9b29f05dad8cb8b953550abef1600fa5ee9e0805"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9b29f05dad8cb8b953550abef1600fa5ee9e0805"}]},"branch":"refs/heads/master"},"913fc0c58ffbc40343093762744f97a1657397ed":{"kind":"REWORK","_number":38,"created":"2024-08-05 17:19:08.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/38"}}},"commit":{"parents":[{"commit":"706db9be728ceae0f4f3f670da029e3bcea892fb","subject":"Add designate-tempest-plugin under devstack_plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/706db9be728ceae0f4f3f670da029e3bcea892fb"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-08-05 17:18:57.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nCo-Authored-By: Omer Schwartz \u003coschwart@redhat.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/913fc0c58ffbc40343093762744f97a1657397ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/913fc0c58ffbc40343093762744f97a1657397ed"}]},"branch":"refs/heads/master"},"0e9294aca19d1909209a8ef730fda60f9914df0f":{"kind":"REWORK","_number":39,"created":"2024-08-09 16:07:38.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/857978/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/857978/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/857978/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/857978/39"}}},"commit":{"parents":[{"commit":"706db9be728ceae0f4f3f670da029e3bcea892fb","subject":"Add designate-tempest-plugin under devstack_plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/706db9be728ceae0f4f3f670da029e3bcea892fb"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-09-15 20:28:18.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-08-09 16:06:54.000000000","tz":120},"subject":"Multi-pools implementation","message":"Multi-pools implementation\n\nThis sets up 2 pools with an instance of named running in both. The\nsecond is denoted by /etc/bind-2 and var/cache/named-2 as its\nlocations. The build up of the /etc/designate/pools.yaml reflects\nthe changes.\n\nCo-Authored-By: Don Kehn \u003cdekehn@gmail.com\u003e\nCo-Authored-By: Omer Schwartz \u003coschwart@redhat.com\u003e\nChange-Id: Icf73e730b31ab26b8be65347239636c0137ab4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0e9294aca19d1909209a8ef730fda60f9914df0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0e9294aca19d1909209a8ef730fda60f9914df0f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
