)]}'
{"id":"openstack%2Fdesignate~858381","triplet_id":"openstack%2Fdesignate~stable%2Fvictoria~I94324104a7ed89e757d48bb250d8b1132026ce46","project":"openstack/designate","branch":"stable/victoria","topic":"fix-queue-definition","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-09-20 00:52:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I94324104a7ed89e757d48bb250d8b1132026ce46","subject":"Update zuul queue configuration","status":"MERGED","created":"2022-09-19 13:34:43.000000000","updated":"2022-09-20 00:54:01.000000000","submitted":"2022-09-20 00:52:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858381-fix-queue-definition","meta_rev_id":"439f0808115f3878778ddcd465976575ca1c5bad","_number":858381,"virtual_id_number":858381,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-20 00:52:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-09-19 23:56:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2022-09-19 23:56:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-19 13:34:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2022-09-19 14:29:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-19 23:56:45.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"73ebbac9d584f6602e4784f0f97be851f11fbb03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-09-19 13:34:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51b33d2ae6619af309270ce011dd5ce48ceb3698","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 14:29:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f7f1e100acb4882bf33798f140ef361\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/703a830cabb14a7cbab88c0f5ca3d204 : SUCCESS in 47m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/90085d511e944971958f9d970ed30f90 : SUCCESS in 47m 40s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/715689c2e6064c0a870093a15cb70e87 : SUCCESS in 51m 15s (non-voting)\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/313e8103052f42f1968e626e4c127e11 : SUCCESS in 39m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e2dae79670cd4bacbe18088076011847 : SUCCESS in 33m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c48e3b43b9ce47c8901aa526bf8c6fae : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6515983e0a64d11a9e5c81d4d417a47 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63ae25c9cee64cdb9af88014e8591c96 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1e299bad7744ee7b5db08107c6753a5 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87f2e04f3d864ede9fb3579d2afdae5a : SUCCESS in 6m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3ea5bafb5fcf46fc8fd88213f60aed16 : SUCCESS in 2m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbc2b3f14e2d42d09c2a7a7ac9101ed0 : SUCCESS in 6m 56s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/94caae12ae5e4378aef0ffd3532f57f9 : SUCCESS in 33m 55s","accounts_in_message":[],"_revision_number":1},{"id":"776e552808f809917edf4001d75a65321eb0d390","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-09-19 23:56:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a9bbc1cda9d17f523f248740bb225f623f9d7665","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 23:57:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"077cee38917e7fd646ae87da7ab7628808cf8082","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 00:52:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d66c865ac03e4cb180e27b8ea2488f2c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6e756512e5e94a089ff3e7e4d9df4614 : SUCCESS in 40m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/887220836cd8464e80fc410ef9565416 : SUCCESS in 49m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/95b852a29b74462fa320340c8d4c41a6 : SUCCESS in 43m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d0b36639bc9c4e7c81fe6e6ac99987b9 : SUCCESS in 46m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/123d5bb6f6464878a4f31741720e8ee7 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62746eb1268e4058a2dd26079ba470f6 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ef3514d08df498cb9153095159082fa : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54ff7a9b424041d5a0e8e0cdd54ef7e7 : SUCCESS in 8m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/01e4f8ce4c69476abdae458b99781388 : SUCCESS in 1m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa919e5fba6046c494824bfd2b896c91 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"52a465c87e3ab1f9a64249e04e091b66c393bf92","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 00:52:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"439f0808115f3878778ddcd465976575ca1c5bad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 00:54:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b1fd6a8879b4331b8575b2a6e5d7fe0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c404dda4d18490cad89f36278442b54 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"099795d2e1497c4043158ffe92fbaa9fbe76ddac","revisions":{"099795d2e1497c4043158ffe92fbaa9fbe76ddac":{"kind":"REWORK","_number":1,"created":"2022-09-19 13:34:43.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/81/858381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/81/858381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/858381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/858381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/858381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/81/858381/1"}}},"commit":{"parents":[{"commit":"a74c2389bf587fc6a36c72d87fba6c958cef2141","subject":"Fix zone update when adding new Bind9 target to pool.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a74c2389bf587fc6a36c72d87fba6c958cef2141"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-19 16:10:12.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-09-19 13:34:09.000000000","tz":120},"subject":"Update zuul queue configuration","message":"Update zuul queue configuration\n\nZuul is changing how queues are declared[1]. This patch updates Designate for the new way to define the queue.\n\n[1] https://lists.zuul-ci.org/pipermail/zuul-discuss/2022-May/001801.html\n\nChange-Id: I94324104a7ed89e757d48bb250d8b1132026ce46\n(cherry picked from commit fe9fb45c21a7be2f88600e1f29fd37f56244cc1a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/099795d2e1497c4043158ffe92fbaa9fbe76ddac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/099795d2e1497c4043158ffe92fbaa9fbe76ddac"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
