)]}'
{"id":"openstack%2Fdesignate~858561","triplet_id":"openstack%2Fdesignate~master~I3e3e1a4701c52003e4adf2738b1ca8598d4cae47","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-09-20 22:35:16.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I3e3e1a4701c52003e4adf2738b1ca8598d4cae47","subject":"Fix minimum versions in requirements.txt","status":"MERGED","created":"2022-09-20 17:54:13.000000000","updated":"2022-09-20 23:51:59.000000000","submitted":"2022-09-20 23:50:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858561","meta_rev_id":"cee934746b04ddecc404712e4568a784d60ea5ee","_number":858561,"virtual_id_number":858561,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-20 23:50:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2022-09-20 18:19:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2022-09-20 22:35:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2022-09-20 22:35:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-20 18:19:06.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2022-09-20 19:17:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e4d62360a4d4e1e527f2555b2934f1ac785581f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-09-20 17:54:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb99bdc447287cbd6fdd458316a211350a01618f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-09-20 17:54:32.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b1ff0cdc080d2ad5c0530c0f879356a6514f475e","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2022-09-20 18:19:06.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"688c885eaf5c16f5b2d11c5195526c480e3919a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 19:17:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dc967b8b3984874aae585fe57bd3eef\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b0649781fa8548caa5b2d12fe1f84c6f : SUCCESS in 41m 45s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b3b393c7398740c1a4e59e61bfb43e94 : SUCCESS in 1h 16m 47s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3d0511322434933a5c15f3853c07f8a : SUCCESS in 1h 04m 05s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/9df04fa27f33463ca43981ea663048d5 : SUCCESS in 55m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5dca1c5ef153422897b61cf2c35d088d : SUCCESS in 50m 29s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/aaaa1d65be214853b0e6a6b85dbfe009 : SUCCESS in 1h 15m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/60fedb66b34340e6a01f3381a7de9101 : SUCCESS in 1h 07m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4aaa7d7cc34447619c41d53e4c109098 : SUCCESS in 51m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e0ea494e36bd4ff3bef3cd8739bcb0b2 : SUCCESS in 51m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b711329b59d647bda8fa03004ce72a29 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15d5630f67b141e5aa8ae2e855d85da0 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/406bde2d35db4304b47312135ff16dcb : SUCCESS in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0f28fc01fc04421b71d733763979b61 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/049d7d5941ee40108692bda6c87036a1 : SUCCESS in 10m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3db942ee728647c6b136c413f8e72bb7 : SUCCESS in 2m 47s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2d0283f2114847a58e0a5e52f093bab5 : SUCCESS in 7m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bdd436ecd9bb3be6c1282062619800e333692d35","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-09-20 22:35:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f97707163c4b35c92b334f871a473a9c70a1b0da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 22:35:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7865da526fe868fa101097e3aabea1e89da57d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 23:50:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a79c3facd0f4ec7a2daa2f1bbb98bc2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ad63fb56205f4e58b226797b2dbddeca : SUCCESS in 48m 33s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/cfad2e140fba4ce9bd92ed8b5b19ac95 : SUCCESS in 50m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d9d39d2db0874d46be80f2d8688d9bb3 : SUCCESS in 46m 57s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ee3df2238a244850bccb8e54c5150484 : SUCCESS in 1h 09m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bb1cac68d56d48ab9d5eb025c6df7d52 : SUCCESS in 44m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/938e30ecb82240c7b6156497af887975 : SUCCESS in 54m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/727ffc2741344ea589d782a42ae3bb6c : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8fa4c09ad4a49fa9114f0f3189d00fa : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a95a7de1f3c4faea3dc32f828c23cc6 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b1d068c3ec54c2bb1cc8ad696b80f57 : SUCCESS in 6m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/eac3b047498b4d87b9b71f01b15e425b : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":1},{"id":"ec0dd2c37240201aca216685bcab262d59f29999","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 23:50:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cee934746b04ddecc404712e4568a784d60ea5ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 23:51:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ca6506c152b4c8389778abf1fd86024\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/691ee48cb2874cbba522fa2a3d98b931 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b3e3bbf9e2c53863932b65a119700e2e4394b533","revisions":{"b3e3bbf9e2c53863932b65a119700e2e4394b533":{"kind":"REWORK","_number":1,"created":"2022-09-20 17:54:13.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/61/858561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/61/858561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/858561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/858561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/858561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/61/858561/1"}}},"commit":{"parents":[{"commit":"b0bff602edd94c8cc37c9f6db45265525498d0ae","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b0bff602edd94c8cc37c9f6db45265525498d0ae"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-09-20 17:53:08.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-09-20 17:54:12.000000000","tz":0},"subject":"Fix minimum versions in requirements.txt","message":"Fix minimum versions in requirements.txt\n\nIt appears we need to bump some minimum versions for changes made in Zed.\nThis patch sets:\nalembic\u003e\u003d1.8.0\n\nChange-Id: I3e3e1a4701c52003e4adf2738b1ca8598d4cae47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b3e3bbf9e2c53863932b65a119700e2e4394b533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b3e3bbf9e2c53863932b65a119700e2e4394b533"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
