)]}'
{"id":"openstack%2Fdesignate~858562","triplet_id":"openstack%2Fdesignate~master~I300a8dd04858476bd348842c2ae9f97d62338682","project":"openstack/designate","branch":"master","topic":"test","hashtags":[],"change_id":"I300a8dd04858476bd348842c2ae9f97d62338682","subject":"DNM: CI test","status":"ABANDONED","created":"2022-09-20 18:01:40.000000000","updated":"2022-10-10 11:00:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7ba3c08c40e286c09e33497a9fea3000a9897aa9","_number":858562,"virtual_id_number":858562,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-09-20 19:19:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-20 19:19:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6287d55eae95eaef9160b9a742b63fd573500674","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-09-20 18:01:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2e308b20c57ab8536eab98916c1f97cc78991d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 19:19:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a9531d944ea4baabc533fa15d22505f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f7c2258a35954a659a2acc4669b1eee3 : SUCCESS in 2m 31s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1453cb43f928413e9126911cf6fdfc97 : SUCCESS in 39m 31s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c3ed4d02d8fd4c069fa5008361d76556 : SUCCESS in 53m 47s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd5b5b233033418699dc8d009e9d7402 : SUCCESS in 1h 00m 29s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/7948701933084605bc7341d2e1ff5f58 : SUCCESS in 35m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/47b874dfb2f64b72987f8371a6be4323 : SUCCESS in 50m 39s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3fffa52d91b446088a62adaf4bac0ff1 : SUCCESS in 59m 49s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/db34a41ea6bb446eaa66bdf153df32cf : SUCCESS in 1h 10m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6a38dd3c9193414d9979db945286f772 : SUCCESS in 41m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ba9ddc1a625542ca836b05ebd1f7cf59 : SUCCESS in 56m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/250c650c89c24ca4a45c26bd618711cd : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa819d68735b43aca38ac07703c81be1 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a4c251a2c984e90b93839e1df3bf0f1 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b358c36ffb9447ab499f4a8882e3235 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6cd5d0920954d76b87e673371f0292f : SUCCESS in 7m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/de3aca4f31914018ba98fa9112fccb01 : SUCCESS in 36m 42s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/074183c17bdc4f64ba96fdb253170396 : SUCCESS in 4m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ba3c08c40e286c09e33497a9fea3000a9897aa9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-10 11:00:25.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"58834cb37ac38c602d20d10b8148d79916d6ab85","revisions":{"58834cb37ac38c602d20d10b8148d79916d6ab85":{"kind":"REWORK","_number":1,"created":"2022-09-20 18:01:40.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/62/858562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/62/858562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/858562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/858562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/858562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/62/858562/1"}}},"commit":{"parents":[{"commit":"7460ae30bf92b5bdf68137bc5de9f7891a33b89b","subject":"Fix compatibility with oslo.db 12.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7460ae30bf92b5bdf68137bc5de9f7891a33b89b"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-09-20 18:01:26.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-09-20 18:01:26.000000000","tz":120},"subject":"DNM: CI test","message":"DNM: CI test\n\nChange-Id: I300a8dd04858476bd348842c2ae9f97d62338682\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58834cb37ac38c602d20d10b8148d79916d6ab85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58834cb37ac38c602d20d10b8148d79916d6ab85"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
