)]}'
{"id":"openstack%2Fdesignate~858588","triplet_id":"openstack%2Fdesignate~master~I7edf1e147637b013f35ba47e4d6b92d53d42fe84","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-09-21 06:15:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7edf1e147637b013f35ba47e4d6b92d53d42fe84","subject":"Fix mdns test_handler setup","status":"MERGED","created":"2022-09-20 23:03:20.000000000","updated":"2022-09-21 06:16:27.000000000","submitted":"2022-09-21 06:15:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858588","meta_rev_id":"9d6e521e2ad6f9591c638b51ad738ad3549f7c40","_number":858588,"virtual_id_number":858588,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-21 06:15:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-09-21 05:09:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-21 04:17:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-09-21 05:09:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-21 00:36:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-21 04:17:16.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-09-21 05:09:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"a1413f5068cb6782285f90ffdac36aa4477f5eef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-09-20 23:03:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e561315d97a5ce43552b0c1f4eaff243eb6da445","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 00:36:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/30808bf9d3154cfeb89c35fd8b23665d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/56b79ed1008142838f56880c1b0e5838 : SUCCESS in 56m 24s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/187d1d463db1447db955c8de927c1bbb : SUCCESS in 1h 19m 35s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e61aa2a0324f417f8d789bb3a6ce32a1 : SUCCESS in 58m 58s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1e81307f77d74d248493f0dc9129cea2 : SUCCESS in 54m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/36af95102f214d198a184dd240dcb1da : SUCCESS in 49m 11s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/17d01438484648d2953a6506898dedc8 : SUCCESS in 1h 11m 39s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/23b041bce70541e49dda3c4764d3c385 : SUCCESS in 1h 03m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/abf11fecf61c4141be493170126b845a : SUCCESS in 50m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/acfa131920bf41a9a4f1ddd4c1c83790 : SUCCESS in 43m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a46b86fd7174a8da4f51e2e106294f0 : FAILURE in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab5b5295d268474d949208facaf3e46b : FAILURE in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a755553141fb4fe6a3a1c46feef40494 : FAILURE in 7m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff0c12c99a8c4529a90b6da3e0202ed6 : FAILURE in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e4b3138e82c4bfeae2572928524926b : SUCCESS in 12m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/327b697412264b0c81d65dab7efc7d2b : SUCCESS in 33m 27s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/54f6677fc6c0457295abe08ecaef2af2 : FAILURE in 7m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2e97e095bdb5f698f972a996c8a0d09e1a460c52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-09-21 00:55:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fc8b2441ce0e1a7e0848ca3872fbd8519627dd5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 02:22:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef1d819f985c4ba69b603d8e0e851dc0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7b07de149a4c4271bff0b992dfb7a950 : SUCCESS in 47m 51s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a5fc3722365549ec9502a6505e7afc37 : SUCCESS in 1h 13m 29s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea2ef24834a94908b1aedb09db17d6d4 : SUCCESS in 1h 03m 44s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/69d7cbd03bc449548ec1ea53c1c69b49 : SUCCESS in 49m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f9f76cc8a1d243e08da392e560aaedfe : SUCCESS in 49m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/1a49bd98da29450dabf93b0f8bf881c0 : SUCCESS in 1h 14m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e2576a6188034bb6995315f3ebab7626 : SUCCESS in 1h 07m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3aa4f63912824efc8d56f3818478dad9 : SUCCESS in 40m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/51c0b0a8fe604a5e93b5c97e4ccc56c8 : SUCCESS in 53m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69503696326347fb844e2670cd6295cb : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f5a572e6a7546d98edc6a5710ee122b : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b1b6d886cc241c1baa882fe6521d74f : SUCCESS in 9m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7c8a946f9634f33bbac3bc054562f44 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af2e3455ced04870bebeded4f51bc56a : SUCCESS in 7m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2e257106c6e34bea9c4e4897aff7cb2f : SUCCESS in 43m 44s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2f56c016af06495096f790151523fdcf : SUCCESS in 4m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"09297b4acfca374c9f2e863fbec3f374cc75eaf8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-09-21 04:17:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e2dddf8396232b754498556e0e3c82f62b6d02b6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-09-21 05:09:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"787f6f9ecc95f179e7c9a58f705880ae5e273ee5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 05:10:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d054aec395cc4594e949fb712cdf61afd9f5a78e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 06:15:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca145d305abe4541a2f6075b3a507b4d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/306e7105b3b441059de7a78e34289ba0 : SUCCESS in 46m 18s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/640048f681de45d3a08bafd7336b1c6e : SUCCESS in 45m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2759545722ba4699954b44e00a19d24b : SUCCESS in 50m 13s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7dc283d1592d492fb406a701c355e529 : SUCCESS in 1h 00m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fab7fa4f85cc4ff896e181ce3f4c6510 : SUCCESS in 47m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d706f29fcdf64d629211213df53273bc : SUCCESS in 49m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0730b63e230c4ec0bd313ed4ac6b91f3 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aaf1e7ff01b74f2488761685d118fc6f : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64212a328b864a47b91b87515dd934b3 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/757f63823bfc4382ae54a84c05e050b8 : SUCCESS in 8m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9b40203063554f71badc9b29fb71f3ef : SUCCESS in 37m 41s","accounts_in_message":[],"_revision_number":2},{"id":"70626282143b170eab34273419536c4d4d9f34ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 06:15:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9d6e521e2ad6f9591c638b51ad738ad3549f7c40","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 06:16:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7f6b97091f54e41a8f0ebdf6daacaf1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2447706ddb494813a2ffb600188edfed : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e6566f86067cbbdfc971dc0f60d2d3db62853965","revisions":{"1b629ac0e1cc4f9cb6c647d936e7bca5948129a2":{"kind":"REWORK","_number":1,"created":"2022-09-20 23:03:20.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/88/858588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/88/858588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/858588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/858588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/858588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/88/858588/1"}}},"commit":{"parents":[{"commit":"b0bff602edd94c8cc37c9f6db45265525498d0ae","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b0bff602edd94c8cc37c9f6db45265525498d0ae"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-09-20 23:00:40.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-09-20 23:03:18.000000000","tz":0},"subject":"Fix mdns test_handler setup","message":"Fix mdns test_handler setup\n\nIf the unit tests were run isolated from the larger test suite, the mdns test_handler unit tests would fail with \"AssertionError: \u0027TRANSPORT\u0027 must not be None\" because the messaging and RPC fixtures were not setup for the test suite.\nThis patch adds the required test fixtures to the mdns test_handler setUp().\n\nChange-Id: I7edf1e147637b013f35ba47e4d6b92d53d42fe84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1b629ac0e1cc4f9cb6c647d936e7bca5948129a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1b629ac0e1cc4f9cb6c647d936e7bca5948129a2"}]},"branch":"refs/heads/master"},"e6566f86067cbbdfc971dc0f60d2d3db62853965":{"kind":"REWORK","_number":2,"created":"2022-09-21 00:55:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/88/858588/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/88/858588/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/858588/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/858588/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/858588/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/88/858588/2"}}},"commit":{"parents":[{"commit":"b0bff602edd94c8cc37c9f6db45265525498d0ae","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b0bff602edd94c8cc37c9f6db45265525498d0ae"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-09-20 23:00:40.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-09-21 00:55:35.000000000","tz":0},"subject":"Fix mdns test_handler setup","message":"Fix mdns test_handler setup\n\nIf the unit tests were run isolated from the larger test suite, the mdns test_handler unit tests would fail with \"AssertionError: \u0027TRANSPORT\u0027 must not be None\" because the messaging and RPC fixtures were not setup for the test suite.\nThis patch adds the required test fixtures to the mdns test_handler setUp().\n\nChange-Id: I7edf1e147637b013f35ba47e4d6b92d53d42fe84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e6566f86067cbbdfc971dc0f60d2d3db62853965"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e6566f86067cbbdfc971dc0f60d2d3db62853965"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
