)]}'
{"id":"openstack%2Fdesignate~860657","triplet_id":"openstack%2Fdesignate~stable%2Fussuri~I94324104a7ed89e757d48bb250d8b1132026ce46","project":"openstack/designate","branch":"stable/ussuri","topic":"fix-queue-definition-stable/ussuri","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-02-23 01:55:53.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I94324104a7ed89e757d48bb250d8b1132026ce46","subject":"Update zuul queue configuration","status":"ABANDONED","created":"2022-10-07 16:44:44.000000000","updated":"2023-02-23 01:55:53.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"cherry_pick_of_change":858381,"cherry_pick_of_patch_set":1,"meta_rev_id":"708d24502b7a7c678b9886d37fa1f9b30b7f348b","_number":860657,"virtual_id_number":860657,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":-2,"date":"2023-02-22 17:45:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2022-10-07 19:34:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2022-10-07 17:05:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2023-02-22 17:44:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-07 16:45:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-07 19:34:48.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2022-10-08 09:55:30.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"5007bd589f086901ef66c29e4c737debded7011c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-10-07 16:44:44.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.\n\nThe following files contain Git conflicts:\n* .zuul.yaml","accounts_in_message":[],"_revision_number":1},{"id":"fc42f3864168052708183b6b4f0ac091b5bcbb88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 16:45:06.000000000","message":"Patch Set 1: Verified-1\n\nUnknown configuration error","accounts_in_message":[],"_revision_number":1},{"id":"fac960c24e516fd0c11c6b23f16e966607e1c554","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-10-07 16:47:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4295209e5da4b4ea3f3faafc78e4ce24822c9118","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-10-07 16:47:57.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"e1179e2f5587fde591c0a6f997eb3ac40392125f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-10-07 17:05:35.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dc62b3bfed416cacfe19e10d05e0eedc158edd48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 17:57:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61eb0c44dc824ae6acdea62c436e41b9\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c9ae9233edb842e59101a8f8546c91f3 : SUCCESS in 36m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bdff6f5367cf464dac983b76fc3a74af : SUCCESS in 35m 52s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/609184e7f64e4f88a783b2941ae4392e : SUCCESS in 1h 08m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5187bfc071a44b6e87814e9a30a03d61 : SUCCESS in 35m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3c95cfe8fc5941029e2c4ffee75ba07c : SUCCESS in 31m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cefcb7bceeb24810bb5e2475b5b7c351 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ffa1bb8393144268a8ae366398cc8da : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58ac01638bcb4fc2aecf95ee7dbe53f5 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/44b7af0e97ef44238092af56b441aeca : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bf86bcbc41d414b921e568a18aaffd4 : SUCCESS in 7m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/199c971cd44245e79148c5cee230de58 : SUCCESS in 2m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8631a4f1327641618639d825b7c89030 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":2},{"id":"5b386ec952e5c5ea0bc9dc91d16cce2ae2bfe3a4","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-10-07 19:34:48.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9d950efba27c7ab6a822cc7e95396876cfd8cb7c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-08 09:55:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1e0c3358958478fb88d366c5333f33d868ccca63","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-22 17:44:47.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6704d89770fb552517fbca622964ff0faf5a6dc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 17:45:12.000000000","message":"Patch Set 2: Verified-2\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/designate on branch stable/ussuri.  The error was:\n\n  The key \"queue\" appears more than once; duplicate keys are not\n  permitted.\n\nThe error appears in the following stanza:\n\n  name: designate-devstack-jobs\n      queue: designate\n      check:\n        jobs:\n          - designate-bind9\n          - designate-pdns4\n          - designate-grenade-pdns4\n          - designate-ipv6-only-pdns4\n          - designate-ipv6-only-bind9\n      queue: designate\n      gate:\n        jobs:\n          - designate-bind9\n          - designate-pdns4\n          - designate-grenade-pdns4\n          - designate-ipv6-only-pdns4\n          - designate-ipv6-only-bind9\n\n  in \"openstack/designate/.zuul.yaml@stable/ussuri\", line 101, column 5","accounts_in_message":[],"_revision_number":2},{"id":"708d24502b7a7c678b9886d37fa1f9b30b7f348b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-23 01:55:53.000000000","message":"Abandoned\n\nfixed in bdbbe6a396682f316304027845c818f94e6f4cd3","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"da492a9bcffc1c6825ff88645445c8fb681a0a90","revisions":{"8dfd38daff25239d914443d40beb5a31bbd8036c":{"kind":"REWORK","_number":1,"created":"2022-10-07 16:44:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/860657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/57/860657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/860657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/860657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/860657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/57/860657/1"}}},"commit":{"parents":[{"commit":"5e569a3c7646042131926b090f570ca5082503e3","subject":"Fix zone update when adding new Bind9 target to pool.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5e569a3c7646042131926b090f570ca5082503e3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-19 16:10:12.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-10-07 16:44:44.000000000","tz":0},"subject":"Update zuul queue configuration","message":"Update zuul queue configuration\n\nZuul is changing how queues are declared[1]. This patch updates Designate for the new way to define the queue.\n\n[1] https://lists.zuul-ci.org/pipermail/zuul-discuss/2022-May/001801.html\n\nChange-Id: I94324104a7ed89e757d48bb250d8b1132026ce46\n(cherry picked from commit fe9fb45c21a7be2f88600e1f29fd37f56244cc1a)\n(cherry picked from commit 099795d2e1497c4043158ffe92fbaa9fbe76ddac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8dfd38daff25239d914443d40beb5a31bbd8036c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8dfd38daff25239d914443d40beb5a31bbd8036c"}]},"branch":"refs/heads/stable/ussuri"},"da492a9bcffc1c6825ff88645445c8fb681a0a90":{"kind":"REWORK","_number":2,"created":"2022-10-07 16:47:36.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/860657/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/57/860657/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/860657/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/860657/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/860657/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/57/860657/2"}}},"commit":{"parents":[{"commit":"5e569a3c7646042131926b090f570ca5082503e3","subject":"Fix zone update when adding new Bind9 target to pool.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5e569a3c7646042131926b090f570ca5082503e3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-19 16:10:12.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-10-07 16:47:35.000000000","tz":0},"subject":"Update zuul queue configuration","message":"Update zuul queue configuration\n\nZuul is changing how queues are declared[1]. This patch updates Designate for the new way to define the queue.\n\n[1] https://lists.zuul-ci.org/pipermail/zuul-discuss/2022-May/001801.html\n\nChange-Id: I94324104a7ed89e757d48bb250d8b1132026ce46\n(cherry picked from commit fe9fb45c21a7be2f88600e1f29fd37f56244cc1a)\n(cherry picked from commit 099795d2e1497c4043158ffe92fbaa9fbe76ddac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/da492a9bcffc1c6825ff88645445c8fb681a0a90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/da492a9bcffc1c6825ff88645445c8fb681a0a90"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
