)]}'
{"id":"openstack%2Fdesignate~860772","triplet_id":"openstack%2Fdesignate~master~Idb04e81ce5954ca8b5e387dc7c0776fdf7d08779","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-11-19 00:27:08.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Idb04e81ce5954ca8b5e387dc7c0776fdf7d08779","subject":"Add basic api version test coverage","status":"MERGED","created":"2022-10-08 20:40:37.000000000","updated":"2022-11-19 02:26:00.000000000","submitted":"2022-11-19 02:24:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860772","meta_rev_id":"437ea5238c308e15bfa1e7158796c8a8a4b6d31f","_number":860772,"virtual_id_number":860772,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-19 02:24:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-10-27 15:09:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2022-11-18 22:14:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-19 00:27:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-08 21:40:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-10 19:11:53.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2022-11-18 22:14:43.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"3ff57b2de42d52e04dcbe7b2cd08440071dd1d99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-08 20:40:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfc86663b26b27ef503a4e755d767b0b29ade1c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-08 20:41:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4ece7575b76587ee46b23b851c1f96cf7af3037a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-08 20:49:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ecada2788f5b27e89ca62bfd1ebbf3f0373a7937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-08 21:40:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e077f1159ff9403f88bae2d281dc86a0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/621c8cf3a3ad4954aef5c34b153b975e : SUCCESS in 37m 46s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0876e76cd4ae435795ad532760d11727 : SUCCESS in 43m 57s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/acace98d697c4eeca4471c3dc04aa8cc : SUCCESS in 40m 32s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/7c167935342e416dae26d23260197e59 : SUCCESS in 37m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6820bc42690948f98b4ec17edcf05724 : SUCCESS in 33m 20s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/27a529481ee44e91ba01476ca78b89a9 : SUCCESS in 49m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/147714581afb44c992206d77402a1f21 : SUCCESS in 48m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/af74d8c546814b94b542cb667c42a313 : SUCCESS in 33m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c651ce82f13e476880e6895a853b9431 : SUCCESS in 31m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/141bbb4569d44370bdb9f080e92fc9c4 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/003237f96371488eaa46822dd23e9789 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c65b9af7bff444beb0840622ef71dfec : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84e695d05b084c839201857df2231f49 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c289ed4e20649a685ad1b413a87ecd0 : SUCCESS in 9m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/036fec28409e40ee968cad574ccbfc35 : SUCCESS in 28m 44s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9bb79493f0434ec4b82c2c316281e508 : SUCCESS in 3m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"197ded49e2e801025f428fb9010ebaf908321113","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-10-10 19:11:53.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d2b18ac996f9e25dea8439e8e513b67d3a94bfcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-27 15:09:59.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c786759cf5a9a23b7096ac817ebee614e357247c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 16:14:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc5f14d23c104a9390f0291e0a55c08a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/120df14fe1504c119e762aace1e33c2e : SUCCESS in 33m 56s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2d1e4c4824ed484dbb79cad001ea206f : SUCCESS in 1h 01m 38s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1a3d84b421f4c85b1babff691f4a146 : SUCCESS in 41m 08s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1105f94339af4d179b04a31d273dc354 : SUCCESS in 41m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/01376163ab004a7396e6d18ff157daef : FAILURE in 36m 31s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/881a6f24a0f046f19f78344e5cb2c244 : SUCCESS in 56m 41s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2b88541acbcb4e68b0291d2542b7b0a1 : SUCCESS in 47m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/320097f97c7f40ffaccbfa48e5fa3068 : SUCCESS in 34m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/395ef3909c0d4939bd45d9e5ee0f0a48 : SUCCESS in 39m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fb23d10a2774ea9b4a1cc801550015b : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/594b1d3a60cd46469b3f42325a686c72 : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0940770437d4e09918b13edb2ffe1db : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/52da396274b2444e8b17e6aa34d5d67b : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1b35050cc2c4222955e2476e4ed2987 : SUCCESS in 8m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3c02dea0e39c460f8469899b4a625370 : SUCCESS in 35m 04s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1f766b3d3c344424b9753ca2c148e266 : SUCCESS in 3m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"018fdf9d432b6c466f11985078e04a4822aaff12","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-27 16:21:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c2e133128be1bf950cc5a06fab4b62fa97ca8afc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 17:33:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1e7cad0a49246fbba11f3969c96f71e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fdb2213c0d2b418194060c8dec1b2988 : SUCCESS in 36m 45s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/61f2de970262472da382a59e4215f46e : SUCCESS in 59m 13s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea261e977b254767acde7291307ac30a : SUCCESS in 37m 53s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/057f34d944f443b7bc51ab7519048dee : SUCCESS in 39m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2161a221e76341b4b2995bbca897c226 : SUCCESS in 37m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/69af2a6bb7224002842067f3fc74795e : SUCCESS in 55m 18s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0caedbe30c2d40f4affce94a3c0cf7e0 : SUCCESS in 50m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7729df4636da4a3d8e23dc0dc8a4b4da : SUCCESS in 41m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b4d71d3562e141caa1d317258324b961 : SUCCESS in 35m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/645486dfba9744e1b940316d7f2417c4 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d60b305796f54a6099ef03bb3fbf354f : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6c6b567eea2461393b8a39a3fa5bd0b : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9913ebacb18f4a13a2b528173c7569ad : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86f4aff0aada413fa33fda3fb297624e : SUCCESS in 7m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/14eb43d8c7184613b627bf053bc50fb8 : SUCCESS in 31m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/86c7b0123e9b4c6d88ec7462979fb872 : SUCCESS in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d8a3d151afb900af8482ca071dc3baea7eaba09d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-18 22:14:43.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4c6827951416af57c04a0c39dd359d1b728b5297","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-11-19 00:27:08.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d26af23a64e02de41d9287c10c350f1d644bfe8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 00:27:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a4575bd96dea7c77a05e7c5c75685e321381c88b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 02:24:04.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29bfbe853c4f459eb56c72c6689aaccc\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/24d77c4959354773bcc10eac77a1d762 : SUCCESS in 36m 00s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/5d3d83e3872442d5b0bad0d05a8d5246 : SUCCESS in 36m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f037e39e64a145d6a325970d67af02ca : SUCCESS in 38m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6a6784eafd134744be061bbb209b82a8 : SUCCESS in 55m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1c6bba4f65ce497fbd709ee79f4c8ef3 : SUCCESS in 36m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f0b374cb8f364854a1e69f11a0ada7aa : SUCCESS in 38m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e4500d56c8642e68f8c19c82ffa46e8 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/281a7a33e2654fc9b9ec7efae420ecdc : SUCCESS in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21e4afadf31845fdb427adcc9c681ab6 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e1c6f3bb80b4f53b36fc07d10c2eb5d : SUCCESS in 7m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d5cb4927129b428f9de06b857c92b701 : SUCCESS in 31m 16s","accounts_in_message":[],"_revision_number":4},{"id":"f5bc4bd616cb0a4c28b415eb37674f82c1ec5b57","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 02:24:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"437ea5238c308e15bfa1e7158796c8a8a4b6d31f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 02:26:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76779a8b8cc246fc952a32fd86a7d355\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c33476a7a7144e89bca2758d54ef019f : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fb876d0ddc12fa89db5e3bec519ad4e93e914b8d","revisions":{"80dd82ccd3dbe1f778ac72da9b5afe879ffb8189":{"kind":"REWORK","_number":1,"created":"2022-10-08 20:40:37.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/72/860772/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/72/860772/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/72/860772/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/72/860772/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/72/860772/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/72/860772/1"}}},"commit":{"parents":[{"commit":"6ca038f350f36ee13ea45f50ddf3624e43e32e6a","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ca038f350f36ee13ea45f50ddf3624e43e32e6a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-08 20:40:23.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-08 20:40:23.000000000","tz":-420},"subject":"Add basic api version test coverage","message":"Add basic api version test coverage\n\nChange-Id: Idb04e81ce5954ca8b5e387dc7c0776fdf7d08779\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/80dd82ccd3dbe1f778ac72da9b5afe879ffb8189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/80dd82ccd3dbe1f778ac72da9b5afe879ffb8189"}]},"branch":"refs/heads/master"},"317cef1e0e42b0c177e3ce38c3f8683854b6416b":{"kind":"REWORK","_number":2,"created":"2022-10-08 20:41:56.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/72/860772/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/72/860772/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/72/860772/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/72/860772/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/72/860772/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/72/860772/2"}}},"commit":{"parents":[{"commit":"6ca038f350f36ee13ea45f50ddf3624e43e32e6a","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ca038f350f36ee13ea45f50ddf3624e43e32e6a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-08 20:40:23.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-08 20:41:49.000000000","tz":-420},"subject":"Add basic api version test coverage","message":"Add basic api version test coverage\n\nChange-Id: Idb04e81ce5954ca8b5e387dc7c0776fdf7d08779\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/317cef1e0e42b0c177e3ce38c3f8683854b6416b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/317cef1e0e42b0c177e3ce38c3f8683854b6416b"}]},"branch":"refs/heads/master"},"6f609a2276f3755e6931a9db17a04f586bd1c0d3":{"kind":"REWORK","_number":3,"created":"2022-10-08 20:49:30.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/72/860772/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/72/860772/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/72/860772/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/72/860772/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/72/860772/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/72/860772/3"}}},"commit":{"parents":[{"commit":"6ca038f350f36ee13ea45f50ddf3624e43e32e6a","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ca038f350f36ee13ea45f50ddf3624e43e32e6a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-08 20:40:23.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-08 20:49:25.000000000","tz":-420},"subject":"Add basic api version test coverage","message":"Add basic api version test coverage\n\nChange-Id: Idb04e81ce5954ca8b5e387dc7c0776fdf7d08779\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f609a2276f3755e6931a9db17a04f586bd1c0d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f609a2276f3755e6931a9db17a04f586bd1c0d3"}]},"branch":"refs/heads/master"},"fb876d0ddc12fa89db5e3bec519ad4e93e914b8d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-10-27 15:09:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/72/860772/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/72/860772/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/72/860772/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/72/860772/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/72/860772/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/72/860772/4"}}},"commit":{"parents":[{"commit":"0b162a4c48ff7806ee0dda0f871a884a32374ed9","subject":"Add Ubuntu 22.04 CI support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0b162a4c48ff7806ee0dda0f871a884a32374ed9"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-08 20:40:23.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-27 15:09:59.000000000","tz":0},"subject":"Add basic api version test coverage","message":"Add basic api version test coverage\n\nChange-Id: Idb04e81ce5954ca8b5e387dc7c0776fdf7d08779\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fb876d0ddc12fa89db5e3bec519ad4e93e914b8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fb876d0ddc12fa89db5e3bec519ad4e93e914b8d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
