)]}'
{"id":"openstack%2Fdesignate~860773","triplet_id":"openstack%2Fdesignate~master~Iea61a0af65afc4bf91c581869984115a6d44011f","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-10-14 23:38:22.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Iea61a0af65afc4bf91c581869984115a6d44011f","subject":"Removed unused SQL functions and better coverage","status":"MERGED","created":"2022-10-08 21:14:48.000000000","updated":"2022-10-15 00:49:23.000000000","submitted":"2022-10-15 00:47:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860773","meta_rev_id":"16fcd233b9c0560164b519f50d5223572262d9bd","_number":860773,"virtual_id_number":860773,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-15 00:47:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-10-10 17:40:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2022-10-14 23:07:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-14 23:38:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-08 23:03:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-10 17:40:55.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2022-10-14 23:07:47.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"e9eccb7e6968c9861b9fe1017029e7eaf57ec4d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-08 21:14:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9baa7ce71ea7d824ce349659d32148d602f851bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-08 21:17:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"44bb4cc585eac86b004aa975eb62dc4aab8ec5c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-08 21:43:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"15bce76dc0af38af35c85d2ef169d283b17bc630","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-08 21:50:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"caf2eb50f983a5900158e8bf721fc2b92b267c2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-08 21:52:09.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"78f87feef9f2d0401c81564614571aaf9b42cce9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-08 23:03:03.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b55fc57149d4e35a67d46434bf1ccb2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8a982081811b43e2bc4204c4fc03e6e1 : SUCCESS in 48m 45s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/101216d2e3824eb7bcae027704db0578 : SUCCESS in 1h 06m 11s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac7d30d5c28e4106a97c04317aa6d8d4 : SUCCESS in 55m 02s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e4028a8ee5d747148642d087acadbb5f : FAILURE in 46m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0496a449b9324aa7922c18d778c58bbf : SUCCESS in 46m 56s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/06196ec857024d94a2b7abc6349aa37c : SUCCESS in 1h 08m 34s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cceea7181b8e4b74a70112e4609b2e63 : SUCCESS in 1h 04m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2ab49ad73d2345dbae2ae829aa127f24 : FAILURE in 50m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2a20369876154787824d756aba9153b0 : SUCCESS in 45m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a47e0cbad974800bebac95785e3d587 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48428eb0c5c3444bbfe048fe3fd52993 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3877b94c5ad747799115cbedc6dafc8a : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2255c0daf3384f72a6dfb547d5978b39 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/958169b01eda41a2b199c3e2f67f1d2b : SUCCESS in 7m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/be875b170a6a40359dfa3f471a510334 : SUCCESS in 25m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/33e19f155cb0405fa19379841e0eb778 : SUCCESS in 4m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"75d85effe1d11a04bdb2c3a2413682685c05e85c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-08 23:07:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e83c76288121a8bb9088067245be11d5b19e8a74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 00:08:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13635df65c8d467eb9e008fbea07d2c0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cad4af05cd5d4fde89f3e6bbfae1b7ff : SUCCESS in 43m 55s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b7dd817ff41f4fe595a8e112374cc702 : SUCCESS in 56m 27s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d56f36288474d49b9599b9b3ec2047a : SUCCESS in 48m 35s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/13350f3a36aa45d489ecaa666223f3b2 : SUCCESS in 38m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/43368bf7dd53488887d6724aeaf5334c : SUCCESS in 39m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/06d6c2fce2cc4cb4bcfe7375565365af : SUCCESS in 57m 12s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/763e9610259e468ca99432c401c832de : SUCCESS in 1h 00m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a03cc2d658b54440aa375c22a3c75188 : SUCCESS in 37m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6d084fb8e75646be8326c3df2c1f593d : SUCCESS in 37m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a939097c9ff46d9ad0a5ca6b8b98cb0 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef0411432fa3424db9db8b7c966bda58 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f428b16cd6542979e36a3dcd47528c4 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77bd2afc22a14adaa11a7443720cd2b1 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/262c3de2cc6c4f8bbf0132d2a489be9f : SUCCESS in 8m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9e30cfbcd81143dd93d1d9ba9d9f225f : SUCCESS in 26m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/eb915f7abaeb4dbd98bae8209b87486e : SUCCESS in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6e1f2db1478e3d3f8fd39fc1ae4b169c4d7322d4","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-10-10 17:40:55.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7ba4b2a05e529d1eaa999697687715edf9f85a82","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-10-14 23:07:47.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7248a264dafa807df178fce7edb2ddf59f33f913","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-14 23:38:22.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"0bdc9dd33181660c78f26191ef089861d29ad606","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 23:38:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"95548919050384967ac5c0eeea378d6a00716a26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-15 00:47:50.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86b05a162f2042cc845f6457575ab2c5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/80f5b04e6fdc47308e390f01692396cd : SUCCESS in 31m 51s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/175daa361f5f46f08e70f35c5d61da9c : SUCCESS in 35m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a86b14fb851547ac9e1e80783e773162 : SUCCESS in 48m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/40df3380131143cbb60cf0fcfea6f08d : SUCCESS in 1h 02m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cfabf6042f344387949da0421612a9d8 : SUCCESS in 39m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4f78fa4cee3d4136a0a51296afa77a56 : SUCCESS in 52m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2edb7f3d5564488909aab47aed8ecf4 : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/644f25a1a36646f5aa88d8f9cd1d3462 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea98a389fe4c4a7b8016b57f9188bcd8 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf0966a35f544bf597a93b082b99ff6e : SUCCESS in 6m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d13b580ec49849598553dfcd8883c56e : SUCCESS in 31m 13s","accounts_in_message":[],"_revision_number":5},{"id":"aaba5263a4c22e88975a11bfd1ef4157cc1369b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-15 00:47:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"16fcd233b9c0560164b519f50d5223572262d9bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-15 00:49:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36d608334cb040d5b6559317c687cbc1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd51ba92d2354388a181931ee39122d0 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ab3b57b044eb5a662f6299e8073dc32311d14c14","revisions":{"c06ce72480c9754cfed75bf093def03535e2c047":{"kind":"REWORK","_number":1,"created":"2022-10-08 21:14:48.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/73/860773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/860773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/860773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/860773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/860773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/860773/1"}}},"commit":{"parents":[{"commit":"6ca038f350f36ee13ea45f50ddf3624e43e32e6a","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ca038f350f36ee13ea45f50ddf3624e43e32e6a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-08 21:13:54.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-08 21:14:40.000000000","tz":-420},"subject":"Removed unused SQL functions and better coverage","message":"Removed unused SQL functions and better coverage\n\nChange-Id: Iea61a0af65afc4bf91c581869984115a6d44011f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c06ce72480c9754cfed75bf093def03535e2c047"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c06ce72480c9754cfed75bf093def03535e2c047"}]},"branch":"refs/heads/master"},"70ef7ec759e97d0b00fd194a8e7ce6cb1ad08e6d":{"kind":"REWORK","_number":2,"created":"2022-10-08 21:17:44.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/73/860773/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/860773/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/860773/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/860773/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/860773/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/860773/2"}}},"commit":{"parents":[{"commit":"6ca038f350f36ee13ea45f50ddf3624e43e32e6a","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ca038f350f36ee13ea45f50ddf3624e43e32e6a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-08 21:13:54.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-08 21:17:38.000000000","tz":-420},"subject":"Removed unused SQL functions and better coverage","message":"Removed unused SQL functions and better coverage\n\nChange-Id: Iea61a0af65afc4bf91c581869984115a6d44011f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/70ef7ec759e97d0b00fd194a8e7ce6cb1ad08e6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/70ef7ec759e97d0b00fd194a8e7ce6cb1ad08e6d"}]},"branch":"refs/heads/master"},"e979fca5fd9fb5577771ff00f105c367fbbbd6be":{"kind":"REWORK","_number":3,"created":"2022-10-08 21:43:29.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/73/860773/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/860773/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/860773/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/860773/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/860773/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/860773/3"}}},"commit":{"parents":[{"commit":"6ca038f350f36ee13ea45f50ddf3624e43e32e6a","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ca038f350f36ee13ea45f50ddf3624e43e32e6a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-08 21:13:54.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-08 21:42:46.000000000","tz":-420},"subject":"Removed unused SQL functions and better coverage","message":"Removed unused SQL functions and better coverage\n\n- Cleaned up indentention.\n- Fixed minor inconsistencies.\n\nChange-Id: Iea61a0af65afc4bf91c581869984115a6d44011f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e979fca5fd9fb5577771ff00f105c367fbbbd6be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e979fca5fd9fb5577771ff00f105c367fbbbd6be"}]},"branch":"refs/heads/master"},"617c561cf2e6698e7e9e56a0b89d4bd9dc151c66":{"kind":"REWORK","_number":4,"created":"2022-10-08 21:50:10.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/73/860773/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/860773/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/860773/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/860773/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/860773/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/860773/4"}}},"commit":{"parents":[{"commit":"6ca038f350f36ee13ea45f50ddf3624e43e32e6a","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ca038f350f36ee13ea45f50ddf3624e43e32e6a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-08 21:13:54.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-08 21:49:49.000000000","tz":-420},"subject":"Removed unused SQL functions and better coverage","message":"Removed unused SQL functions and better coverage\n\n- Cleaned up indentention.\n- Fixed minor inconsistencies.\n\nChange-Id: Iea61a0af65afc4bf91c581869984115a6d44011f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/617c561cf2e6698e7e9e56a0b89d4bd9dc151c66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/617c561cf2e6698e7e9e56a0b89d4bd9dc151c66"}]},"branch":"refs/heads/master"},"ab3b57b044eb5a662f6299e8073dc32311d14c14":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-10-08 21:52:09.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/73/860773/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/860773/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/860773/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/860773/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/860773/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/860773/5"}}},"commit":{"parents":[{"commit":"6ca038f350f36ee13ea45f50ddf3624e43e32e6a","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ca038f350f36ee13ea45f50ddf3624e43e32e6a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-08 21:13:54.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-08 21:51:48.000000000","tz":-420},"subject":"Removed unused SQL functions and better coverage","message":"Removed unused SQL functions and better coverage\n\n- Cleaned up indentation.\n- Cleaned up sql syntax.\n- Fixed minor inconsistencies.\n\nChange-Id: Iea61a0af65afc4bf91c581869984115a6d44011f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab3b57b044eb5a662f6299e8073dc32311d14c14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab3b57b044eb5a662f6299e8073dc32311d14c14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
