)]}'
{"id":"openstack%2Fdesignate~860774","triplet_id":"openstack%2Fdesignate~master~I5cf8681f63bcf9895fd1873f296630dd27ae9aca","project":"openstack/designate","branch":"master","topic":"tune","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-10-09 05:13:15.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I5cf8681f63bcf9895fd1873f296630dd27ae9aca","subject":"Cleaned up and optimized sqlalchemy base","status":"ABANDONED","created":"2022-10-09 01:58:37.000000000","updated":"2022-10-09 05:29:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"43ef9379f300fc7b61d92a0bf750c19bb7a4ec17","_number":860774,"virtual_id_number":860774,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-10-09 03:23:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-09 03:23:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"18f38bc554916e26d835af6eba0430457678a14d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-09 01:58:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbb9075610fdb09ca5e5c302acc3d3d2aaa02c59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 03:23:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/27d7376b558b496caa9f7e571642ea0c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/036d6d0637f345708fc78902ac5a27d2 : FAILURE in 59m 02s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e80c871ad5004a79902994837a397db0 : SUCCESS in 1h 07m 21s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5596b929c68f4e92874086f31ccffd16 : SUCCESS in 59m 49s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2255fd754cb64455ad7c6714966b72e6 : SUCCESS in 48m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9394a83e92c24ca3881d043e369344a8 : FAILURE in 48m 13s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/1d95007cf0824200ba17d9f6d5be6c87 : FAILURE in 1h 10m 52s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6eff199c434043cd97a21aa04fbe7e15 : SUCCESS in 1h 18m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cab7daca970e41f78d93da2f21a5eb2f : SUCCESS in 54m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8b7fd9fee419444692accdbb36cc3877 : SUCCESS in 40m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d83e69a0043c4796b1a65106039cf70f : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c114676d69e474789844df1800cc893 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29d0acf10e104269a95af7d5fb7dd8b0 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa7e718caa2140f8a761b0ce8a93711a : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c139bebf8587444996e13be29e37357a : SUCCESS in 8m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9e1be58f89db4d93b6af6a387bbd3143 : SUCCESS in 30m 52s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/197dca5234b14cdbbc99d5d65e8fb876 : SUCCESS in 4m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"68fa1ad68059ccc534ccf0c76a0f41980d7b88ed","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-09 05:13:15.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"43ef9379f300fc7b61d92a0bf750c19bb7a4ec17","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-09 05:29:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e0b552f0cba419d6c2fcb78be4d1e9d9819a41ca","revisions":{"e0b552f0cba419d6c2fcb78be4d1e9d9819a41ca":{"kind":"REWORK","_number":1,"created":"2022-10-09 01:58:37.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/74/860774/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/860774/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/860774/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/860774/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/860774/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/860774/1"}}},"commit":{"parents":[{"commit":"ab3b57b044eb5a662f6299e8073dc32311d14c14","subject":"Removed unused SQL functions and better coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab3b57b044eb5a662f6299e8073dc32311d14c14"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-09 01:57:31.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-09 01:57:31.000000000","tz":-420},"subject":"Cleaned up and optimized sqlalchemy base","message":"Cleaned up and optimized sqlalchemy base\n\n- Cleaned up indentation.\n- Cleaned up sql syntax.\n- Fixed minor inconsistencies.\n\nChange-Id: I5cf8681f63bcf9895fd1873f296630dd27ae9aca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e0b552f0cba419d6c2fcb78be4d1e9d9819a41ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e0b552f0cba419d6c2fcb78be4d1e9d9819a41ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
