)]}'
{"id":"openstack%2Fdesignate~860783","triplet_id":"openstack%2Fdesignate~master~Icae82ed323d25d85a35e12b8b4c980f0f71d5d85","project":"openstack/designate","branch":"master","topic":"unicode","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-10-12 03:36:48.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Icae82ed323d25d85a35e12b8b4c980f0f71d5d85","subject":"Remove unicode","status":"MERGED","created":"2022-10-11 04:29:05.000000000","updated":"2022-10-12 05:42:58.000000000","submitted":"2022-10-12 05:41:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860783-unicode","meta_rev_id":"e5073084bd10ee6b724fd30dd97b17d6b30be4d5","_number":860783,"virtual_id_number":860783,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-12 05:41:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-10-11 16:14:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-11 16:14:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-10-12 02:01:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-11 05:52:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-11 14:19:31.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-10-11 15:08:15.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"676c7e1b754954d1c99d269126b0365075c1c720","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-11 04:29:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9713c74f99cf6b50686d99a85cb1193b8b17e75f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-11 04:41:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9b7b68fa10e22eb08d0de7a763743a6bdfe9136c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 05:52:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/294886a183d04f4a98945cb34b632919\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ff8edde1a9ce4fdab6d6d090bcd2aeb4 : SUCCESS in 4m 09s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8a7c16f700184db2905c362fbf40ef49 : SUCCESS in 49m 14s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1f66a1af2ba945fb9e134d69a6868801 : SUCCESS in 45m 03s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5a2bb9fa72043e893c35c01eca89c53 : SUCCESS in 35m 33s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ed7aae26eb904b5a89fb6cc0d61f20d2 : SUCCESS in 51m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/dc51097dde5442b0b9c85af4682ea621 : SUCCESS in 46m 41s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/939c602d36d942c9b96594bd56c85ac4 : SUCCESS in 1h 04m 53s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4653f2a778ae47518296fd06e73daf85 : SUCCESS in 1h 03m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/84d78d5a698e417b98774f58ebab1468 : SUCCESS in 49m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fcc147251758485289c9cb670e9ca1ec : SUCCESS in 48m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85a4b12cebdc48f4aa2384212048aa16 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b97035a09aa543f19800163e1d01d822 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/482cddabca8d440da304eb132d1ac8f3 : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee14e5b45fdf44f5bab3f71b546d302b : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19ed4a2e06b14d62a0783060c81009c9 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21ebce3ce2ed48dea725cefb50e9bb73 : SUCCESS in 4m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2a81ee571a4e498183201095035863fd : SUCCESS in 32m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/caa1cf68d21741da836ba2f28bdd18a6 : SUCCESS in 4m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c27433796e329368367a423e1212397180f123fc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-10-11 14:19:31.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3a9ac921015d893466ff088afa2136067c3456d9","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-10-11 15:08:15.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"721abd39a9813d156b735e78fe077d92da14a8cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-11 16:14:50.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4a880315a69ce6ca462f2aaa23ceddbd9e670bf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 17:09:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1558e51528a4a568158213a8cf292a7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8e023b6691cd462eb5ae13acb67ae7b4 : SUCCESS in 3m 30s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/60307f596c934a44a7c0ff6e97cd32cf : SUCCESS in 36m 34s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a0903bbadb06493e883dadfb54495854 : SUCCESS in 48m 39s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5fa881dc02c24fd182efeb344bceb6f1 : FAILURE in 42m 34s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/9738f7706a5545359e6b0279573f34d5 : SUCCESS in 43m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6e189abedaf84f6098bda4d15898cf10 : SUCCESS in 35m 28s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/83aa547915f34a2f927904d736972146 : SUCCESS in 47m 15s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2d774e8fcf024dce994324a05f6746b4 : SUCCESS in 48m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/09dc989e206a4782b106a33b2bf627d7 : FAILURE in 29m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e3e660b83f0c4b56a7e60837d0967b2a : SUCCESS in 32m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2626d843b054fbb92ae44113998235d : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/add5e5242e504b27a702a8eec9d937f9 : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9eab0a3452974dcb9fae87e2c269c58d : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d00419531bc14e4c912dfe69844c1020 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b1aada21dac4a778f7611f1096de2ab : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/825cb20888904438945147af88d83116 : SUCCESS in 4m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dfd7367415794512a3fcea9c5b814702 : SUCCESS in 27m 39s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/39ec2ba0f86d41569ddbcb5641adde14 : SUCCESS in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0c6bda38e956aa5e25aa1f5aa5a6cbf391f31611","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-11 18:46:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f2826674a4b41a8596852617879c199f95d0dd3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 19:47:26.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/902350bbef3b4976b2b581b9a3b5d77d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/719fe6bcea064311abed2c74018f9467 : SUCCESS in 3m 11s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/58a5d3c4f4c7453b817cd2fbbb02a122 : SUCCESS in 32m 35s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/af749641cf0a489f8d06f6dbc024a7ed : SUCCESS in 55m 40s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/24d7da65a0994e0ba32fc54c0c4aa9b1 : SUCCESS in 41m 29s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/127443fc54bb4be89fc724f4022437fe : SUCCESS in 30m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ddb90e9fa4ed49b48d139e4e1348b41c : SUCCESS in 41m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/46aca6c546494152a3f253d24cd1a518 : SUCCESS in 52m 09s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4416a95b81794c878a5d2af24c3e0671 : SUCCESS in 56m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8857e87d5e164dcdb5a9ef8ebccd98a5 : SUCCESS in 32m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cbcdfb6314474074bf19f00fff368516 : SUCCESS in 43m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2290253d8ba541fc89027ca62ab89dad : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/836f9d89cc0f4954a7fb1df57fd009f5 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a631b172c0842b18e458528e34bc9d2 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4db2d5a122b644909c42fe2ac69241e6 : FAILURE in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e76d4d02281a4af08ae9d9521f59b4a3 : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6cdca81bdb34768b5f3dde25e0dc0f0 : SUCCESS in 4m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/854c3c8a63e94b988fa7c1bcfb000b3e : SUCCESS in 31m 24s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c4adf5a6372349338af4d88ff27454b6 : SUCCESS in 4m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"86eb1619b322469b6f08b9bfae4969d896c6092f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-11 20:18:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dc43a74071ab4635765c47e6b30fe1d6d56434be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 21:11:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75e985faef30406e9f51945d308e228a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/834cd83f37874a1ebe535875dc79094d : SUCCESS in 3m 29s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e14c7af87a214179950aae00160b6050 : SUCCESS in 38m 51s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/74460527ebee4bcc97946222920e10d2 : SUCCESS in 45m 10s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/843edab6258b4b0cbee6ff1381d778e5 : FAILURE in 41m 42s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/6a31ac03d6c049488db37d5ddf73a51b : SUCCESS in 34m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d3abe8b4b9124b4cb72f7735811ec9d3 : SUCCESS in 36m 11s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2723ed2e43e94813be3f9249ae629177 : SUCCESS in 50m 34s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b29b9551eadb4f34ac35e95ecf9eb987 : SUCCESS in 42m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/39f133303e5d4b28a073036946f0e09b : SUCCESS in 32m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3999c57d9ab5413b992575c8261c3195 : SUCCESS in 31m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/853f68bd19ec43b5a0107e5415469b74 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e28df6082cb4d3a8751a3e9fe7fc211 : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79e9dc06a8764794aa970b4a9c0fc9cc : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3362f91d87f044dc92d31ba1cd52ca48 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8faa7950eba49528614a46bf82847d7 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7aa8cb59bfc44d8f9de587c56a6ca688 : SUCCESS in 4m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f460e6c4d035427d8a3638aee4b8ca1d : SUCCESS in 30m 01s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3003e1b9f819466fb7a1f53dc49e8493 : SUCCESS in 3m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6b577fb3529edf89d394b194ed5a88f253c8d270","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-12 02:01:24.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4df77d558c39ff33c6865ff8db33a48793c5b21a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 02:05:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"01d28d4c8acc81c22880f23fce086632f5fe2f18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 03:25:02.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b68d55eae164bf2b9a14136146e8781\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d6c914e2a3c34300b1b259ce84350464 : SUCCESS in 4m 45s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a53cc7177eed4f769caf68d88df4dc20 : FAILURE in 34m 51s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/21edf03807af4dde9de2ca496e5a4a8a : SUCCESS in 40m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4f8d356e8bac45849f7a540ebecd41ae : SUCCESS in 37m 02s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/dae200930dd54a04ad1680591eccc538 : SUCCESS in 52m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6cdb212c1998467d8d5f137e845771c8 : SUCCESS in 40m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c11cb7619da34165b9f416691699c157 : SUCCESS in 53m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e131fae381948c492d043f489563654 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a0f0bbe410e4d7fabe667cae6dd0f0a : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/733fa133cda2476dba55c26dca8e7cdb : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9fb194db9c84e94baa433d75b2ce1cb : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f4feaa8ba904b63be0bafc62b699c78 : SUCCESS in 3m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3674cdb50de444e8b17d1c6836c12407 : SUCCESS in 30m 14s","accounts_in_message":[],"_revision_number":3},{"id":"5485fdafcf75bbaa6d4e074c347140ad64508056","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-12 03:36:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2592d346c7137f04c7ef47d231f46d539187263e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 04:35:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd5cc0ae177f41fa99828656b9e816b2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1e85d4cf49734a7a8129b6c4a3ba447d : SUCCESS in 4m 56s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9f955e0b919b46679c97f71a1dec6758 : SUCCESS in 44m 24s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/056da85726a549c98cb3541d8ad6cc2c : SUCCESS in 52m 10s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8bbb3f16009b41f29e474c72eb02d959 : SUCCESS in 41m 30s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/9561c61fc165451286e0c918eacb93da : SUCCESS in 45m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d47091921a304ca7b47712a9c9780d04 : SUCCESS in 42m 30s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/05a097f528f248078350d7ab2ab19575 : SUCCESS in 55m 07s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f6da78c704094797a697c63e72d336ba : SUCCESS in 52m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/51e28574848f49b1841cf43b0d03c9cf : SUCCESS in 38m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/35e1a506d3da4a39a1123ba626d2db68 : SUCCESS in 39m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8021c02cede473da7fe3a6193d050dd : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a67896538a4446c496cc86abfdb7f0ee : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b277f59e6e84192a00f42e8e14a386a : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/111364e2438c4defad6de3e4206a2164 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21184eadded449b29075bab4270177c8 : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d76c2f5ea4fe45e6a4a0ebbeb9134e53 : SUCCESS in 5m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fc37de416711483d820cda4a6d74afa8 : SUCCESS in 34m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/de07aa9e6862477e8a4e320e72da8633 : SUCCESS in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8e912544cd0399aad2a6b1974ade6eb03efb98b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 04:35:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fd0a6ba48c917d5d5634bc5ef51651ca3b12b4a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 05:41:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e37b8a872f9a4163948022019e19240a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c687e99976664ae29657b34999b8239b : SUCCESS in 4m 11s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2bbfbb21f4384fee8f9a34be5f15ef49 : SUCCESS in 47m 03s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/85179fcb229b4185a0502509e274fb64 : SUCCESS in 47m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e913816c4ced4bb4bf8ca6b12f303f62 : SUCCESS in 42m 42s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a06bd353161d4944972b1074ea2bbc26 : SUCCESS in 1h 00m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c66f2e7cce1a4470b4b0382ef910c529 : SUCCESS in 43m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/03d953f8df6b4839a7b4e7749034ecb0 : SUCCESS in 55m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7049441289b74a4899a0db650e9633f8 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26647c65147d419da426bcfbb13f1503 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b6ccabdc21c40f8a8fc1f752110f1fd : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f0ee961566d488686c91c0a4caca7cb : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72cc12a5e3014b0aba848d91ef6758fa : SUCCESS in 4m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/04967e09c10f4f2f9242c25579b38690 : SUCCESS in 31m 07s","accounts_in_message":[],"_revision_number":3},{"id":"ea4be4a46f68589727ea151bf41f349ebc960092","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 05:41:47.000000000","message":"Change has been successfully merged\n\n2 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":3},{"id":"e5073084bd10ee6b724fd30dd97b17d6b30be4d5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 05:42:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15fe64a7c1a647f0bb99552fd660c59d\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/c696d8929e574d1c9ee2997972cc8292 : SUCCESS in 42s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c30080042b224e22a54575c326811839 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9db92b0718c4ebb8cd62445a7c0f4ff : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6fee42ecac2ff1f3a8393ab069c1043ce891f4ef","revisions":{"a37a0c80fe2093bca55f1e456bf642d163cbc4cc":{"kind":"REWORK","_number":1,"created":"2022-10-11 04:29:05.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/83/860783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/83/860783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/83/860783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/83/860783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/83/860783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/83/860783/1"}}},"commit":{"parents":[{"commit":"0ca13a84e380b8799d310e7262e464031aef96e7","subject":"Cleaned up and optimized sqlalchemy base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0ca13a84e380b8799d310e7262e464031aef96e7"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-11 04:28:37.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-11 04:28:37.000000000","tz":-420},"subject":"Remove unicode","message":"Remove unicode\n\nNo longer necessary as we no longer support Python 2.7.\n\nChange-Id: Icae82ed323d25d85a35e12b8b4c980f0f71d5d85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a37a0c80fe2093bca55f1e456bf642d163cbc4cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a37a0c80fe2093bca55f1e456bf642d163cbc4cc"}]},"branch":"refs/heads/master"},"9a2410d0e263147da1fe63cfb4640fc39a6cf51c":{"kind":"REWORK","_number":2,"created":"2022-10-11 04:41:48.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/83/860783/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/83/860783/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/83/860783/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/83/860783/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/83/860783/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/83/860783/2"}}},"commit":{"parents":[{"commit":"0ca13a84e380b8799d310e7262e464031aef96e7","subject":"Cleaned up and optimized sqlalchemy base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0ca13a84e380b8799d310e7262e464031aef96e7"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-11 04:28:37.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-11 04:41:37.000000000","tz":-420},"subject":"Remove unicode","message":"Remove unicode\n\nNo longer necessary as we no longer support Python 2.7.\n\nChange-Id: Icae82ed323d25d85a35e12b8b4c980f0f71d5d85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a2410d0e263147da1fe63cfb4640fc39a6cf51c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a2410d0e263147da1fe63cfb4640fc39a6cf51c"}]},"branch":"refs/heads/master"},"6fee42ecac2ff1f3a8393ab069c1043ce891f4ef":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-10-11 16:14:50.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/83/860783/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/83/860783/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/83/860783/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/83/860783/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/83/860783/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/83/860783/3"}}},"commit":{"parents":[{"commit":"6ca038f350f36ee13ea45f50ddf3624e43e32e6a","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ca038f350f36ee13ea45f50ddf3624e43e32e6a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-11 04:28:37.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-10-11 16:14:50.000000000","tz":0},"subject":"Remove unicode","message":"Remove unicode\n\nNo longer necessary as we no longer support Python 2.7.\n\nChange-Id: Icae82ed323d25d85a35e12b8b4c980f0f71d5d85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fee42ecac2ff1f3a8393ab069c1043ce891f4ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fee42ecac2ff1f3a8393ab069c1043ce891f4ef"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
