)]}'
{"id":"openstack%2Fdesignate~860964","triplet_id":"openstack%2Fdesignate~stable%2Fwallaby~If7aa4f2172b07a3538f3aea3a7df5a291d8c93b1","project":"openstack/designate","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-02-23 09:04:13.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"18551":{"account":{"_account_id":18551,"name":"Szymon Datko","email":"szymon.datko@gmail.com","username":"sdatko"},"last_update":"2023-09-20 15:02:42.000000000","reason":"\u003cGERRIT_ACCOUNT_18551\u003e replied on the change","reason_account":{"_account_id":18551,"name":"Szymon Datko","email":"szymon.datko@gmail.com","username":"sdatko"}},"13995":{"account":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"last_update":"2024-03-05 19:03:18.000000000","reason":"Change was abandoned"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-03-05 19:03:18.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If7aa4f2172b07a3538f3aea3a7df5a291d8c93b1","subject":"Restrict virtualenv to avoid suds-jurko dep issue","status":"ABANDONED","created":"2022-10-11 16:26:43.000000000","updated":"2024-03-05 19:03:18.000000000","total_comment_count":14,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"42dafc8376dea856bff3fc50aedba7511c29647c","_number":860964,"virtual_id_number":860964,"owner":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-10-13 17:05:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2022-10-17 09:14:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-11 16:30:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2022-10-11 16:31:07.000000000","updated_by":{"_account_id":18551,"name":"Szymon Datko","email":"szymon.datko@gmail.com","username":"sdatko"},"reviewer":{"_account_id":18551,"name":"Szymon Datko","email":"szymon.datko@gmail.com","username":"sdatko"},"state":"REVIEWER"},{"updated":"2022-10-11 17:11:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-13 09:48:38.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-10-17 18:07:27.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2023-09-20 15:02:42.000000000","updated_by":{"_account_id":18551,"name":"Szymon Datko","email":"szymon.datko@gmail.com","username":"sdatko"},"reviewer":{"_account_id":18551,"name":"Szymon Datko","email":"szymon.datko@gmail.com","username":"sdatko"},"state":"REMOVED"},{"updated":"2024-03-05 19:03:18.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"0016a079dc024e93890bc7f8b1db53b58386e520","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2022-10-11 16:26:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"322f24b76d763016dcf4b486de6a539af1e3dc05","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-10-11 16:30:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"27fd7cea8ab563f97eb1d9d69f2895a59315f121","author":{"_account_id":18551,"name":"Szymon Datko","email":"szymon.datko@gmail.com","username":"sdatko"},"date":"2022-10-11 16:31:07.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f5fd75435c687977e3c0608ae775ce7a4b0df3fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2022-10-11 16:33:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"177e230d78a973350a0af0c891ae69479142d833","author":{"_account_id":18551,"name":"Szymon Datko","email":"szymon.datko@gmail.com","username":"sdatko"},"date":"2022-10-11 16:41:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"568f3620ad516350553b64726c32ff30a372b0b2","author":{"_account_id":18551,"name":"Szymon Datko","email":"szymon.datko@gmail.com","username":"sdatko"},"date":"2022-10-11 16:41:38.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1b72ee16dbe2d29ba33fc305a2aac7459fb55c1c","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2022-10-11 16:50:09.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f38493672235c11b1031b2611188170b6547c7ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 17:11:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e6e33db4d4449398954252f4fe52490\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0ea9298ed41a4647bde5b25f1709a16b : SUCCESS in 34m 41s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/90267fdb79174d38a3649dc8e4eb9577 : RETRY_LIMIT in 7m 06s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/cbd7b910d14e451a979d6cd80371e5f3 : SUCCESS in 33m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bff6dcda599c461f8e72b6f234078b7c : SUCCESS in 33m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b76604d579c549bfa6c1fa7da6843339 : SUCCESS in 35m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ad2800ee3ce4474b8c0cb97f9ab19de3 : SUCCESS in 36m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c4f3c5019db43289855741d8a6ec8df : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3af71bbece3c4b1498a4976af4b07cc7 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8713ddbbc66e4bed96912116ccc5e0a5 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bb4833084754b4da7078c9117e02d4a : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c409e5fa1bf40d5b856024d5acd11f4 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd5493acdd3f447fbddab0ffd05a5417 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14ce3ffd66aa432ca3344e18f7050ca9 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":2},{"id":"edd1de206349e605ffd07be686e23c003ab9f875","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-10-11 17:12:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b8bb11cc5ce0401d3302d518b3d4ff40f2afa808","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2022-10-11 18:25:46.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"20646097acfe2d2c914f2b1064dc1d2623caa732","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2022-10-11 18:26:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e529cc93ce457c604cb4b96b4e263ad06c1bfe48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 19:28:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb01907d74eb4306a9de83b8842200b0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/424001e76bee461091ccf9cb3b1db6bd : SUCCESS in 36m 39s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/955626fd6bbb407e9a8820a6a5c2d582 : RETRY_LIMIT in 7m 29s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/90c6d58d04c64d3b91fcc31679aa67f7 : SUCCESS in 46m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/21658d72adb14f83856e2788d45d6a00 : SUCCESS in 45m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9ff19be5767a4166ab806ba3567e4361 : SUCCESS in 43m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f9967681fd5d47039a2e73ddc82baeda : FAILURE in 56m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c6c21e00f1440f6b4b742fadddacc9a : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/497227b188ab4ce2bf6fd1857a84b1c7 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f500f225144f4007b8ef0f05366dd817 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/693890b089324d729fab68e25dd0e366 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/144225b6aaf04b29b237367e24d9217c : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f182b1e11f8d4e81840f731c3d50f842 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8dd59128535844fb86292c921f2c799b : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":3},{"id":"6222ba7623187345edaa33b112e8801bd3d4c0d4","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2022-10-12 20:00:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e7fdcdb460aa42acc74a5c4339a195b0b0e3e81a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 20:50:46.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3db661e06d6b43a19d56f08e5a88d6ce\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/26fff69a9ac14db18ab999f4b66dbd40 : SUCCESS in 40m 48s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/7e1ed043ee884d279bbc846762039758 : RETRY_LIMIT in 8m 55s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/5c392838f96246e2a828163858ca38c8 : FAILURE in 37m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cc28fa3d0257450c849b110f392713d6 : SUCCESS in 33m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/956f8e34d3dc40278a6b0a81fc99565d : SUCCESS in 30m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/dc1fd532002a4e37b46e0d71c626b839 : SUCCESS in 33m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6a11cc01909462eba50b1c17e131f66 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e26daab3dcaf46aea786332af8873ff8 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c943f4f9861463b9c094aeefffc98ce : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cc7d54eb4964f1aa95539f7462b107a : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f0b35ffe6914bd5a65b8cdd42d4372d : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76a6f0f98a8048559b4400cd70e0f1f7 : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbfba6a71a0c4cbda4610aecf9779c83 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":3},{"id":"7a6e61d0a6abe0b9c3a4d5035329872460e6320b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-13 09:48:38.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e531f4c0c6be53cdfa57473b869676ddc14d591b","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2022-10-13 16:03:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"52e04767c1b9e4b4001e3bd52dec17f9e8a9942b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2022-10-13 16:05:02.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"246b7b74995b6cbf05bd004340690cc949caf8b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 17:05:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a79ad672da548fc9409fcac586231df\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ee93794591124575a17760380af00315 : SUCCESS in 48m 45s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/3d99ff553a52419f9b8039612af19ff8 : RETRY_LIMIT in 8m 21s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/d0708652570b480da017b34c00df491e : SUCCESS in 49m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/515b3a4d561443ad960e095fd48179bd : SUCCESS in 46m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/495537867cf54a86b6b707f84483a254 : SUCCESS in 46m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3c7e523d6ec9482e9395296e77bb7555 : SUCCESS in 53m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e77636ccf7fd42419f15b5ae35f154e9 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eebf848cb8643708151ee68147de157 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eda716cd3cd942419244a19ce80d7243 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b5161fbc0ed48ef98b6402d91d514bd : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d53b17fe9b3240138adb60b1187e88d3 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee13d41237d1480eadc05a631992b2b1 : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca66af3962ad4f6abf4422445c836a21 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":4},{"id":"5c24f87c1b8905f41e43d31b8193ba67f2cb1089","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-17 09:14:55.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5ecb6a36e8f0987dbdb2b68eded540c570b82de8","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2022-10-17 14:54:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"50993584cda88e0cb2c2e66cf8eeada28029b6aa","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-17 18:07:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dbcb998deb3166ca02f6ccb1d3d9162255e53bc2","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":18551,"name":"Szymon Datko","email":"szymon.datko@gmail.com","username":"sdatko"},"date":"2023-09-20 15:02:42.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_18551\u003e.","accounts_in_message":[{"_account_id":18551,"name":"Szymon Datko","email":"szymon.datko@gmail.com","username":"sdatko"}],"_revision_number":4},{"id":"42dafc8376dea856bff3fc50aedba7511c29647c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-05 19:03:18.000000000","message":"Abandoned\n\nstable/wallaby branch of openstack/designate is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/wallaby if you want to further work on this patch.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2559524b710c6c85afc7915765c8eaff04b3392b","revisions":{"e128fd7622dcc09f7de709818c7f6dd19cc29660":{"kind":"REWORK","_number":1,"created":"2022-10-11 16:26:43.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/64/860964/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/64/860964/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/860964/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/860964/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/860964/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/64/860964/1"}}},"commit":{"parents":[{"commit":"3f4847384d3e7c6bcb7f8d13ee62b0f9141c91fc","subject":"Merge \"Check zone_id in notification_handler to avoid UUID format issue.\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3f4847384d3e7c6bcb7f8d13ee62b0f9141c91fc"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2022-10-11 16:24:15.000000000","tz":-240},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2022-10-11 16:26:43.000000000","tz":-240},"subject":"Restrict virtualenv to avoid suds-jurko dep issue","message":"Restrict virtualenv to avoid suds-jurko dep issue\n\nSince an unconstrained virtualenv can cause breakage in CI systems\nusing much later versions of packages, constrain it in tox.ini.\nThis resolves an issue for stable/wallaby that was later resolved in\nyoga [1] without requiring an invasive backport.\n\n[1] https://review.opendev.org/c/openstack/designate/+/813380\n\nChange-Id: If7aa4f2172b07a3538f3aea3a7df5a291d8c93b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e128fd7622dcc09f7de709818c7f6dd19cc29660"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e128fd7622dcc09f7de709818c7f6dd19cc29660"}]},"branch":"refs/heads/stable/wallaby"},"469fb33a415ec895e72a3343cbf3255db401d14f":{"kind":"REWORK","_number":2,"created":"2022-10-11 16:33:52.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/64/860964/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/64/860964/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/860964/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/860964/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/860964/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/64/860964/2"}}},"commit":{"parents":[{"commit":"3f4847384d3e7c6bcb7f8d13ee62b0f9141c91fc","subject":"Merge \"Check zone_id in notification_handler to avoid UUID format issue.\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3f4847384d3e7c6bcb7f8d13ee62b0f9141c91fc"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2022-10-11 16:24:15.000000000","tz":-240},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2022-10-11 16:31:58.000000000","tz":-240},"subject":"Restrict virtualenv to avoid suds-jurko dep issue","message":"Restrict virtualenv to avoid suds-jurko dep issue\n\nSince an unconstrained virtualenv can cause breakage in CI systems\nusing much later versions of packages, constrain it in tox.ini.\nThis resolves an issue for stable/wallaby that was later resolved in\nyoga [1] without requiring an invasive backport.\n\nThis also implies a bump of tox version to one released in 2018.\n\n[1] https://review.opendev.org/c/openstack/designate/+/813380\n\nChange-Id: If7aa4f2172b07a3538f3aea3a7df5a291d8c93b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/469fb33a415ec895e72a3343cbf3255db401d14f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/469fb33a415ec895e72a3343cbf3255db401d14f"}]},"branch":"refs/heads/stable/wallaby"},"ea8c75fe28fe810365edb85181dc3f616869e6ea":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-10-11 18:25:46.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/64/860964/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/64/860964/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/860964/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/860964/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/860964/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/64/860964/3"}}},"commit":{"parents":[{"commit":"3f4847384d3e7c6bcb7f8d13ee62b0f9141c91fc","subject":"Merge \"Check zone_id in notification_handler to avoid UUID format issue.\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3f4847384d3e7c6bcb7f8d13ee62b0f9141c91fc"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2022-10-11 16:24:15.000000000","tz":-240},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2022-10-11 18:25:24.000000000","tz":-240},"subject":"Restrict virtualenv to avoid suds-jurko dep issue","message":"Restrict virtualenv to avoid suds-jurko dep issue\n\nSince an unconstrained virtualenv can cause breakage in downstream\nCI systems using much later versions of packages, constrain it in\ntox.ini.  This resolves an issue for stable/wallaby that was later\nresolved in yoga [1] without requiring an invasive backport.\n\nThis also implies a bump of tox version to one released in 2018.\n\n[1] https://review.opendev.org/c/openstack/designate/+/813380\n\nChange-Id: If7aa4f2172b07a3538f3aea3a7df5a291d8c93b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ea8c75fe28fe810365edb85181dc3f616869e6ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ea8c75fe28fe810365edb85181dc3f616869e6ea"}]},"branch":"refs/heads/stable/wallaby"},"2559524b710c6c85afc7915765c8eaff04b3392b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-10-13 16:05:02.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/64/860964/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/64/860964/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/860964/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/860964/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/860964/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/64/860964/4"}}},"commit":{"parents":[{"commit":"3f4847384d3e7c6bcb7f8d13ee62b0f9141c91fc","subject":"Merge \"Check zone_id in notification_handler to avoid UUID format issue.\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3f4847384d3e7c6bcb7f8d13ee62b0f9141c91fc"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2022-10-11 16:24:15.000000000","tz":-240},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2022-10-13 16:03:57.000000000","tz":-240},"subject":"Restrict virtualenv to avoid suds-jurko dep issue","message":"Restrict virtualenv to avoid suds-jurko dep issue\n\nSince an unconstrained virtualenv can cause breakage in downstream\nCI systems using much later versions of packages, constrain it in\ntox.ini.  This resolves an issue for stable/wallaby that was later\nresolved in yoga [1] without requiring an invasive backport.\n\nThis also implies a bump of tox version to one released in 2018.\n\n[1] https://review.opendev.org/c/openstack/designate/+/813380\n\nChange-Id: If7aa4f2172b07a3538f3aea3a7df5a291d8c93b1\nCloses-Bug: #1992825\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2559524b710c6c85afc7915765c8eaff04b3392b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2559524b710c6c85afc7915765c8eaff04b3392b"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
