)]}'
{"id":"openstack%2Fdesignate~861098","triplet_id":"openstack%2Fdesignate~master~I9cc94506a63371ad78b204030e5e87339982e9e7","project":"openstack/designate","branch":"master","topic":"bug/1992677","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-10-13 09:13:02.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-10-13 11:36:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9cc94506a63371ad78b204030e5e87339982e9e7","subject":"Fixes compatibility with jsonschema \u003e\u003d 4.16.0","status":"MERGED","created":"2022-10-12 16:53:58.000000000","updated":"2022-10-13 11:37:27.000000000","submitted":"2022-10-13 11:36:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861098-bug/1992677","meta_rev_id":"7dba95eafa6e2fb189762e5bc1aed964c94be297","_number":861098,"virtual_id_number":861098,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-13 11:36:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-10-13 06:19:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-10-13 06:19:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-12 18:04:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-12 20:53:41.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-10-13 06:19:40.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"e26ac3af334313cd288e64fef755442d1b231df3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-10-12 16:53:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c774bf5b7eb428381b75b001ceb918515257b510","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 18:04:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a47e6a6392084d4692d7322d81c52b7f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/00f0691e3c3a44a29c5ad27114eb2424 : SUCCESS in 34m 17s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c64697dd96504f18b4099479b6e2aaaa : SUCCESS in 1h 04m 19s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/63bbdb4eb25d438e9ec217ff074a3155 : SUCCESS in 46m 44s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/677e234148e14a95b17145533a2f1cc5 : SUCCESS in 35m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/845c7f9600a7490cbfa8a12343e4ed3f : SUCCESS in 35m 45s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e4de73e829c34e78adc9ac8397db8bef : SUCCESS in 1h 02m 04s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b4a14a7554114eaa9a031686f00563a8 : SUCCESS in 46m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/46d9acc50d68462b80c601e25f47d74d : SUCCESS in 43m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ee676de3ef744ee9989b655776aadd37 : SUCCESS in 34m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b6ad09a83454a22a22bde5e27272017 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e412c716090448aaf89ccfe846e9c22 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1a4e8b6ee2540139ddf23f24ed78adb : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/206cb1fc8fba472dafa08baeb60fe96a : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d46b2c782aea49a08c99c330da0341cc : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/431ea0d6646e4cdd8374882cf2671284 : SUCCESS in 3m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7dba49d4492c4ac081881c081e7be23b : SUCCESS in 28m 11s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a1b55d908ed1481bbaf22a4d4c9db0a5 : SUCCESS in 4m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"19e260e67ce5613bc506ddb73a0fbe1a5e03c8a6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-12 20:53:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ac2c0ac74ab0dd30999b7b7834921b8876b7bab1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-13 06:17:29.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"24bb6bd2b13a957cf3014f500f0b986b53e724d9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-13 06:19:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"29a3eaf1219bda97b87ff1d789da5ba426210e0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 07:32:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d70599d2969c4ae7a5b27bb0b22f84a6\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dab2d449a1dc408797a8c26bfd78ddf5 : SUCCESS in 41m 38s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b3bb20c8bc6c47d1b1de894e907047dc : SUCCESS in 1h 08m 51s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/04c8d8a1ff42477ba1c30d1998bb86f5 : SUCCESS in 42m 47s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e09b1917772d4cb4b6cf20a237d17c9b : SUCCESS in 42m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/04db3d8950dd4d74a08a09997c99a9b2 : SUCCESS in 37m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a141cfd2266f488097552ae6055348d5 : SUCCESS in 1h 04m 18s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/abf4aa0cdcc840f798c7f2536538d1d9 : SUCCESS in 1h 00m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bfbecc17e54f454293a41b3b9e30dd77 : SUCCESS in 48m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b2e5ae746e9b42caae6f8750e32b7efc : SUCCESS in 40m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e85350f106944a8a2a58cbe0267e6ee : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/392497fb53de41c687422edf2e37b9cb : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2561386fdd0741f2a5dbbbb31530b6e1 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b621166cf7a4e7db7d34aec7dfaf709 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/441f7ae322644cd39ffcbf7a86d8762b : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5fd645329744cca84bc89e00bf6ab88 : SUCCESS in 3m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/db13d92b59ae41618213c1707d42dd7c : SUCCESS in 28m 37s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/29607b08be154bfaab674ebb2c1b8711 : SUCCESS in 4m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d43c6f99507bc500a12df476ed7bda6d44fc8b98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 07:33:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b47b93f1ce540be2d15749c17515df7f271e6a5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 08:45:34.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d490940589b42578bc3e70db0a06afe\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/282e40f99f8648dfb074d0541cdf9929 : SUCCESS in 47m 13s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e3863314fe9c4854ad61e89635cc2b8a : FAILURE in 52m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/69e4335000e84bd4870c3ce894e9a462 : SUCCESS in 54m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7e0bc0b5f0274245b78052bbcb948d9d : SUCCESS in 1h 06m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5da90ff3adb7476bb82d9255d769050c : SUCCESS in 43m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5e21d0b16dd04e2da0e2bd0083b09e49 : SUCCESS in 52m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/749fce60b56e4ceeb273c09b1cb8096c : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c5956e1cb9341339279efcf6194f1f4 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ced4573552a44c4ad02ab5c7b02bcc8 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8a8be2f27274ecca364ff8ff60d76a1 : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4b6031f91b842f89c8de5964875e589 : SUCCESS in 3m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ce8edd5e69354a6aa0046c6154c5161a : SUCCESS in 29m 35s","accounts_in_message":[],"_revision_number":2},{"id":"9a1d2260f61982503953277ddc13c47cc4bfbd78","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-13 09:13:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4f1013f21240ab0c36c36348a3d31c91fb5414a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 10:29:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5338461afc6944c59f15c4020e23690f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c8717bd77a1d40b7a5473da822c98b15 : SUCCESS in 48m 05s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4b06835d669745fd9c5682831e6ea638 : SUCCESS in 1h 10m 12s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/60d2fef704df43cba8280be10480ce28 : SUCCESS in 1h 11m 16s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f2fe9739f9654b1dbbac2a67471397df : SUCCESS in 45m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1694a3756c284a9791a61ea1eaaa10a5 : SUCCESS in 40m 05s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/793105ab8ac94ad0b24ca4fb908ee9c0 : SUCCESS in 1h 03m 14s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7254f11b52824a47bc2ee70e15b1c4ab : SUCCESS in 1h 06m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/68749bb9f4d644a79f08b110145a79e5 : SUCCESS in 44m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5fb9575d5492414d8020e287ff2a4be3 : SUCCESS in 45m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12c5b914833947f0b624dd6eedd6777a : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/665067ff228044c1b32c2d4f99caa76c : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3beac9fa0b7d4df6b369212d0892c1bf : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a23f746178764e07997bb4d3cf82ccdd : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14e8c401ed354f0184eeaf06094da2b4 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aff5d6111b4e4a5f9f6765393bd0d3e5 : SUCCESS in 4m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/85d5d885acd84a5d9edc6fdb15385abe : SUCCESS in 31m 15s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/97037228658149df812b28730be085f9 : SUCCESS in 3m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f528df01db0c378dc88647547ea4c481a14950c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 10:30:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f4003d3318883042d5b444da6d2cc724e07c6295","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 11:36:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/648da1917b824415be533257abab3962\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0824562ec82241c68a1ba38bbf6e4f01 : SUCCESS in 48m 13s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/7bf708595a874a699e94440f59aa051d : SUCCESS in 51m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0cce7bd14a5f44f8afff5d5b82254d5c : SUCCESS in 50m 07s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fac4740332aa4c32b6e721124c276660 : SUCCESS in 1h 01m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cbc3563280144b4f828a54ae5cd544bf : SUCCESS in 46m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/59f7b0f7702543ae8d5d30ee1da2d97d : SUCCESS in 43m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4635bbceb8e6494f8859c53bad2bf6d2 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84b249c07eab4e97a6235f6e54c22a23 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef543ef55def4200a9c3f48035851b47 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/170415e960fd406fb9702714ff60d05e : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a8e33090dd04aada2f0e8a276844f66 : SUCCESS in 3m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5f1141a8d201492ab042ba053ed7d124 : SUCCESS in 26m 11s","accounts_in_message":[],"_revision_number":2},{"id":"a4f14dc8abb631f4c5c39581a59c3be858bcf829","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 11:36:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7dba95eafa6e2fb189762e5bc1aed964c94be297","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 11:37:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b3a915faa12451383bc19b81e967ad5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cace3a32086c4ded8bdee2259cf33009 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8714505ec60a4350bc600c90037dabbf : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d6eabef4deeb4632d39aec91a1a83ce9fbe0853a","revisions":{"9f973a0e86ec9dd469452fe023006fe8f0511f6c":{"kind":"REWORK","_number":1,"created":"2022-10-12 16:53:58.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/98/861098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/98/861098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/861098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/861098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/861098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/98/861098/1"}}},"commit":{"parents":[{"commit":"6fee42ecac2ff1f3a8393ab069c1043ce891f4ef","subject":"Remove unicode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fee42ecac2ff1f3a8393ab069c1043ce891f4ef"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-10-12 16:50:57.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-10-12 16:53:57.000000000","tz":0},"subject":"Fixes compatibility with jsonschema \u003e\u003d 4.16.0","message":"Fixes compatibility with jsonschema \u003e\u003d 4.16.0\n\nJsonschema has changed the type of a parameter from a simple tuple to a pyrsistent map object. Since we were passing in the defaults anyway, this patch removes those parameters from the jsonschema call. This makes designate compatible with jsonschema \u003e\u003d 4.16.0.\n\nCloses-Bug: #1992677\nChange-Id: I9cc94506a63371ad78b204030e5e87339982e9e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9f973a0e86ec9dd469452fe023006fe8f0511f6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9f973a0e86ec9dd469452fe023006fe8f0511f6c"}]},"branch":"refs/heads/master"},"d6eabef4deeb4632d39aec91a1a83ce9fbe0853a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-10-13 06:17:29.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/98/861098/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/98/861098/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/861098/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/861098/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/861098/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/98/861098/2"}}},"commit":{"parents":[{"commit":"6fee42ecac2ff1f3a8393ab069c1043ce891f4ef","subject":"Remove unicode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fee42ecac2ff1f3a8393ab069c1043ce891f4ef"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-10-12 16:50:57.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2022-10-13 06:17:29.000000000","tz":0},"subject":"Fixes compatibility with jsonschema \u003e\u003d 4.16.0","message":"Fixes compatibility with jsonschema \u003e\u003d 4.16.0\n\nJsonschema has changed the type of a parameter from a simple\ntuple to a pyrsistent map object. Since we were passing in the\ndefaults anyway, this patch removes those parameters from the\njsonschema call. This makes designate compatible with\njsonschema \u003e\u003d 4.16.0.\n\nCloses-Bug: #1992677\nChange-Id: I9cc94506a63371ad78b204030e5e87339982e9e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d6eabef4deeb4632d39aec91a1a83ce9fbe0853a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d6eabef4deeb4632d39aec91a1a83ce9fbe0853a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
