)]}'
{"id":"openstack%2Fdesignate~862426","triplet_id":"openstack%2Fdesignate~stable%2Fvictoria~I1fe73145e4a4699939244bc857600efa264f2340","project":"openstack/designate","branch":"stable/victoria","topic":"queue-config-template-victoria","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-10-28 10:49:32.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I1fe73145e4a4699939244bc857600efa264f2340","subject":"CI: Move queue configuration into project-template","status":"MERGED","created":"2022-10-23 10:27:08.000000000","updated":"2022-10-28 11:54:34.000000000","submitted":"2022-10-28 11:51:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862426-queue-config-template-victoria","meta_rev_id":"7ec9984c5fdf704259dee57d39e25b58c27f98e1","_number":862426,"virtual_id_number":862426,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-28 11:51:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-10-24 16:13:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-24 03:41:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2022-10-28 10:49:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-23 10:27:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-24 03:41:16.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-10-24 16:13:26.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"25add5d342743e41a7fc3d0a8b625de8f92831c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-23 10:27:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e770813c362d38054d71544e7791bbea70743869","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-23 10:27:29.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/designate on branch stable/victoria.  The error was:\n\n  The key \"queue\" appears more than once; duplicate keys are not\n  permitted.\n\nThe error appears in the following stanza:\n\n  name: designate-devstack-jobs\n      queue: designate\n      check:\n        jobs:\n          - designate-bind9\n          - designate-pdns4\n          - designate-grenade-pdns4:\n              voting: false\n          - designate-ipv6-only-pdns4\n          - designate-ipv6-only-bind9\n      queue: designate\n      gate:\n        jobs:\n          - designate-bind9\n          - designate-pdns4\n          - designate-ipv6-only-pdns4\n          - designate-ipv6-only-bind9\n\n  in \"openstack/designate/.zuul.yaml@stable/victoria\", line 126, column 5","accounts_in_message":[],"_revision_number":1},{"id":"5a93ffb66495e1119c5f7200fbda4ff99bb1183c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-23 11:46:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f055f7729454aabe08ea1e8868599682889e2335","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-23 12:39:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0a35510e567498098648a75a9927879\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f8b032fc65e6488e8ba1d4146eb11d54 : SUCCESS in 37m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a78943c4d7174f1ca23fe17fb92a51a7 : SUCCESS in 29m 04s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cebbeb195efe4f6c81d56f0c42f96898 : SUCCESS in 51m 40s (non-voting)\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/833d3d83cc9748678aa3f238f5123258 : SUCCESS in 28m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c6e4a03b3de544739c7f0d603523dc7c : SUCCESS in 35m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/940371ffa68d471cb0ff28c41ab17d93 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/037f1946d70c478d9d6a09867575e1bf : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73c91bfdfb1247d1a396ca40586217d7 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ee9af16290b41b28ebb7eb841598198 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b165ab9dc1e491387ff8ae005c09411 : SUCCESS in 7m 24s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/4a1a32b71c844c3893250eb026150e3e : SUCCESS in 24m 04s","accounts_in_message":[],"_revision_number":2},{"id":"cb9965b0b67735e8ce89088e4bdea0e7c0a98553","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-24 03:41:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9ab131d1f5cb88b473028a02fbaec85ff7ce755b","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-10-24 16:13:26.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c8f2db7e6f9a7fff3b7bef88b222c23814f03b4b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-28 10:49:32.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"38545a36b422a92f86b7c68df280c22692710a81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 10:50:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7ea1e5f6b49912a9ed678eeedf80335110134d3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 11:51:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/673d734c5d3748a0bdd95f8839540707\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f003bc5736cd48d2a3ab3262b016d5b5 : SUCCESS in 33m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3134ad342b044bb39890aa427456aecc : SUCCESS in 37m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/23ab82e74f8647e5915f79dd0bf18be2 : SUCCESS in 33m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e5fd179d66924024bcadef31f8e07462 : SUCCESS in 36m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96c67544dc01467abf1c922c9ef47ba5 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f529faf921ec40b39e144092f7cc0581 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45f1063ca18c4846b395192ac7481354 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da4f4890f25b4977b7798f3e64cabf55 : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":2},{"id":"c8ff1a3c0ebd16b7113cf4d2d1646a0f3018c025","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 11:51:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7ec9984c5fdf704259dee57d39e25b58c27f98e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 11:54:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc8b028b5f394deab53bd127624070a1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94da955d490541599c9257d5567c08b8 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"59ef8361466817c9485a311080c8c2f88667bbc5","revisions":{"6f65a946e84abe72a28ab5d6edf16ae70690cfb8":{"kind":"REWORK","_number":1,"created":"2022-10-23 10:27:08.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/26/862426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/862426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/862426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/862426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/862426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/862426/1"}}},"commit":{"parents":[{"commit":"099795d2e1497c4043158ffe92fbaa9fbe76ddac","subject":"Update zuul queue configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/099795d2e1497c4043158ffe92fbaa9fbe76ddac"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-10-14 10:15:56.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-10-23 10:26:09.000000000","tz":120},"subject":"CI: Move queue configuration into project-template","message":"CI: Move queue configuration into project-template\n\nWith this it can be reused by the other designate repos and we can\nreduce the replication everywhere.\n\nConflicts:\n  .zuul.yaml\n\nChange-Id: I1fe73145e4a4699939244bc857600efa264f2340\n(cherry picked from commit c907cfa1e68249c21e8cad9f58ff3dc3d9cea6df)\n(cherry picked from commit 96d788c6733b3e3f78fab96ebc5e0aae2203842a)\n(cherry picked from commit 013ccb4bb1410987be7d2b917a31b6a0c7ff4ad4)\n(cherry picked from commit 6dff8c6fdb31b1aa74137e52f9fe4864173cdff3)\n(cherry picked from commit d3263165889f4c91d8edd7b83e47abf8f0f662f4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f65a946e84abe72a28ab5d6edf16ae70690cfb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f65a946e84abe72a28ab5d6edf16ae70690cfb8"}]},"branch":"refs/heads/stable/victoria"},"59ef8361466817c9485a311080c8c2f88667bbc5":{"kind":"REWORK","_number":2,"created":"2022-10-23 11:46:59.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/26/862426/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/862426/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/862426/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/862426/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/862426/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/862426/2"}}},"commit":{"parents":[{"commit":"099795d2e1497c4043158ffe92fbaa9fbe76ddac","subject":"Update zuul queue configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/099795d2e1497c4043158ffe92fbaa9fbe76ddac"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-10-14 10:15:56.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-10-23 11:45:57.000000000","tz":120},"subject":"CI: Move queue configuration into project-template","message":"CI: Move queue configuration into project-template\n\nWith this it can be reused by the other designate repos and we can\nreduce the replication everywhere.\n\nConflicts:\n  .zuul.yaml\n\nChange-Id: I1fe73145e4a4699939244bc857600efa264f2340\n(cherry picked from commit c907cfa1e68249c21e8cad9f58ff3dc3d9cea6df)\n(cherry picked from commit 96d788c6733b3e3f78fab96ebc5e0aae2203842a)\n(cherry picked from commit 5f921b78896e68e473ab0c139d534f02a4105840)\n(cherry picked from commit 019b1ae9cb54f06f7b754024687e9f0fdfad9534)\n(cherry picked from commit 12453a34b385646bfc0dd16cac7203c4c587798a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59ef8361466817c9485a311080c8c2f88667bbc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59ef8361466817c9485a311080c8c2f88667bbc5"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
