)]}'
{"id":"openstack%2Fdesignate~863252","triplet_id":"openstack%2Fdesignate~stable%2Fxena~I1c8dfec7bd2b5f80b12d7229e47ef8ff70bd283b","project":"openstack/designate","branch":"stable/xena","attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-02-20 23:38:00.000000000","reason":"\u003cGERRIT_ACCOUNT_5572\u003e replied on the change","reason_account":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"}}},"removed_from_attention_set":{"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2023-02-20 23:37:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1c8dfec7bd2b5f80b12d7229e47ef8ff70bd283b","subject":"Switch CentOS jobs over to CentOS 9 Stream","status":"MERGED","created":"2022-11-02 10:54:37.000000000","updated":"2023-02-20 23:39:23.000000000","submitted":"2023-02-20 23:37:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863252","meta_rev_id":"95224c521f74303d54edf399967e7ea9f8b43904","_number":863252,"virtual_id_number":863252,"owner":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-20 23:37:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2023-02-20 23:38:00.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2023-02-20 19:25:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-20 19:25:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2023-02-20 19:25:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-02 10:54:37.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2022-11-02 12:00:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-03 17:13:07.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-11-07 22:43:53.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2023-02-10 18:00:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"674a24fb1160da72cf627b191ee59dddbb12b25d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-11-02 10:54:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9e5b7d00068cb87457ccae2b1fe8dfb5fe0682a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 12:00:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6844f831b6846e2b85a050a0f56ecb2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9ea30d4573ef43c5b5a5323e224d5aa6 : SUCCESS in 39m 57s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4950410c8cf2402ca35242d4a58d22d0 : SUCCESS in 49m 38s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/9174d44f744644a3acf89b67a34a601e : SUCCESS in 39m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/261d6945b0df406c9df43e082d35f747 : SUCCESS in 39m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/aa6a9aaec5384c8d811d7b1a427bd299 : SUCCESS in 1h 03m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/94434ca1d742421eb03d9beefa90349e : SUCCESS in 44m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a648f1c2aad44e21ad92b4f5c1dce1d5 : SUCCESS in 49m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b52aedef41cf4d8dbb9ab5619a59a4ca : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2d136db25f544f1842483fc37ea69fc : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40b7d5a487b84aa6a8258f6f624cd0b5 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5658bbeb29324d0d85d929ae2c4e9ecf : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9107601702bf4b709e5742dd6987c1fa : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/619cffac3c6a4e6996ece6610ed0f8e5 : SUCCESS in 7m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/27a5c14b7a1749b586a8bfc4ac6c0000 : SUCCESS in 34m 06s","accounts_in_message":[],"_revision_number":1},{"id":"fe8db2c88fcda6b2c9cdea22ba60cb3dcdcb6c92","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-11-03 17:13:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"afb9c4d27b205689c73ee299d3b438c8d6a7682a","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-11-07 22:43:53.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ffa237e3a34dc18e4b4b6cf464f0c7fe867d432c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-10 18:00:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa44a01d25ce8534d4c20136b28824fbda1cc35b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-20 19:25:02.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"cbef997409ee4c33a974dedc3283a73b0a957e69","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-20 19:25:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f960e1491214b57557f7bca51eb9581349d6d00e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 20:36:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f5d270da36b479d969928fe932413d3\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a3eac9772eab476d9c8e2791674efb7f : SUCCESS in 44m 52s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3c283f49a5b44baca856da6faff905a1 : SUCCESS in 1h 04m 54s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/bdfde58dfd504a06815081e9f062934d : SUCCESS in 38m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/21b395c74cfd48ada2401249830ed541 : SUCCESS in 37m 23s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c6229b74a9e84678b5963a215a247f53 : FAILURE in 1h 02m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d4770606c3e847e7ac36b59464655df5 : SUCCESS in 38m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bef129466ccb43b78b4812a18990956c : SUCCESS in 48m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/773758555699479abb8c9371c4147890 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f081f83e66ad4659b748dd5a3927613a : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/982e1b3a6ee444b7884e22ab0daa68ed : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e33b9494d7eb46a4aff196ff1a05458b : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95623a640ba641738580004c06c65a74 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f08c6332d5774313af51eae3b05a2dea : SUCCESS in 6m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5fb3ade7f30c469daabf756cac3cf8cf : SUCCESS in 33m 25s","accounts_in_message":[],"_revision_number":2},{"id":"d03bf46369aa22a29a54189dc746de6de5aa4c00","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-20 21:22:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"63147c815ce3e61c755c12694c9e01baf1ef5a81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 22:30:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cbe20036b05493690cc6ef19b42f7db\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fa61c5ef2be643aeb472a7fb28e502ca : SUCCESS in 39m 54s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3f9438f7a6a54d17b71ab8033c26b9d8 : SUCCESS in 1h 01m 40s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4c9f29d2281847c2b40c0698a563ae99 : SUCCESS in 35m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bb27d0d050124722962dcebb41696ee4 : SUCCESS in 34m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4f34b66e583c480ea367c9824ff09e68 : SUCCESS in 1h 01m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/74eea8afb2004df28514c65ecc341a30 : SUCCESS in 40m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3f0d5c86cb854b458662745d1bdca565 : SUCCESS in 40m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09df95f91cea43dc9b4f8ce44e6aa62a : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cd3b6e60c984a5fa74b64bff33a611f : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96807097b18f4d5c897990c1c74d5b2b : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19044399576046fbb7d18a2b5c9c9718 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5496ef024eda4d5d8b48a74af255550f : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3aa40b37f5345a2a0224d3ce0354390 : SUCCESS in 6m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/caeb02090ec14cf8a93af7e81080b069 : SUCCESS in 33m 43s","accounts_in_message":[],"_revision_number":2},{"id":"09462783becd172ac733dae109f80a719937c361","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 22:31:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"754c66eac598f5b80650455037173f2e37f19683","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 23:37:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80f56d2ec98a47409dbf530319d6d658\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/498f47c73da341c49dfd6855b3435072 : SUCCESS in 38m 37s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/0a344e190b56478bb68def46da76fa35 : SUCCESS in 36m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/13b00f6922f34034a0b18af6107af7c9 : SUCCESS in 34m 53s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/590a5795682e4b31b61317b420da4076 : SUCCESS in 1h 05m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8408ba4734da48c393ecca2aca39a63e : SUCCESS in 36m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2edcef34078647abaa3da4fb2f83f00a : SUCCESS in 42m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08d91c60873a427e804a5319f3823276 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/080222e9a2ca4297bc900394f38f9e63 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fe55ea8da1c4892a598b66d1d5182b8 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddcc595118e24c738660cce518a1421a : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":2},{"id":"b327863052f09cb979eed8a27941db0e307d711f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 23:37:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"83c4327d82c75446cdb989bfc629e9faa1b8e950","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-02-20 23:38:00.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"95224c521f74303d54edf399967e7ea9f8b43904","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 23:39:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90a0cf8e84554723a1c48ebc94159eb6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf8376f8d02b40f2abd0725398ccce93 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"970175de37e7d14835c1027f9c889142ba3df7bc","revisions":{"554b2e7e28a9fd0fff6314157dc401f67bfa5be6":{"kind":"REWORK","_number":1,"created":"2022-11-02 10:54:37.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/52/863252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/863252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/863252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/863252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/863252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/863252/1"}}},"commit":{"parents":[{"commit":"019b1ae9cb54f06f7b754024687e9f0fdfad9534","subject":"CI: Move queue configuration into project-template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/019b1ae9cb54f06f7b754024687e9f0fdfad9534"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-26 23:09:45.000000000","tz":0},"committer":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-11-02 10:57:27.000000000","tz":0},"subject":"Switch CentOS jobs over to CentOS 9 Stream","message":"Switch CentOS jobs over to CentOS 9 Stream\n\nSince Devstack is dropping CentOS 8 Stream[1], we need to move our jobs over to CentOS 9 Stream.\nNow that there is a \"devstack-single-node-centos-9-stream\" nodeset we can remove our local\nCentOS 9 Stream nodeset.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2022-April/028321.html\n\nChange-Id: I1c8dfec7bd2b5f80b12d7229e47ef8ff70bd283b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/554b2e7e28a9fd0fff6314157dc401f67bfa5be6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/554b2e7e28a9fd0fff6314157dc401f67bfa5be6"}]},"branch":"refs/heads/stable/xena"},"970175de37e7d14835c1027f9c889142ba3df7bc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-20 19:25:02.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/52/863252/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/863252/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/863252/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/863252/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/863252/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/863252/2"}}},"commit":{"parents":[{"commit":"2b94cf3ca15c6521386978dfd6504d614e880e1f","subject":"Fix dns.resolver import for mdns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2b94cf3ca15c6521386978dfd6504d614e880e1f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-26 23:09:45.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-02-20 19:25:02.000000000","tz":0},"subject":"Switch CentOS jobs over to CentOS 9 Stream","message":"Switch CentOS jobs over to CentOS 9 Stream\n\nSince Devstack is dropping CentOS 8 Stream[1], we need to move our jobs over to CentOS 9 Stream.\nNow that there is a \"devstack-single-node-centos-9-stream\" nodeset we can remove our local\nCentOS 9 Stream nodeset.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2022-April/028321.html\n\nChange-Id: I1c8dfec7bd2b5f80b12d7229e47ef8ff70bd283b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/970175de37e7d14835c1027f9c889142ba3df7bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/970175de37e7d14835c1027f9c889142ba3df7bc"}]},"branch":"refs/heads/stable/xena","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
