)]}'
{"id":"openstack%2Fdesignate~864058","triplet_id":"openstack%2Fdesignate~stable%2Fvictoria~I052fb7ad90b4134ebe4c4e6a4784064c176eaaee","project":"openstack/designate","branch":"stable/victoria","topic":"bug/1940563","attention_set":{},"removed_from_attention_set":{"32927":{"account":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"last_update":"2022-11-28 21:20:32.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I052fb7ad90b4134ebe4c4e6a4784064c176eaaee","subject":"Specify AXFR transfer timeout for dnspython v1.","status":"ABANDONED","created":"2022-11-08 19:26:17.000000000","updated":"2022-11-28 21:23:13.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"33552763dcbb989c927a0f46026ebb304fd11ee0","_number":864058,"virtual_id_number":864058,"owner":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-11-09 13:26:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":-1,"date":"2022-11-22 18:23:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-08 20:10:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-22 18:23:42.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"b01c86310370211a7b77e4fadb1d6313a0f87acb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-11-08 19:26:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98d637a3e1fa947be5fe33f2cb10d0c913e0a0e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 20:10:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c80d998b0b4b417a9e17280221132742\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8701e7666c8c430898c637c5b42e6d8b : SUCCESS in 31m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/32d62ddbfbfa45f4b7790a0846883dde : SUCCESS in 30m 38s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4f36d12786654b4e81b82a0d5858d8d1 : SUCCESS in 42m 28s (non-voting)\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6145015a89e64fb2a8d932a8dda4f616 : SUCCESS in 31m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6ff0741427d84cf3a41a3db89236ca87 : SUCCESS in 28m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8482b4d8fd2b447dbcd643ef0dd75a2b : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/240a1e32355a476f929ed2077fa59312 : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/060fb61ebe8549e98e107d73bfbad98a : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be26b32cc29c470597d6d2a7690c0434 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5db1aa78e1c347e8a30c43a0168d520b : SUCCESS in 6m 51s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/6d8de1c8f8ce4f7aae8284acc019851f : SUCCESS in 22m 17s","accounts_in_message":[],"_revision_number":1},{"id":"81673b3d272ed6a08ec3d5e7197e418d38a40939","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-11-09 12:13:41.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"17490e845a937e1822dd5097cf16898b61f61ce7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-11-09 12:14:11.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"db6132a602ea6a092e86ada250a025790d868dd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 13:26:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7c06fbb80ff440e8054ce0ebd6e7a5f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fdb723432e4749a0aa91b17271bff5f6 : SUCCESS in 42m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ec9abab875544f2580eb01fb4a5afbca : SUCCESS in 40m 53s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/75fd11aab44347559ea0afb40c239ba8 : SUCCESS in 1h 07m 19s (non-voting)\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2428b3fa672a45589cd44f8959c7f6a1 : SUCCESS in 40m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d304abe114f44976a315388abdfc6c9c : SUCCESS in 40m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfb923f92f894adeb7647441a6291786 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e5fb6bf8c5f41c08e110ae35d23b1b9 : SUCCESS in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/089696ff204e44228ebc0da1dd8edd30 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/702f8ad03c964502834c8adc59a05a4b : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8560382781a3414db751b0200f9f7634 : SUCCESS in 6m 26s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/6c5f8a0e239d402e90b9b12955dc9c29 : SUCCESS in 23m 28s","accounts_in_message":[],"_revision_number":3},{"id":"24d927a6c77d7fb9de8970ed9bda31e28cc6102f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-22 18:23:42.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"fe416d66d2da460b067ee95253412a93663bb916","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-11-28 21:20:32.000000000","message":"Abandoned\n\nplease review https://review.opendev.org/c/openstack/designate/+/865843","accounts_in_message":[],"_revision_number":3},{"id":"33552763dcbb989c927a0f46026ebb304fd11ee0","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-11-28 21:23:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dda3c31465683f5f64403b75d7d599db17af6e93","revisions":{"48d2454c00bfc08a5cdb65121497f09886f8de08":{"kind":"REWORK","_number":1,"created":"2022-11-08 19:26:17.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/58/864058/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/58/864058/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/58/864058/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/58/864058/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/58/864058/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/58/864058/1"}}},"commit":{"parents":[{"commit":"59ef8361466817c9485a311080c8c2f88667bbc5","subject":"CI: Move queue configuration into project-template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59ef8361466817c9485a311080c8c2f88667bbc5"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-11-08 19:23:28.000000000","tz":180},"committer":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-11-08 19:23:28.000000000","tz":180},"subject":"Specify AXFR transfer timeout for dnspython v1.","message":"Specify AXFR transfer timeout for dnspython v1.\n\nIf we specify message timeout and do not specify AXFR transfer timeout\nwe face with error in dnspython v1:\nhttps://github.com/rthalley/dnspython/issues/390\nThe bug was fixed in v2:\nhttps://github.com/rthalley/dnspython/commit/9fbf9b223dc26262b1cf23f623f89283ee9c8375\nIt\u0027s risky to update dnspython to version 2 because there are changes in interfaces\nand eventlet requires updates too to be compatible with dnspython.\nSo it\u0027s simpler to set transfer timeout.\nBy default there is no timeout at all, so we set infinity.\n\nCloses-Bug: 1940563\nChange-Id: I052fb7ad90b4134ebe4c4e6a4784064c176eaaee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/48d2454c00bfc08a5cdb65121497f09886f8de08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/48d2454c00bfc08a5cdb65121497f09886f8de08"}]},"branch":"refs/heads/stable/victoria"},"ffd6fbf393685bd70c67ea42097c185a2811ee92":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-11-09 12:13:41.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/58/864058/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/58/864058/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/58/864058/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/58/864058/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/58/864058/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/58/864058/2"}}},"commit":{"parents":[{"commit":"59ef8361466817c9485a311080c8c2f88667bbc5","subject":"CI: Move queue configuration into project-template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59ef8361466817c9485a311080c8c2f88667bbc5"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-11-08 19:23:28.000000000","tz":180},"committer":{"name":"mitya-eremeev-2","email":"mitossvyaz@mail.ru","date":"2022-11-09 12:13:41.000000000","tz":0},"subject":"Specify AXFR transfer timeout for dnspython v1.","message":"Specify AXFR transfer timeout for dnspython v1.\n\nIf we specify message timeout and do not specify AXFR transfer timeout\nwe face with error in dnspython v1:\nhttps://github.com/rthalley/dnspython/issues/390\nThe bug was fixed in v2:\nhttps://github.com/rthalley/dnspython/commit/9fbf9b223dc26262b1cf23f623f89283ee9c8375\nIt\u0027s risky to update dnspython to version 2 because there are changes in interfaces\nand eventlet requires updates too to be compatible with dnspython.\nSo it\u0027s simpler to set transfer timeout.\n\nCloses-Bug: 1940563\nChange-Id: I052fb7ad90b4134ebe4c4e6a4784064c176eaaee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffd6fbf393685bd70c67ea42097c185a2811ee92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffd6fbf393685bd70c67ea42097c185a2811ee92"}]},"branch":"refs/heads/stable/victoria","description":"Edit commit message"},"dda3c31465683f5f64403b75d7d599db17af6e93":{"kind":"REWORK","_number":3,"created":"2022-11-09 12:14:11.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/58/864058/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/58/864058/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/58/864058/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/58/864058/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/58/864058/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/58/864058/3"}}},"commit":{"parents":[{"commit":"59ef8361466817c9485a311080c8c2f88667bbc5","subject":"CI: Move queue configuration into project-template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59ef8361466817c9485a311080c8c2f88667bbc5"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-11-08 19:23:28.000000000","tz":180},"committer":{"name":"mitya-eremeev-2","email":"mitossvyaz@mail.ru","date":"2022-11-09 12:14:11.000000000","tz":0},"subject":"Specify AXFR transfer timeout for dnspython v1.","message":"Specify AXFR transfer timeout for dnspython v1.\n\nIf we specify message timeout and do not specify AXFR transfer timeout\nwe face with error in dnspython v1:\nhttps://github.com/rthalley/dnspython/issues/390\nThe bug was fixed in v2:\nhttps://github.com/rthalley/dnspython/commit/9fbf9b223dc26262b1cf23f623f89283ee9c8375\nIt\u0027s risky to update dnspython to version 2 because there are changes in interfaces\nand eventlet requires updates too to be compatible with dnspython.\nSo it\u0027s simpler to set transfer timeout.\n\nCloses-Bug: 1940563\nChange-Id: I052fb7ad90b4134ebe4c4e6a4784064c176eaaee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dda3c31465683f5f64403b75d7d599db17af6e93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dda3c31465683f5f64403b75d7d599db17af6e93"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
