)]}'
{"id":"openstack%2Fdesignate~864816","triplet_id":"openstack%2Fdesignate~master~I5b9cfb4b86bd7c342fd50d1b50dd12dce0c8e81a","project":"openstack/designate","branch":"master","topic":"sqlalchemy2","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-03-11 05:00:05.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-03-20 21:14:24.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I5b9cfb4b86bd7c342fd50d1b50dd12dce0c8e81a","subject":"Update for sqlalchemy v2.x compatibility","status":"MERGED","created":"2022-11-21 02:27:16.000000000","updated":"2023-03-20 22:58:53.000000000","submitted":"2023-03-20 22:57:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864816-sqlalchemy2","meta_rev_id":"f458d214288ab96d771a123c3a4c890f826fcaca","_number":864816,"virtual_id_number":864816,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-20 22:57:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-11 05:06:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-03-20 19:38:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-11 05:06:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-21 03:44:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-11 05:06:40.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"7652ceb3dc99a720155c4067108b1b3a5011ecb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-11-21 02:27:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"940e25b4b2f9d0bda638b272ca461e8bd398e1ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 03:44:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce68716dbcda4d078d65622ce315e6bd\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3408d081d39f47698da06a6bb71d8f76 : SUCCESS in 45m 46s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b5009d5e28d94124a3fd2bcc5bd542e1 : SUCCESS in 1h 01m 15s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/801468fa6e0a4ff2a8d62235891fec2e : SUCCESS in 47m 11s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/94f3d3b74da34af4bd85f37fa85d7971 : SUCCESS in 41m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/39991c34c4ff4b3dbc80a01849737f98 : SUCCESS in 44m 34s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5d90f2ae5b324a22b1278fef3a1f374b : SUCCESS in 1h 05m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b90b9a713e7142a6b3b66d132b91a62b : SUCCESS in 58m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a0d9356e18434d538eee9a091382e7f7 : SUCCESS in 39m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7f6e77c192b44db3a7821735941b2bb0 : SUCCESS in 37m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a76cc4af3350460a99a53d6a95d8d8a3 : FAILURE in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/746db19557c946028bf0f5d21b6619a0 : FAILURE in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2eff7af78ee0407fa0cbef580a2c8d82 : FAILURE in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b40b693ef5164afd8345f8b6dcdd0f8d : FAILURE in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a745a457a1d7462cb82a48c31ceabe90 : SUCCESS in 8m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a5ce4d0e48b54814a896f394c61314e0 : SUCCESS in 2m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cef038d0420946cdbffdee1e9d8c9484 : SUCCESS in 26m 05s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/89e2489906dd43c191aab124dba36894 : FAILURE in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"536d8ebbcd8563cfe2674228297cf36d1c21029f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-11-21 04:16:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dd165c1a114cb5e42fac584884df6f40df554b73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-11-21 04:23:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2bfcb919c9a5048657ed395ac4777320680c2c49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-11-21 04:23:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"37ae4fdedc30b4dbbd0f9c04b7f37506505edd1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-11-21 04:31:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6ebc844a8cbefd0589b78d54273dfc1bf84c1dbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 05:33:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e59bc5d4d4a4b1b927982f8c756922d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/98900b1a700f473e86a4f3a012395ab4 : SUCCESS in 36m 45s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5d1a8705531b4845b710ce0e319f1b83 : SUCCESS in 42m 26s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/473f267415284252b45efd929c48e356 : SUCCESS in 42m 14s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/15798d53765c4df6a23b1d5f07387f76 : SUCCESS in 37m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5c24c38e0c214526903488abfa8054ab : SUCCESS in 30m 03s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3790dcfa8b0b4ac0a3c2e3963df9ffb6 : SUCCESS in 1h 00m 33s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e92323888d604b6086ced7edca84035e : SUCCESS in 1h 00m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5ac9b59ced034a6aa7587928636e717e : SUCCESS in 36m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2bd559d8eeb548d180066a08e2bfa4fb : SUCCESS in 39m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f9d4054b27348deb1e26fceea6a59d7 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9809d3eeac6e418c9c57c3c3f0d3b9e4 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/320a59b2687f429c933f4d562e1d7d2d : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59b936d2426746b78e3c2a39b0851ad1 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70176a4ea5c44e6d9a60e373e6839dd1 : SUCCESS in 8m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/58f0776652a14eee8d07eeb897fcaeb7 : SUCCESS in 1m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c38f99040658488fa8252d6e59640ca8 : SUCCESS in 26m 44s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/36297c7819c94b2182b8b3d1358dcff7 : SUCCESS in 4m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b7ef2b04b25dbc91c4049e8ddd64bfab14bc0172","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-11-21 08:44:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"18aee321c1f4a46a932d64ac38195cdbb91edded","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-11-21 08:56:40.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6fc0e2177d0afabb848ef8257313cb9896ed2412","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 09:55:48.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1db18e6a744c4c06a25e70fa8a4e4601\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/121d0d1e4d984abe893d9b0988132c17 : SUCCESS in 35m 51s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ec834b9f918c41f7ae31b8783e939077 : SUCCESS in 58m 16s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/00b8e8e295ca44cba4187fed0da59e4d : SUCCESS in 39m 30s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3a111195b5bb4aecba68865dbd2fea46 : SUCCESS in 37m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1a373b7194ca4d63b7631f46014f3144 : SUCCESS in 34m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/817a6a6fe52b4b3983b1d551e1ce620f : SUCCESS in 55m 58s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/33a8095327a3437a9195c96a4b633898 : FAILURE in 11m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/58da9d610f0b46a4a4de355f3e13dc95 : FAILURE in 36m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e0796371cae947f2a6ffe1720b2f8c32 : SUCCESS in 37m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7026206a2bb74fddb75b53b7f76cc692 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7bc2dc7020045b0ba7b3f055e0f9b24 : FAILURE in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb8f14479be949b684b231da704dc4df : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a6f35c8c1b3456ba9719082ec6a7a98 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaa4795bcda641dc995a0667bda933b1 : SUCCESS in 8m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c00be467ae914fcea5933725d3820d1e : SUCCESS in 2m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b23d8f47e4814d80a089bffbcd8912b7 : SUCCESS in 25m 45s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3281736edc1b49dab21674dff6bc1d77 : SUCCESS in 5m 07s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6b811812f53a10cd99d3f5be5411eb20bf0af316","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-11-21 19:46:33.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8f521edd6da23104886df7b8c5e4f54be592753e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 20:49:33.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d6c895b429948799990b5908d09a40e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f3cbbc866eb24afebd9bae828a98ca7d : SUCCESS in 32m 42s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9d48e5d1387e4f32bac31d6eed292fa0 : SUCCESS in 44m 52s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd08bc3d6ade4292ae190fa0496b7efa : SUCCESS in 40m 57s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3a1b431722f24aedbde794f71a3b18ac : SUCCESS in 36m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/69e114b4d11f4ed4b7b2af111c3caa6f : SUCCESS in 37m 59s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/bb644e7ef0004d26bb01aed4a807bfbc : SUCCESS in 55m 01s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4344014c4d6b49518d5877a353e35d36 : SUCCESS in 49m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a18a7452f5a54d429dc141b1738d0a73 : SUCCESS in 36m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/09b485c50da443f391040f32d606999b : SUCCESS in 38m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa21c2a1794943758776fdd225356f1d : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f27bb216b75a4740ba1849c1b28e2b66 : FAILURE in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8cd9915d0b6488f87bdc32a04a9b63c : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/712f004739ee4889a8d1007fd1c8b44d : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/645491ee4b1f4a2da0a3a3a24990253f : SUCCESS in 6m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d956708c29e44bf78214b1cc0d48b721 : SUCCESS in 1m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a448237fa5e749378e83a07d27cdf70e : SUCCESS in 31m 00s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/47b503bfa6db47d3a511f25f0b59fc24 : SUCCESS in 4m 18s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ad06d96cf9fa07f4d4d2c652fe6b33b3dba0a037","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-01-01 21:15:59.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"52ee77928f0b76a4a077c2f532deef0ea703eed4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-01 22:17:39.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bce4d8c6e8ab4f13b2351f281673e219\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9b0b2d3db4e942c5be235254a9fe3ccc : SUCCESS in 30m 04s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e21218e62c164bd38a329e8e5c543cc9 : SUCCESS in 51m 12s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5e3cf50e6ca491b87b6eecad47b0110 : SUCCESS in 33m 23s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/df6bc29bd1f341a6a1d5146736ec3d22 : SUCCESS in 29m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/121755474fa54d8c83596ea7a8710578 : SUCCESS in 32m 09s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e5c66a6b2a264b648b037c3c95f21600 : SUCCESS in 49m 53s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d30433109f4e4bac94a0b99c8e5fadc8 : SUCCESS in 46m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8de6460fc5df40fabed25fb119e2c80e : SUCCESS in 32m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/15b1513a6fef4d39926d16a9f0971a17 : SUCCESS in 30m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2597a79f615149a6982b9e122217db49 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be30ed0f3a024e139359de053b1b5201 : FAILURE in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6815cbf92d84c22af440280b8da6d10 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b804798ce8c4741a031be88faa99162 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61a5ce3aae7940adb6f8e522e4ef2b99 : SUCCESS in 6m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2f840b7fec4149bda334640df89c6551 : SUCCESS in 1m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9981bfc3709b4efd899d52b7a3cbf473 : SUCCESS in 29m 29s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/88fc2da688644e038ab16e8afbf7bf50 : FAILURE in 3m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"389f2747a9c846743b28cbd910d755cfc42f79c8","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-01-04 13:49:13.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":7},{"id":"a3fafdb56aff380c9f3fcda33e4ec152b6b78859","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-05 22:52:56.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"bdd78321ddc0a5fdc2d951e96dfb7ecda10120c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 00:08:33.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25dc99be16a3451797b815b150d3bfb3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/482db9263c47411690b347c8e7bf2f92 : SUCCESS in 4m 01s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4400465b31d24a3c84f2f6f6437e62ec : SUCCESS in 45m 45s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2bcae0d5955c475eab6abf48ce172ad3 : SUCCESS in 1h 04m 53s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a49ac3c46464ed0b923b9bbcfebb914 : SUCCESS in 55m 46s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/825cd2942db44147b41ee7ce5865ed61 : SUCCESS in 52m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a0d78137ecdc426e94e88202d2e65f6e : SUCCESS in 47m 25s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/1c9e679be48c4d19b31ee1f12b526678 : SUCCESS in 1h 14m 55s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ccce506b2cca4dd1b387af8206aa97ad : SUCCESS in 1h 14m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a72daa77b58a4cb8a3d3dd694e2a3475 : SUCCESS in 44m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6b8ece16281243738ed264d8080d1648 : SUCCESS in 45m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/435d84181cfd4918bd3493cb60e68ca3 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0e8a94142f44bf1ae3fdad500d3d0ca : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/248b2fd2df4347e2afd92c305d5e42e1 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cfb28171ee6415aa47c516b3bd0aef1 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4c1fa3490814fce94468f65ef83ca08 : SUCCESS in 8m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ec7bb9a394f34510bc7ac146c3860c9e : SUCCESS in 2m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e74bf93a5f444b28fa004b2e5db0227 : SUCCESS in 4m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d2268098191446afa2f68782cfa4ccf8 : SUCCESS in 30m 38s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1c29928ce8084b8181de939162bdb572 : SUCCESS in 4m 39s (non-voting)\n- designate-tox-sqlalchemy-latest https://zuul.opendev.org/t/openstack/build/dbda483391e34693926827d2f0ef5e9f : SUCCESS in 6m 22s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"5bb3282a8a332e33e6eee9b43308e1166ea520a6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-06 17:06:44.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"5ce41d482f77c8bdeb2ed3d082035881e4726eb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 18:19:58.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a2374eee6f04fa282fb19d313146509\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b59f9303bc034ffd84c0bc4c88c23b37 : SUCCESS in 3m 41s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7e8bd561bdee43df97f5a090ccadce8a : SUCCESS in 33m 45s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0ba9edd73a0a41b9a0d1edfb36318ce6 : SUCCESS in 54m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c5fcba16ce748a2a6c7ad6ef3ac16c0 : SUCCESS in 42m 46s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/9529b0c7087e4c30bca2e5a4e2c4962d : SUCCESS in 37m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/093e0551fb4f426bb4b5103dd9395e8d : SUCCESS in 44m 37s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/fa2d16b4b3434d779d607cafaea7ffa0 : SUCCESS in 49m 07s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d68b5c91b6624fc7a7aad54f7fd44e40 : SUCCESS in 1h 12m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/81529285cb4b4cacb8b85323659f20bc : SUCCESS in 30m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/74e83b2927b54afa882a18e22c0fe510 : SUCCESS in 43m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2aa8078199ed48eda5b99a8bf4ebfc6e : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5c39819501c4cccb840df29c58f73a6 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e774fa1fe69343e0ba69b524e056f9f0 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09abc12c891a4b3a92194ab32cfc2db4 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b085b7fb35cd4d449a4eadc6ec74a5c9 : SUCCESS in 7m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a6a16e3d4bda4dafb5519911f1607040 : SUCCESS in 1m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6885f36fed8499fafee474b138d91d2 : SUCCESS in 3m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fb3e3d45a1a540a29e2765d0bb5fb3c1 : SUCCESS in 32m 47s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/fb4f809dd90149f8a07c1c6e2436e8de : SUCCESS in 4m 13s (non-voting)\n- designate-tox-sqlalchemy-latest https://zuul.opendev.org/t/openstack/build/deeff6b36b6c456b8ddd8a5c2a57dacd : SUCCESS in 4m 22s\n- designate-bind9-sqlalchemy-latest https://zuul.opendev.org/t/openstack/build/c8d54551a93e4bbea0c08519d9da6b61 : FAILURE in 29m 09s","accounts_in_message":[],"_revision_number":9},{"id":"f85d678c1bbecc0df8a67de28ea0dfe8b05eadfd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-06 18:22:40.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"29a9a781a2a1cd86e812bf6fa211f2fcfa70dc2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 19:36:06.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/de53583db6d1405db2e086b4cf8504c6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/79310a270a724562b193f2faf73b517c : SUCCESS in 4m 07s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b11aaeedc8cc496a81ed3193568ed73f : SUCCESS in 44m 35s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6edc385761c54f03a984f6310b379efa : SUCCESS in 57m 14s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0ec7e5f00bb4b5e92694c37a202a2d9 : SUCCESS in 42m 48s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f85c3791736542af8d4ceff21d1eda0a : SUCCESS in 41m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4d7478eb86d1448999e6f8ac2d83d119 : SUCCESS in 37m 42s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/434451a0d30f42ef8cac944d601157cc : SUCCESS in 1h 06m 09s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7ad58b4f60dd4e1eb1b84bed5655b759 : SUCCESS in 53m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/507ab6c8f3c74716ac4a395151fae11a : SUCCESS in 48m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/857e4e60b9844697800970a882c3a868 : SUCCESS in 35m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5968d5cd8db04affbd92ebd9c4af0d17 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a89ee5d7edd1494e8647e779d5d4fcb9 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7efa947b4c5a4630979abfce4e7bc88b : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d74fc36038c4886a711d190e8c62650 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/171efde62b36429fb85869de85772555 : SUCCESS in 7m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6237856261e14103a7de669aa7fb5d4a : SUCCESS in 1m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9990693634ab4f4abd6467a88598d3e7 : SUCCESS in 5m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f938b66df9014cb299d37b2ee891b422 : SUCCESS in 27m 05s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/57307fd59e044c59bdd0d9394e088a14 : SUCCESS in 6m 20s (non-voting)\n- designate-tox-sqlalchemy-latest https://zuul.opendev.org/t/openstack/build/6cfaa0c58e694f109d1eb828ec67b8d7 : SUCCESS in 5m 02s\n- designate-bind9-sqlalchemy-latest https://zuul.opendev.org/t/openstack/build/4ff4127139534246b1f65a473b91ace0 : FAILURE in 32m 57s","accounts_in_message":[],"_revision_number":10},{"id":"81e66ba4e3ebae7a66d736dc54cf74e22dddd97d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-06 21:15:09.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"c994f2dc2860dca72e1522b579aa80c55f0c965b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 22:27:11.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ea617ebf8ca479c8aa1ac1e1fa84337\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c3fff2e6fb5b47879f6caac3e88d6100 : SUCCESS in 3m 22s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/84dac2ce19864dce87a052eb0072b8f9 : SUCCESS in 44m 55s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8773cd871667401ba1f7e02ec3cbf754 : SUCCESS in 1h 06m 46s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb97bf947741491fbe4ecea411c66360 : SUCCESS in 50m 20s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ea966e600a91464a901f76267193814b : SUCCESS in 56m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/05cedf0ef3c44cf68bcfbee42857248c : SUCCESS in 46m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ad8ac6294d7e49ec9825d6c2910838b2 : SUCCESS in 1h 04m 59s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/80747eced2f6435688999a38d015f8b3 : SUCCESS in 1h 04m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/762541f6958a41fe91200fba527e1649 : SUCCESS in 44m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6415a20eabe145b88016c56556bb9905 : SUCCESS in 45m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d95669862fd43658320b471978012c7 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a615ac8baa1f4199a07727b0388175b5 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1fe1f34811b45aab5cce3f91917e974 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2eea8f99dd9b4a5db7406f346b35246d : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b207c64c0f441c48265729a67711be9 : SUCCESS in 7m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5fd7836c49a140a29531cb049b75a5db : SUCCESS in 1m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3aee5e603a314269be7bfeffc09275f7 : SUCCESS in 4m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c573c8d5b5a84d95aa3aa66515960bfc : SUCCESS in 29m 01s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/cf4ae41c14e944db84985ae5bd5fc29a : SUCCESS in 4m 12s (non-voting)\n- designate-tox-sqlalchemy-latest https://zuul.opendev.org/t/openstack/build/75a92cb3065b4cd4834ceec3aa02c638 : SUCCESS in 4m 44s\n- designate-bind9-sqlalchemy-latest https://zuul.opendev.org/t/openstack/build/649e283b739f44d9a34836ca983d250d : FAILURE in 37m 27s","accounts_in_message":[],"_revision_number":11},{"id":"ee70e78a78eb65b5ceea23499313ff28cd753de7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-06 23:01:31.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"2e28101d469a3f5edecfab6cd503e6278a9cc9ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 00:10:06.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d16e4fcbf7449a9a4bbc763581bb2a6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ad68327732654a8c9690158dbffb466c : SUCCESS in 3m 42s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8fca844791c243499cf3bf2ec0efbef7 : SUCCESS in 47m 45s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/57c266d6ec0a4e2fa9782e3480b4310b : SUCCESS in 56m 45s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f0ea70340664c7e96a9c4ef2d020616 : SUCCESS in 37m 51s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/5d930425bcf344db87a81801b6a1873c : SUCCESS in 49m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d86122157aa241a0bb5ee2ccd4d9bd83 : SUCCESS in 33m 00s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b65166b71fb447789f5a8358cd545bae : SUCCESS in 1h 01m 16s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/08ddc2a811c9446a96d34be0d6782a17 : SUCCESS in 59m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/314da0d09ba84cccb2a6b3a1a25cdd00 : SUCCESS in 30m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4695306046834d8cb22acb1994424187 : SUCCESS in 35m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/169ec10dee724d96bdddfc0acb2a46dd : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c319e662a89d40fb90d0fd60af137e2e : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07215802782448e4bbab7e006be0987d : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53be7f9c61804466bba05b0431087b68 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7e3df530d5148fdbf7bed7eacccfd03 : SUCCESS in 7m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f57f4f2e00a54df6b65ac88c992ac13d : SUCCESS in 1m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93e6b905ebed42d4b59aeac5f8979cd2 : SUCCESS in 4m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ba05077f78634ccaa1dc94b2ff0c5872 : SUCCESS in 27m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/22a063e46d6349efbf1d33d0bfa1466a : SUCCESS in 4m 20s (non-voting)\n- designate-tox-sqlalchemy-latest https://zuul.opendev.org/t/openstack/build/d5b4a9061054477c90291ff99eb3f2a2 : SUCCESS in 4m 56s\n- designate-bind9-sqlalchemy-latest https://zuul.opendev.org/t/openstack/build/ce4de38a467f43bbb9dedb9f4c505931 : SUCCESS in 44m 45s","accounts_in_message":[],"_revision_number":12},{"id":"3041c9f791bb65a909106c2f0d08f9e5369bf8da","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-09 00:11:40.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":12},{"id":"e0eeed4b886318fbf7d3b10b72a560a660c7050f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-10 22:53:27.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"5da0de5ec19aec688479bd91d1d365194697e852","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-10 23:02:21.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"3367020c7f78a290c907d5a8b9b1c0bb1936cb07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 00:13:43.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c86677aaa2e41dc99b70ccebc367c33\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/69bf21503a5c43398707e8943875e790 : SUCCESS in 34m 11s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7f846af76ab240bc98782f868f26b01b : SUCCESS in 1h 08m 18s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd7569a2fdc5446d9561930e79e4f0af : SUCCESS in 34m 21s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8cde3572b4e148f5941dc45b58b2ccc9 : SUCCESS in 45m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/95421d009512403ea0359414a552390c : SUCCESS in 33m 47s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ef74348844f349e582de7db4a6cb53e5 : SUCCESS in 1h 10m 03s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/48e91527e2cd4bff96952b4c7f3fa9e6 : SUCCESS in 1h 11m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e760044dc36944ba914009abce76caa5 : SUCCESS in 30m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/72f8257c99634df6b36383381822a2ba : SUCCESS in 46m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f37e2aee85664d68b7df7b2d64f741ec : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6a451fb5e13467dbe6581c6bc256630 : FAILURE in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdd95f0bdd5e4c3cae1e79e665ab18c8 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca1242eaacb24ff5add9d2bc6f0c3f20 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac14ebbc76984b2c9f577a80e8ece3d1 : SUCCESS in 7m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/90da4dbf71644a778946896f0f017352 : SUCCESS in 1m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/567a52233134416abdfb2e324406bbf5 : SUCCESS in 4m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d27b2f8f3baf45ce9c91b2b5fd68e1e3 : SUCCESS in 25m 16s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/39c804aae56d4fe3b5f0197127ee7b98 : SUCCESS in 4m 08s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"ae9cc0bb0bc07375d898e3a65d3ff46deb2a2fc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-11 03:50:19.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"a6774ceea174026449e5b43f574fad459e9883f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 05:00:05.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb717e6bd141404dad6ab4da75b27ebd\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/53139c54c6f2464fb6b408c9e89c0d12 : SUCCESS in 41m 12s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/05990f1880934b7e80cc3f064b329ad4 : SUCCESS in 1h 03m 20s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f71eca5cc5db46aaa4eff75a2c0219a4 : SUCCESS in 51m 55s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/7f85f072ba7a44c587982309d8b069b0 : SUCCESS in 40m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e717072bdeb24caaaf60e2902acaa236 : SUCCESS in 31m 48s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c834b369d24c4f918ea62b6a13a53251 : SUCCESS in 46m 07s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/02fd7a52254e427ab4da311419605622 : SUCCESS in 47m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/da99390d29a54f1da40a9816481dbd49 : SUCCESS in 49m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/46628b330e5345799a8451b274b82d9a : SUCCESS in 31m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a174aaea0134b51b24e2b6befbab074 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5b76e34e9dc47be83bc051d2cfecd9c : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/270a756d92bc4db1a5c222b87b3d6452 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e544920785c84d40aa8a39484ace595a : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9b62a8c374645aa805c710073ce2948 : SUCCESS in 6m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4fa42aec24824b56bc43657bbc7526a4 : SUCCESS in 2m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2ca5e26ae964bab8f17c8917e68b478 : SUCCESS in 4m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bfb7ed174f8b40cd8409129682c087b9 : SUCCESS in 29m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/72a27f1c8e4644cdbbd87bcf253985ee : SUCCESS in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"dfe9e57a49aed134324f467293e35d22d6e95ce2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-11 05:06:40.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"203fbc0d07e142914c3c01a6a1ab564ffdd3a851","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 05:06:52.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e72f6f7ba30e4c1bbc41538110331d6c\n\n\nWarning:\n  Change 874858 in project openstack/oslo.db does not share a change queue with 864816 in project openstack/designate","accounts_in_message":[],"_revision_number":14},{"id":"77f7aec4dc814373ebe7271786e6c2cbb6ec03ad","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-20 19:38:47.000000000","message":"Patch Set 14: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"c743fc33f0ad00058313136caf8793338ab927b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 20:45:38.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e4b11520dc44cdbb49bb156a505c1c3\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/75602e7edf1546c7adc3d9ce2952ee7a : SUCCESS in 45m 31s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/032a9b30a13340c490cff5f61bc7c9b9 : SUCCESS in 55m 54s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0b678eff9824cec83a96b40991cfbb5 : SUCCESS in 58m 43s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ee10a0cb3b174bfb968bbf62de861e3d : SUCCESS in 44m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/eb1e5c58350b4842996e0387d8c0c35c : SUCCESS in 45m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d9d119a1dfe74864aae8169da86e5e28 : SUCCESS in 1h 04m 10s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9b0187d057544c51873b3870ac2ebf2e : SUCCESS in 1h 06m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4687d2606c614b86b3c125c0af89f06d : SUCCESS in 45m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e1c4c3eb9c0640b5bdabf1d799e0a8da : SUCCESS in 51m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a843ad412093432385189d9663e7bbb2 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99b3eb9ba17f438ab53b2f821fc8359a : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc7e42f6a78c4adcafd489c3c300f0bd : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2d7f97564e149749cb1d8377afdf29e : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d36ab6551fd4344b51076b5ab61d6b9 : SUCCESS in 9m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/635fd07bba61482ca4ecd795a9364efb : SUCCESS in 1m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51ad5912022a4ebdaefe69e10aef65d9 : SUCCESS in 4m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b0e316e01dd54f019d5eb2f57d91aca9 : SUCCESS in 29m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0c136f502ad843f3a6759ada661a5a3a : SUCCESS in 4m 23s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"898efc3667987459e9a57bab1b61fc0038ccb3ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 20:46:11.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"60870392d5589ab36c5a80a9c75321c4c20cc778","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 20:54:50.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f69fd7aad2c24af5928dfe75bfebd22f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/538dd853f0604c598ea7dd39283fb487 : CANCELED\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/c6dfde2d81e3418896670939493f3548 : CANCELED\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ce38cf546d22414d995769d8b269b928 : CANCELED\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6e06707bb6ff405aa2dc8dc9fac43ba3 : CANCELED\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/65ab41a07a3e4ec7b1d34775f73af185 : CANCELED\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2500673d4df74b15a6bc44bb4901e458 : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7107bedf51654f08ba427282f829e6ab : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00580c2ede2e42c6aec7aaf5fb6ffd91 : SUCCESS in 6m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1b7efb20d8745b59dd20936b449bd37 : FAILURE in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf71be5707de4ec9be566ba6139d23ca : CANCELED\n- requirements-check https://zuul.opendev.org/t/openstack/build/184f61d8d36540f6a7868bcf43b6265d : SUCCESS in 2m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f02031da51ce4f21af78fcce8fd5f15c : CANCELED\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2349308d3ea644bf8577aa53afb55b3a : CANCELED","accounts_in_message":[],"_revision_number":14},{"id":"01419e37694619261317a270d8147f79d9ed025f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-20 21:14:24.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"c1d6cb644de4088006341afbb50ce43bef3de2a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 22:11:49.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/366dab4df40248679a94c546d8a5b460\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/88a2d6a8780c4a2d87b18bb2dadbf6a1 : SUCCESS in 45m 52s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9a43b931d0d645899a495d91045735a4 : SUCCESS in 46m 13s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3791972d7b4341298803616f92a8b5e4 : SUCCESS in 40m 27s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/56e04635a5b94502bb6b856b33c5fb08 : SUCCESS in 35m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/48c5e330623e43d5bc18f8459a198f84 : SUCCESS in 35m 47s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/32b2168f32ee437baa86490845eba3a4 : SUCCESS in 54m 49s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fcf5ab9f3ae843c28525626553b9493a : SUCCESS in 55m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6c54f03623714c9191b9ceebdfa59b5c : SUCCESS in 46m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3b2919128d1148a89081f3b80af39e8c : SUCCESS in 41m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d032f400c234ea6839feb39593fa6f3 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59df2f6a968a432296a0c11f1b4c5147 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d754fc113a34f5d99dfffcf3dc088a7 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d445994ad9b4eb3a30c22ce894bff8a : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef8c1ef4657a41ab82f00fbb337bea51 : SUCCESS in 7m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8a1f8b4cdd1847d08bcfe900ac0c87c6 : SUCCESS in 1m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f0658265e6e41a29d91991517c5bf77 : SUCCESS in 4m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2ea6660cc88845d1bb5c340b60065e22 : SUCCESS in 36m 42s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0787c0f5ca3d423aaac953fe54a659b4 : SUCCESS in 4m 44s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"66944c6a5b17a2fb55e76bdefaf2c3952562c938","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 22:12:23.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"f1bb8b269115da44d7b954b1ac5c688b15887de3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 22:57:48.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2403abc1539a436bafe302e9e33db99d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ec243b1e876d4fd8a46f7430b1357cac : SUCCESS in 38m 09s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/5f4c890672e444cf9062a7fe9618f47a : SUCCESS in 29m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c490d824aeca4b3b8120e2a9758e6c23 : SUCCESS in 29m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/19c364c2156e4abfb2d1977d50b5420c : SUCCESS in 44m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/417a895694a84885bcb5139338bda60a : SUCCESS in 28m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b884e08f2de04cdc81cf930e60603256 : SUCCESS in 30m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c680bdd7c51649fe8a2631c7c114c04a : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3952f50f79b4fd696ed63e740a21aca : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a3fad6a7a5b411b822ca6717f5d5895 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fce437b21104f67af59e37346637a74 : SUCCESS in 8m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/88c37eae86c04a7da3b75e17058f1735 : SUCCESS in 1m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8ad497c96144d37a2972092f2db5fa8 : SUCCESS in 4m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/71dce36636d248468879c86048495c92 : SUCCESS in 25m 34s","accounts_in_message":[],"_revision_number":14},{"id":"c89ed350ff99f99c1ebe67c5cabcd3e8ca6d5106","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 22:57:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"f458d214288ab96d771a123c3a4c890f826fcaca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 22:58:53.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/173a67db63d44a0a85dc97adf3fae067\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4097b6939f4441ed894e9534e7c8db15 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8124cc5d19574987b18bf9dc40b48671 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"05a112177d491ed39e2a5c3001719ad05ffe16fc","revisions":{"f0bebc21a6b2708036f75792f77731aec3cc48f6":{"kind":"REWORK","_number":1,"created":"2022-11-21 02:27:16.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/864816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/864816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/864816/1"}}},"commit":{"parents":[{"commit":"03c729c4791d8fd0feb115fa18b239d993771497","subject":"Cleaned up and optimized sqlalchemy base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/03c729c4791d8fd0feb115fa18b239d993771497"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-21 02:26:52.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-21 02:26:52.000000000","tz":-480},"subject":"[DNM] Messing with sqlalchemy v2.0","message":"[DNM] Messing with sqlalchemy v2.0\n\nChange-Id: I5b9cfb4b86bd7c342fd50d1b50dd12dce0c8e81a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0bebc21a6b2708036f75792f77731aec3cc48f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0bebc21a6b2708036f75792f77731aec3cc48f6"}]},"branch":"refs/heads/master"},"420dd45fe479706b32d1f4f4013b708497f4ffee":{"kind":"REWORK","_number":2,"created":"2022-11-21 04:16:35.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/864816/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/864816/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/864816/2"}}},"commit":{"parents":[{"commit":"03c729c4791d8fd0feb115fa18b239d993771497","subject":"Cleaned up and optimized sqlalchemy base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/03c729c4791d8fd0feb115fa18b239d993771497"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-21 02:26:52.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-21 04:16:29.000000000","tz":-480},"subject":"[DNM] Messing with sqlalchemy v2.0","message":"[DNM] Messing with sqlalchemy v2.0\n\nChange-Id: I5b9cfb4b86bd7c342fd50d1b50dd12dce0c8e81a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/420dd45fe479706b32d1f4f4013b708497f4ffee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/420dd45fe479706b32d1f4f4013b708497f4ffee"}]},"branch":"refs/heads/master"},"2032b78f6857c88f7f438d673001946bd90a4f1b":{"kind":"REWORK","_number":3,"created":"2022-11-21 04:23:22.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/864816/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/864816/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/864816/3"}}},"commit":{"parents":[{"commit":"03c729c4791d8fd0feb115fa18b239d993771497","subject":"Cleaned up and optimized sqlalchemy base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/03c729c4791d8fd0feb115fa18b239d993771497"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-21 02:26:52.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-21 04:23:16.000000000","tz":-480},"subject":"[DNM] Messing with sqlalchemy v2.0","message":"[DNM] Messing with sqlalchemy v2.0\n\nChange-Id: I5b9cfb4b86bd7c342fd50d1b50dd12dce0c8e81a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2032b78f6857c88f7f438d673001946bd90a4f1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2032b78f6857c88f7f438d673001946bd90a4f1b"}]},"branch":"refs/heads/master"},"7a6a6ec0be8569b4276cd83a87278b1804ae0c6c":{"kind":"REWORK","_number":4,"created":"2022-11-21 04:23:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/864816/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/864816/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/864816/4"}}},"commit":{"parents":[{"commit":"03c729c4791d8fd0feb115fa18b239d993771497","subject":"Cleaned up and optimized sqlalchemy base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/03c729c4791d8fd0feb115fa18b239d993771497"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-21 02:26:52.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-21 04:23:45.000000000","tz":-480},"subject":"[DNM] Messing with sqlalchemy v2.0","message":"[DNM] Messing with sqlalchemy v2.0\n\nChange-Id: I5b9cfb4b86bd7c342fd50d1b50dd12dce0c8e81a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7a6a6ec0be8569b4276cd83a87278b1804ae0c6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7a6a6ec0be8569b4276cd83a87278b1804ae0c6c"}]},"branch":"refs/heads/master"},"44548cbf0c8aaa0b0c4a8da87256c59665b2af48":{"kind":"REWORK","_number":5,"created":"2022-11-21 04:31:43.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/864816/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/864816/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/864816/5"}}},"commit":{"parents":[{"commit":"03c729c4791d8fd0feb115fa18b239d993771497","subject":"Cleaned up and optimized sqlalchemy base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/03c729c4791d8fd0feb115fa18b239d993771497"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-21 02:26:52.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-21 04:31:32.000000000","tz":-480},"subject":"[DNM] Messing with sqlalchemy v2.0","message":"[DNM] Messing with sqlalchemy v2.0\n\nChange-Id: I5b9cfb4b86bd7c342fd50d1b50dd12dce0c8e81a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/44548cbf0c8aaa0b0c4a8da87256c59665b2af48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/44548cbf0c8aaa0b0c4a8da87256c59665b2af48"}]},"branch":"refs/heads/master"},"e48a4767cfe7f9d9e8172236f7d4a872fd2ec0f0":{"kind":"REWORK","_number":6,"created":"2022-11-21 08:44:22.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/864816/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/864816/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/864816/6"}}},"commit":{"parents":[{"commit":"03c729c4791d8fd0feb115fa18b239d993771497","subject":"Cleaned up and optimized sqlalchemy base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/03c729c4791d8fd0feb115fa18b239d993771497"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-21 02:26:52.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-21 08:44:14.000000000","tz":-480},"subject":"[DNM] Messing with sqlalchemy v2.0","message":"[DNM] Messing with sqlalchemy v2.0\n\nChange-Id: I5b9cfb4b86bd7c342fd50d1b50dd12dce0c8e81a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e48a4767cfe7f9d9e8172236f7d4a872fd2ec0f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e48a4767cfe7f9d9e8172236f7d4a872fd2ec0f0"}]},"branch":"refs/heads/master"},"fa7815f641366dc977fb09e79e6c12e43cda1c58":{"kind":"REWORK","_number":7,"created":"2022-11-21 08:56:40.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/864816/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/864816/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/864816/7"}}},"commit":{"parents":[{"commit":"03c729c4791d8fd0feb115fa18b239d993771497","subject":"Cleaned up and optimized sqlalchemy base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/03c729c4791d8fd0feb115fa18b239d993771497"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-21 02:26:52.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-21 08:56:34.000000000","tz":-480},"subject":"[DNM] Messing with sqlalchemy v2.0","message":"[DNM] Messing with sqlalchemy v2.0\n\nChange-Id: I5b9cfb4b86bd7c342fd50d1b50dd12dce0c8e81a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa7815f641366dc977fb09e79e6c12e43cda1c58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa7815f641366dc977fb09e79e6c12e43cda1c58"}]},"branch":"refs/heads/master"},"7d9f5d8b5ff4da4a41f38a849915cd958fb633da":{"kind":"REWORK","_number":8,"created":"2023-03-05 22:52:56.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/16/864816/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/864816/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/864816/8"}}},"commit":{"parents":[{"commit":"8910a2014be52808cb3f4a42953b2176e852d0ad","subject":"Add hacking check for line continuation backslash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8910a2014be52808cb3f4a42953b2176e852d0ad"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-21 02:26:52.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-05 22:52:54.000000000","tz":0},"subject":"[DNM] Messing with sqlalchemy v2.0","message":"[DNM] Messing with sqlalchemy v2.0\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.db/+/874858\nChange-Id: I5b9cfb4b86bd7c342fd50d1b50dd12dce0c8e81a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7d9f5d8b5ff4da4a41f38a849915cd958fb633da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7d9f5d8b5ff4da4a41f38a849915cd958fb633da"}]},"branch":"refs/heads/master"},"d388a43e30222a568b79ddd2021cb7fb0859ce7e":{"kind":"REWORK","_number":9,"created":"2023-03-06 17:06:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/16/864816/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/864816/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/864816/9"}}},"commit":{"parents":[{"commit":"8910a2014be52808cb3f4a42953b2176e852d0ad","subject":"Add hacking check for line continuation backslash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8910a2014be52808cb3f4a42953b2176e852d0ad"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-21 02:26:52.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-06 17:06:37.000000000","tz":0},"subject":"[DNM] Messing with sqlalchemy v2.0","message":"[DNM] Messing with sqlalchemy v2.0\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.db/+/874858\nChange-Id: I5b9cfb4b86bd7c342fd50d1b50dd12dce0c8e81a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d388a43e30222a568b79ddd2021cb7fb0859ce7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d388a43e30222a568b79ddd2021cb7fb0859ce7e"}]},"branch":"refs/heads/master"},"8dd38815203bd3a4c97be81940eade9e1bb92593":{"kind":"REWORK","_number":10,"created":"2023-03-06 18:22:40.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/16/864816/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/864816/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/864816/10"}}},"commit":{"parents":[{"commit":"8910a2014be52808cb3f4a42953b2176e852d0ad","subject":"Add hacking check for line continuation backslash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8910a2014be52808cb3f4a42953b2176e852d0ad"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-21 02:26:52.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-06 18:22:09.000000000","tz":0},"subject":"[DNM] Messing with sqlalchemy v2.0","message":"[DNM] Messing with sqlalchemy v2.0\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.db/+/874858\nChange-Id: I5b9cfb4b86bd7c342fd50d1b50dd12dce0c8e81a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8dd38815203bd3a4c97be81940eade9e1bb92593"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8dd38815203bd3a4c97be81940eade9e1bb92593"}]},"branch":"refs/heads/master"},"4841273d7fc6fd53b64d7a0dfb9fb116a59ca569":{"kind":"REWORK","_number":11,"created":"2023-03-06 21:15:09.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/16/864816/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/864816/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/864816/11"}}},"commit":{"parents":[{"commit":"0d735bbecd775ffade2823f287da8b91c918aa44","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0d735bbecd775ffade2823f287da8b91c918aa44"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-21 02:26:52.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-06 21:14:45.000000000","tz":0},"subject":"[DNM] Messing with sqlalchemy v2.0","message":"[DNM] Messing with sqlalchemy v2.0\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.db/+/874858\nChange-Id: I5b9cfb4b86bd7c342fd50d1b50dd12dce0c8e81a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4841273d7fc6fd53b64d7a0dfb9fb116a59ca569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4841273d7fc6fd53b64d7a0dfb9fb116a59ca569"}]},"branch":"refs/heads/master"},"d1cb947412317b9b0b1023cb0cf2444e99bf3371":{"kind":"REWORK","_number":12,"created":"2023-03-06 23:01:31.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/16/864816/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/864816/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/864816/12"}}},"commit":{"parents":[{"commit":"0d735bbecd775ffade2823f287da8b91c918aa44","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0d735bbecd775ffade2823f287da8b91c918aa44"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-21 02:26:52.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-06 23:01:14.000000000","tz":0},"subject":"[DNM] Messing with sqlalchemy v2.0","message":"[DNM] Messing with sqlalchemy v2.0\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.db/+/874858\nChange-Id: I5b9cfb4b86bd7c342fd50d1b50dd12dce0c8e81a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d1cb947412317b9b0b1023cb0cf2444e99bf3371"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d1cb947412317b9b0b1023cb0cf2444e99bf3371"}]},"branch":"refs/heads/master"},"655ad75b81f894dac05d09858c0fd6c1cf163654":{"kind":"REWORK","_number":13,"created":"2023-03-10 22:53:27.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/16/864816/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/864816/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/864816/13"}}},"commit":{"parents":[{"commit":"0d735bbecd775ffade2823f287da8b91c918aa44","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0d735bbecd775ffade2823f287da8b91c918aa44"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-21 02:26:52.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-10 22:53:25.000000000","tz":0},"subject":"Update for sqlalchemy v2.x compatibility","message":"Update for sqlalchemy v2.x compatibility\n\nThis patch updates Designate to be compatible with SQLAlchemy 2.x.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.db/+/874858\nChange-Id: I5b9cfb4b86bd7c342fd50d1b50dd12dce0c8e81a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/655ad75b81f894dac05d09858c0fd6c1cf163654"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/655ad75b81f894dac05d09858c0fd6c1cf163654"}]},"branch":"refs/heads/master"},"05a112177d491ed39e2a5c3001719ad05ffe16fc":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2023-03-11 03:50:19.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/864816/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/864816/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/864816/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/864816/14"}}},"commit":{"parents":[{"commit":"758956d5b9321c72d63cbf3d8a1987d5dc6e6bb3","subject":"Removed unsupported test B309 from bandit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/758956d5b9321c72d63cbf3d8a1987d5dc6e6bb3"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-21 02:26:52.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-03-11 03:50:19.000000000","tz":0},"subject":"Update for sqlalchemy v2.x compatibility","message":"Update for sqlalchemy v2.x compatibility\n\nThis patch updates Designate to be compatible with SQLAlchemy 2.x.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.db/+/874858\nChange-Id: I5b9cfb4b86bd7c342fd50d1b50dd12dce0c8e81a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/05a112177d491ed39e2a5c3001719ad05ffe16fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/05a112177d491ed39e2a5c3001719ad05ffe16fc"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
