)]}'
{"id":"openstack%2Fdesignate~865355","triplet_id":"openstack%2Fdesignate~master~I44dc0499e03431261d5b596206858d94cc4803f9","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-11-23 06:34:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I44dc0499e03431261d5b596206858d94cc4803f9","subject":"Fix configuration leak in unit tests","status":"MERGED","created":"2022-11-23 01:10:49.000000000","updated":"2022-11-23 06:35:39.000000000","submitted":"2022-11-23 06:34:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865355","meta_rev_id":"e16e38a8a23d36402710890b5f4101361f4b37ef","_number":865355,"virtual_id_number":865355,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-23 06:34:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-11-23 02:15:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-23 05:40:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-23 05:40:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-23 02:15:23.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2022-11-23 02:16:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-23 05:40:54.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"a297ec82abf0a2775d87911dd208b10dc11940dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-23 01:10:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99057fc49dff18244f60eaa103db25dc1b501ea7","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-11-23 02:15:23.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b03b64308ff846de6969432efbd28190c252907b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 02:16:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94011ca38c56483aa8cf6eff3867425f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/61a4053e991e4b7489e3c2cb378f3b55 : SUCCESS in 41m 16s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c035dae515234c8a824206e025473258 : FAILURE in 1h 02m 50s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/4178a0d29b134f2795a86fb39d357b37 : SUCCESS in 41m 02s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/6f38c2c1c1d0490293b85f9a0770615c : SUCCESS in 39m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2156a3894c9f45c9b9f1392a852d5974 : SUCCESS in 36m 48s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/91e929cb18d74fd592ff95e4e321b2f5 : SUCCESS in 56m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/59e66d6a51dd4b9689fd9c6764a723bf : SUCCESS in 57m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7a677e7354394edd8c6bc5ac4ff08d2b : SUCCESS in 37m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bbd2cbccc1724177bb4431e17f117f88 : SUCCESS in 36m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/782962eeca8f4a9eb6a3f41d0c0ea3d6 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c0bc3959af843c48b79dbb8e1760861 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4880fcdf4da43d19f4970cc415a124b : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b41a58c928ed47328abfcaab83f2cb63 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eddbe819bfea4cfbb258de31530f57c1 : SUCCESS in 7m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/42137c3eb4a54e21926ce53a9dfa0f0f : SUCCESS in 30m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a862dad310db42cfab2aff95f94a022c : SUCCESS in 4m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9463131e7c5aced00af099f067786c758fbd38a4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-11-23 05:40:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6b941cbbf358250e3efe76879b43b1a46af5dc5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 05:41:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"042b72f019937ab75e9cf74e0afd6d30a28b3e30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 06:34:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2afa9dc39a74ba4806fa0ff7507a91e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1f7b1db84b2e42abb031acad3a681ff4 : SUCCESS in 33m 57s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/136e552d269842b59f506ea9b7d58480 : SUCCESS in 36m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bbbe0d2722a148149de37f3b16e548a4 : SUCCESS in 30m 44s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ec8cb83ebcda47f7bde0e967ef971a18 : SUCCESS in 46m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7ccd396e50904ff79015741a5980fa56 : SUCCESS in 33m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/92d97ba8dc6246ee827e4b8c84e7f3e5 : SUCCESS in 36m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/196dc40f764b420bbd0b3de357c403f4 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0b2c7453e1749988c988e575a248a55 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9fda42a816a5473182b0e2d88ff0a09b : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7852c5e312864366a9cd1a018382bbc3 : SUCCESS in 7m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6f14c25111594bb7856d210e6a29d09c : SUCCESS in 30m 28s","accounts_in_message":[],"_revision_number":1},{"id":"cc0fbd28c1339a65e04371b98a330ddadf6cd36e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 06:34:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e16e38a8a23d36402710890b5f4101361f4b37ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 06:35:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0c3cb1f9ef74544a9891d45e1c9a2d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/081b87d62b354d5da1164ef3d8ad179e : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3cdc8f2d6063648d42c256f400ea836b5923a7b","revisions":{"c3cdc8f2d6063648d42c256f400ea836b5923a7b":{"kind":"REWORK","_number":1,"created":"2022-11-23 01:10:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/55/865355/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/865355/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/865355/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/865355/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/865355/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/865355/1"}}},"commit":{"parents":[{"commit":"03c729c4791d8fd0feb115fa18b239d993771497","subject":"Cleaned up and optimized sqlalchemy base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/03c729c4791d8fd0feb115fa18b239d993771497"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-11-23 01:04:40.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-11-23 01:10:47.000000000","tz":0},"subject":"Fix configuration leak in unit tests","message":"Fix configuration leak in unit tests\n\nThere was a configuration leak in the agent based bind 9 tests, where if\nyou have a /etc/designate/designate.conf file with settings other than\nthe default the tests would fail.\nThis patch adds a configuration fixture setting override for the two\ntests to ensure consistent testing.\n\nChange-Id: I44dc0499e03431261d5b596206858d94cc4803f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c3cdc8f2d6063648d42c256f400ea836b5923a7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c3cdc8f2d6063648d42c256f400ea836b5923a7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
