)]}'
{"id":"openstack%2Fdesignate~866834","triplet_id":"openstack%2Fdesignate~master~I1fc02225b6676b68137245df2279340f215221f4","project":"openstack/designate","branch":"master","topic":"mod-proxy-http","attention_set":{},"removed_from_attention_set":{"8108":{"account":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"last_update":"2023-01-03 17:31:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1fc02225b6676b68137245df2279340f215221f4","subject":"Make sure proxy_http is always loaded for devstack","status":"MERGED","created":"2022-12-07 08:59:11.000000000","updated":"2023-01-03 17:32:59.000000000","submitted":"2023-01-03 17:31:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866834-mod-proxy-http","meta_rev_id":"b1bcceae17dbe5108028825db566ebc19b6ef265","_number":866834,"virtual_id_number":866834,"owner":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-03 17:31:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2023-01-03 08:32:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2023-01-03 08:32:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-03 08:32:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-01-03 16:29:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-07 10:12:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-07 19:09:26.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2022-12-07 19:24:54.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-12-07 19:33:18.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2023-01-03 16:29:09.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"c1798bc494a2de66a011488f605955fcf348f047","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-12-07 08:59:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f1ca56715d13e75d8c770fda331d12ad39ae140","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 10:12:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d89160ec57a4593957ac61bdfa5295d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fdd64dc1e85c4b7a9c65f80785887a86 : SUCCESS in 36m 30s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a7e3e793c72a4cbc992393f62bc4e611 : SUCCESS in 42m 23s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/07b37ffaeabc4df6a0a90a17afec1995 : SUCCESS in 39m 49s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/b8b198a993a54b689b30ec8822bd2e07 : SUCCESS in 37m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/364c8ac423834fec9f2b6e52b41c9624 : SUCCESS in 40m 11s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/fb316ce649e54270bfa44e03a7d82e22 : SUCCESS in 51m 35s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a663436976b841e396f2c277fb312284 : SUCCESS in 47m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4157b1d7691c454aa86627e0d4fbc63d : SUCCESS in 34m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/098dd6dd5762425abd5016a3fe52d461 : SUCCESS in 40m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9662b6ed57794eb48bd6043df05cace0 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6339eddd795647aeb31c319bce60353e : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/369e15e1d9e9496dac0ebea1dfe7a6bd : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0966c7747c749cf9aaf07e8682e178c : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d122ae504a2a4199923f764929a9468d : SUCCESS in 8m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/82d5499078414c7e89bfc60870448407 : SUCCESS in 26m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/99f2aeb9f27a41a9a6401f769aa10e41 : SUCCESS in 4m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a794273fd6a8da993bc11fed3ebf36af9eca618","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2022-12-07 19:09:26.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"346ecd9ace3b4970c079134b8b7b8c11ff213a5d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-12-07 19:24:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"675d177da0ad713190c6d3e8e3aed3008ec204ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 19:25:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5f377b91b5801108b058e664e16f4982fec8b4ab","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-12-07 19:33:18.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"48cbe2be25836514e9443b4664fd2acfc1bfb41b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 20:06:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c72fbb7bf84b423f97fe2ad57dd69354\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/52962b9082344669b951df4fef91ce7e : SUCCESS in 34m 22s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/d5cbbfeed4804d4e8e207f927754947c : SUCCESS in 32m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/78340cb1a3c64b32a7941a4416255f48 : SUCCESS in 29m 27s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1f888c0eb55d4f97a019bf2b6bfda3d2 : FAILURE in 18m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9e35e53ea42940d9b0e7942c19dae20c : SUCCESS in 29m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/79ecb20d9fb04d2688442a10356e1cd1 : SUCCESS in 31m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78be0a2814434beb9c7f518f78d8692f : FAILURE in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd9ac72ea4534e52bf6581f4f22f3f12 : FAILURE in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c3d322e89754e41a09282ce294a259d : FAILURE in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eed4ab0c69ad4e42be5b6d95c999392f : FAILURE in 3m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c433f1c03fdd4dcba4a0f23ef7d1298d : SUCCESS in 36m 56s","accounts_in_message":[],"_revision_number":1},{"id":"7b9e799e21f3b186bdf6eac1ad92e3b633471941","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-12-07 21:38:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"02a39e8b44b936edfe194792bb75a54e174fde40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 22:59:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa10d754a6d6415b82c780cf3025e67c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7d26312234a34390b3493e6f82dd70a0 : SUCCESS in 46m 46s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/42925e8cdec941f787da64d292fd702b : SUCCESS in 1h 14m 37s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f483eac57094c529525a7904ce5850c : SUCCESS in 56m 57s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/70002d5b5adb401da609907356a1aefb : SUCCESS in 45m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/864f64e556d547ccbb1354f90acb1c06 : SUCCESS in 42m 00s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/841a1baed1374fe6828042ef5ca9e033 : FAILURE in 38m 55s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/dbd5fd7e2c27415b87eb52b3c04336b2 : FAILURE in 33m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d494b4799f35427fb9a9c8096504434c : SUCCESS in 42m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d4306ca9bd7c4d149c6def7d23e77270 : SUCCESS in 46m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ab4f59555e249eaaed569ea37f11987 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e92c6b223588410381b645d3863a2999 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be526dcc12cc4421ae3360b26c6939be : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a904aa3c6ff14012a959898487305df2 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eae7e073aa25406282dc7e88e37fb792 : SUCCESS in 7m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9f68ef068387462ea2dcdf982bbae2fd : SUCCESS in 26m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/97de52cdda2942c6adfe93c0dcfb0fa7 : SUCCESS in 4m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5fc911d1ac2ace234fe86849a86ceff896d4dbb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-12-14 02:59:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"efed92daf5d22ebe6498b0df9084a76812653ed5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 04:01:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd7ab4565c284e32a7c31c4bce33c414\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/90546fe55e7749b892122abe5401d4d5 : SUCCESS in 41m 06s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/500fc52ac0034316af271997ae4eed1d : SUCCESS in 1h 00m 34s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/802379463d964d8086eea4fcd3a7d80a : SUCCESS in 50m 08s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/cd91cc5c2cec4850b8dfc3594be962fc : SUCCESS in 42m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f575908bf601490cb036542f3104dd37 : SUCCESS in 35m 37s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ecbecd1aaf904040a61f34db5a012dfd : SUCCESS in 57m 55s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c3727a1db68b498ead4ce1a4fc9a1435 : SUCCESS in 55m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3ee1b35327914fe2bff8af5e13cc0984 : SUCCESS in 44m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c978d09c19f4459f8e7a3a486cd59a99 : SUCCESS in 38m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6dab64013704ccdb6365d93ad1b497c : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ad19897783340c1af2b0bfc6fe3c5aa : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b8585cab33d4227a1bb6964f541e878 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55164bf670cc49398970908c251f82cd : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03a0600acbdc4cd0adf2ebadde54ad11 : SUCCESS in 7m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cb69752362d14a3ea6a0f8818dfb9775 : SUCCESS in 27m 48s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/dc56961a67fb4282b0000f1e346b6028 : SUCCESS in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7102d931c63f382b37ac290132291c2606881c76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2023-01-03 08:32:36.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"5ecc32acd695c44cea9e5ce5d9bfc451d1b0987c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 09:36:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e98d9afeabe46cd8c074349f355b970\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fe558ee13d2d4e008bb8d016232ee1bd : SUCCESS in 47m 24s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/cb9ac7b494164bbc8782a26bfbfba73d : SUCCESS in 54m 53s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c977f3b56dc4d4fa2138ed6cd2b3a25 : SUCCESS in 55m 29s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/efa29818417747648e3c612384699b69 : SUCCESS in 44m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/28d482ee1e5e45daa70e452f38121639 : SUCCESS in 44m 42s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c495568f7dd2481f8e20e12c5aeb199d : SUCCESS in 57m 24s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/37104ac298d24e2c86de05b418f17eab : SUCCESS in 56m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f0def781ada2429f8137b8c29f43c242 : SUCCESS in 46m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1b83df332b3d4975b44f84ff6b5617fc : SUCCESS in 39m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a42aa8f8c1574a3f9f5f0a8fc5aef008 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb028c249c8d475c8e919075b13a2c90 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d292e0a4dbe14a59af4fdef46be0f51b : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad206fc200ab45a89ca341723db291b6 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31bc7f0782c84007a1d6130c5f6a41b2 : SUCCESS in 7m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1325fd7d06e4492cbc934b43b66e1cda : SUCCESS in 24m 42s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/bdd49703bab84b4b88eb3b9070a82e2b : FAILURE in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1e1ac10c646f3fd80fc449a448d786736946005b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-01-03 16:29:09.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3c9e94b2a2a720fb6aa3cf44456ead11f1d2736a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 16:30:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b92b20ddaf48156528b53e4875086055799f622e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 17:31:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a50fd66847d4ae684f732fbfe05d976\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/aca7306018f942e39069c4325c4d7a0b : SUCCESS in 54m 59s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4f5b57ffdc2e47a3a40074895f33b5b1 : SUCCESS in 39m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/196e117842a142b981c424592e9e85d0 : SUCCESS in 39m 08s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/733d56575dbf48529a1bdf540b1ea81c : SUCCESS in 55m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f85b018b8dac4efca166af17848441e7 : SUCCESS in 33m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/919f71b8cbbb4e62949d78629da6b23b : SUCCESS in 47m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89091826f8254099b8693506e9721fea : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1236aae5148e4cd8b63a62b20fe9acab : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f16d64961bc045b3abe26b905b6d643e : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/087cad920c414c7e82178ad767383a78 : SUCCESS in 7m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e77b7d983df34ca7a54213dc293903bb : SUCCESS in 25m 20s","accounts_in_message":[],"_revision_number":3},{"id":"02cd2881240d052823103944830a7f0b36f007ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 17:31:47.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":3},{"id":"b1bcceae17dbe5108028825db566ebc19b6ef265","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 17:32:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6eb407b12b9445e869f11f4e2c0b81f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef4b7fc636cf40ed955d3b6d76f72eb5 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"04321d3324c0c290c79cc22df38ea5a767e05687","revisions":{"93ec394300345fc5c80db5bec723440ee61345b3":{"kind":"REWORK","_number":1,"created":"2022-12-07 08:59:11.000000000","uploader":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"ref":"refs/changes/34/866834/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/866834/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/866834/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/866834/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/866834/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/866834/1"}}},"commit":{"parents":[{"commit":"0483a533216c85314a0f3102d6600d12647f0c11","subject":"Merge \"Fix configuration leak in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0483a533216c85314a0f3102d6600d12647f0c11"}]}],"author":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2022-12-07 08:20:14.000000000","tz":540},"committer":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2022-12-07 08:56:36.000000000","tz":540},"subject":"Make sure proxy_http is always loaded for devstack","message":"Make sure proxy_http is always loaded for devstack\n\nWith the minimal Designate deployment with devstack, keystone module\ndoesn\u0027t load proxy_http so Designate API requests may fail. Explicitly\nenable the module to make sure the apache2 setup always works for\nDesignate.\n\nCloses-Bug: #1999019\nChange-Id: I1fc02225b6676b68137245df2279340f215221f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/93ec394300345fc5c80db5bec723440ee61345b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/93ec394300345fc5c80db5bec723440ee61345b3"}]},"branch":"refs/heads/master"},"171bccf7e6ff47fd5db2d79af1d0c236bb3fad29":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-12-14 02:59:00.000000000","uploader":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"ref":"refs/changes/34/866834/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/866834/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/866834/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/866834/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/866834/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/866834/2"}}},"commit":{"parents":[{"commit":"e6bf93a80af9d167c286382fa34a68706a5b2376","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e6bf93a80af9d167c286382fa34a68706a5b2376"}]}],"author":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2022-12-07 08:20:14.000000000","tz":540},"committer":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2022-12-14 02:59:00.000000000","tz":0},"subject":"Make sure proxy_http is always loaded for devstack","message":"Make sure proxy_http is always loaded for devstack\n\nWith the minimal Designate deployment with devstack, keystone module\ndoesn\u0027t load proxy_http so Designate API requests may fail. Explicitly\nenable the module to make sure the apache2 setup always works for\nDesignate.\n\nCloses-Bug: #1999019\nChange-Id: I1fc02225b6676b68137245df2279340f215221f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/171bccf7e6ff47fd5db2d79af1d0c236bb3fad29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/171bccf7e6ff47fd5db2d79af1d0c236bb3fad29"}]},"branch":"refs/heads/master","description":"Rebase"},"04321d3324c0c290c79cc22df38ea5a767e05687":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-01-03 08:32:36.000000000","uploader":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"ref":"refs/changes/34/866834/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/866834/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/866834/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/866834/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/866834/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/866834/3"}}},"commit":{"parents":[{"commit":"c6ddf55f7f0e32f7b8d24173cd963bbd37c11e0a","subject":"Fix tox4 error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c6ddf55f7f0e32f7b8d24173cd963bbd37c11e0a"}]}],"author":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2022-12-07 08:20:14.000000000","tz":540},"committer":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2023-01-03 08:32:36.000000000","tz":0},"subject":"Make sure proxy_http is always loaded for devstack","message":"Make sure proxy_http is always loaded for devstack\n\nWith the minimal Designate deployment with devstack, keystone module\ndoesn\u0027t load proxy_http so Designate API requests may fail. Explicitly\nenable the module to make sure the apache2 setup always works for\nDesignate.\n\nCloses-Bug: #1999019\nChange-Id: I1fc02225b6676b68137245df2279340f215221f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/04321d3324c0c290c79cc22df38ea5a767e05687"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/04321d3324c0c290c79cc22df38ea5a767e05687"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
